)]}'
{"id":"openstack%2Foctavia~872741","triplet_id":"openstack%2Foctavia~stable%2Fyoga~Ifa2ba3a9ba7505c4da6e391c63ca3342a606e4ed","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-02-07 22:02:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifa2ba3a9ba7505c4da6e391c63ca3342a606e4ed","subject":"Pin pylint on stable/branches","status":"MERGED","created":"2023-02-06 10:02:21.000000000","updated":"2023-02-07 22:03:27.000000000","submitted":"2023-02-07 22:02:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872741","meta_rev_id":"0c966d0bab3c0e0d8630c179680894390ec08aa1","_number":872741,"virtual_id_number":872741,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-07 22:02:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-02-07 15:43:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-06 10:02:39.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-02-07 15:43:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-06 10:02:39.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-02-06 12:15:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-07 15:43:54.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"00cc82e11488374b96cdb10bb7eb5a4d1935cdd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-06 10:02:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3cbcc82b8e70757c63d1eb091de10fe2a91e2a4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-02-06 10:02:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2ac9a74bbfbe896e4cb05430d4e37910f23227d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 12:15:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d82e2a1468434d0e9e12b0e597d75ff2\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a0ebe5e2a71d4f8cb524614b2100ad2b : SUCCESS in 1m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31d232b2d24f4aff8f67e4e278fe967c : SUCCESS in 8m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/753971f0e9964cc994c9e85b817f77e8 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79e3958c3f774ae1b5dc7ac0699bdef2 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60409a8dc16b47cca83fe9aad3df9c8f : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b74a4ad456b4ef5bd744cc1a0a6d32c : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22dbaae9c4ce4cdfa2df1fe66c0a21f7 : SUCCESS in 10m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45c18ff466364c8a9a0b245886490684 : SUCCESS in 4m 09s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c621d7fb9f9c47b293d500aa5fdc1169 : SUCCESS in 4m 45s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/aa2fefa8143b48bcbc73a9108f0e8908 : SUCCESS in 4m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c29c2e1a2bec4704959abda53e52047c : SUCCESS in 4m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2781ab90248642398360fe69ddf92aa8 : SUCCESS in 7m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/43300b4376634c83813dfb7bf921a0d8 : SUCCESS in 1h 20m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27c4c86225bd4a6ba1894ae3ae43b800 : SUCCESS in 1h 40m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6f054801b46747309bfd81a520400b48 : SUCCESS in 38m 30s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ffbe96eafb304e10a781b2abc90918cd : SUCCESS in 1h 14m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3f3af3ca13904330a02bdcb350a3d8f1 : SUCCESS in 35m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2bb742e6de424be69f6287d9061ce804 : SUCCESS in 2h 00m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c5c7be3413a14ec4a2dab3cf77fcf5d6 : FAILURE in 2h 10m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3097105aedc243ce9a67c1570a5ab0b7 : SUCCESS in 1h 52m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/deb160d5a1754023a2878258571d721f : SUCCESS in 1h 37m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c63adc709a714395885a277969baa7f4 : SUCCESS in 1h 55m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/14efa89b2d4a4cfcb2f837c2f45bbb78 : FAILURE in 41m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f780fe95264f07be3368af8a3e8cb732644eba2c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-07 15:43:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"95520c6b46fce9bf2fc3a751d102e0a10528ccb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 15:44:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bfaf331c95c955e0ff95497400e7a9812def6971","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 22:02:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b10e0b6843a42e1863fa573495619b1\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b8f54a9b33d84e0787b305d059f438e3 : SUCCESS in 2m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/931288434b434268a81d0954bffcba37 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dfe74bf1b6546be828d635931dbaae9 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63d2495d4e87469e95160aa5eb49071d : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7bfa89424b7c4beb97601e8799dfffc4 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6574fbec0d44538b092fd0a9642ae63 : SUCCESS in 11m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b46f86cd7b90424d8a2f73d67c7dfcce : SUCCESS in 7m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dfa94bc08bf54b2ab052a62e681458cc : SUCCESS in 3m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8e28b77e30db4f90ab3dbd092a064bd3 : SUCCESS in 9m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/80fb1734e1954e6f841ef3ab73540cf8 : SUCCESS in 1h 39m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d614b0f261b14578b8751b7b06f0e76c : SUCCESS in 2h 00m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/449480c573f94494956bd4cf6e7975f5 : SUCCESS in 52m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9a16e63cc3a34b6c95d515a99c2fdf43 : SUCCESS in 42m 49s","accounts_in_message":[],"_revision_number":1},{"id":"f9746894adcc858bbd107ff49b32bac3cd12aa63","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 22:02:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0c966d0bab3c0e0d8630c179680894390ec08aa1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 22:03:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53005f6b0ae64f92a3c79a9af6ecd767\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e6fe10424fc49dd956db424d384e6a8 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f72f9e8cbf874544acce1179ae87c020 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"23a7ca77c785f4a67457848dfc6da60d1a3b013b","revisions":{"23a7ca77c785f4a67457848dfc6da60d1a3b013b":{"kind":"REWORK","_number":1,"created":"2023-02-06 10:02:21.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/41/872741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/872741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/872741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/872741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/872741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/872741/1"}}},"commit":{"parents":[{"commit":"bc724fbebdcd85e4040b449ba2569e5810d92e96","subject":"Merge \"Filter out details from taskflow logs with v2+jobboard\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc724fbebdcd85e4040b449ba2569e5810d92e96"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-06 10:00:35.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-06 10:02:12.000000000","tz":60},"subject":"Pin pylint on stable/branches","message":"Pin pylint on stable/branches\n\npylint release 2.16.x adds new checkers, pin pylint to the previously\nused version.\n\nChange-Id: Ifa2ba3a9ba7505c4da6e391c63ca3342a606e4ed\n(cherry picked from commit 9d2a236f52a7ada5e18647ccef1b959a3fd4d0bf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23a7ca77c785f4a67457848dfc6da60d1a3b013b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23a7ca77c785f4a67457848dfc6da60d1a3b013b"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
