)]}'
{"id":"openstack%2Foctavia~872866","triplet_id":"openstack%2Foctavia~stable%2Fzed~I012ac70d511906a7401f92a3112c931b271dbefa","project":"openstack/octavia","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-02-08 10:48:24.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I012ac70d511906a7401f92a3112c931b271dbefa","subject":"Fix grenade job \u0026 pin pylint on stable/branches","status":"MERGED","created":"2023-02-07 10:19:43.000000000","updated":"2023-02-08 12:55:10.000000000","submitted":"2023-02-08 12:53:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872866","meta_rev_id":"b07346dc175a77e210b97c2176157ad0ee9b470b","_number":872866,"virtual_id_number":872866,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-08 12:53:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-02-08 10:48:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-08 10:44:56.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2023-02-08 10:48:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-07 10:39:37.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-02-07 12:57:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"451bde2e657cb9ac2ef0c0b5d199dababb19ad76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-07 10:19:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80f08de3f98987498616fc7a0851ad4ac19c33ca","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-02-07 10:39:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d4cd2e4cd729094a0b3709066646b27433ac4cad","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-07 12:24:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08c18104a738c82357ae89a8a83e0ff3d88a6580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 12:57:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ff49e120c2d45749b56407857992fbc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba7b783875bf4649bdbcf4b971a5d7c1 : SUCCESS in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1063a1192644bb899fd368cd8216c3c : FAILURE in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8b0f2e717ee4faeb65592366e017c21 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2de782b933364b6f91459fa142203701 : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3eb02a69ab54796a83cc91498fed149 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4dc966d3c1c41b38eb15fc4758834ac : SUCCESS in 12m 07s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e3718d6ad0374845849ffaa4bc1fc6f9 : SUCCESS in 7m 17s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a5035f23e48648e3a93d42191ed9a530 : SUCCESS in 6m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/92f1c03f4005409eb91b0fdd78f5a657 : SUCCESS in 4m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5cde9e8cebfd4520bd52db10973ba307 : SUCCESS in 7m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/68b00f2272da453d9d579cde4f8f13a0 : SUCCESS in 7m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/715f64ca3cb940d0a57829cffb99881f : SUCCESS in 7m 13s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1221da858eb445cfa615f5478b98a3a7 : SUCCESS in 1h 20m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/024c2aa00870449f97493a1628d5a104 : SUCCESS in 2h 02m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3ae12b5a5cd24509a2a0d395d6a012db : SUCCESS in 45m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/31858e252e004a09b439259fa6d7a1f8 : SUCCESS in 49m 22s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a21cdb05edd346988ce5158d2ebc58b5 : SUCCESS in 1h 11m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/66c13ebb143e4399b20cdfcde1f086ca : SUCCESS in 45m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/75b02f3a438e456d872e90f692cb2cd2 : SUCCESS in 1h 59m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1aecb10fde62431ea02974069e556479 : SUCCESS in 2h 02m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/05dc07ec248e4813978b33323ba905c4 : SUCCESS in 1h 55m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/acab1fbf739943d49d3fbf104ff28507 : SUCCESS in 1h 53m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2afc5d8cfb5c4be296ece34a5cc08558 : SUCCESS in 1h 52m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c3076c955624a6bbf9bac21883afc53 : TIMED_OUT in 2h 33m 38s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/amphora_info.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalived.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalivedlvs.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/loadbalancer.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/plug.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/server.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/util.py\n  Comments left for invalid file octavia/amphorae/backends/utils/haproxy_query.py\n  Comments left for invalid file octavia/amphorae/drivers/health/heartbeat_udp.py\n  Comments left for invalid file octavia/api/common/types.py\n  Comments left for invalid file octavia/api/drivers/amphora_driver/v1/driver.py\n  Comments left for invalid file octavia/api/v2/controllers/health_monitor.py\n  Comments left for invalid file octavia/api/v2/controllers/listener.py\n  Comments left for invalid file octavia/common/validate.py\n  Comments left for invalid file octavia/controller/worker/v1/flows/load_balancer_flows.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/flows/load_balancer_flows.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/database_tasks.py\n  Comments left for invalid file octavia/distributor/drivers/noop_driver/driver.py","accounts_in_message":[],"_revision_number":1},{"id":"fec1862935b6238078a02b3fba424078de114858","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-08 07:21:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"87ef96743f9505f6badc3fd96bb262f030ba02bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 10:42:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9403a595ce324c108f7c6d7d83eca21f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/923ad8bda9194b5cbe5a6dc8e618d5ca : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b27b2c6fd490420188a6ee604b90aebc : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbead28220dc49d395e480c2bb4aa209 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c9bad32dfe341ddb659e99883169869 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2455d6be1983440297a81f78d4d2097b : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/481ef5a96c9c410c866d957ab6d51699 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f56c305b64d5405fa139a88cf34c135d : SUCCESS in 12m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eba204e8dc6346c2b610089fec83dc1d : SUCCESS in 4m 54s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/75aa3cc1bf7e4715854253b19cddcea3 : SUCCESS in 4m 50s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/cbc00b43036745f3a049419d2ba61d47 : SUCCESS in 8m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/668ce6ec08c64b40a2fefdaf1cb512e1 : SUCCESS in 4m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7a402798870e46b3be86eccb924ff24e : SUCCESS in 8m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b8fae9a1548c4ac1a0897010913a583b : SUCCESS in 8m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/864893035c22418089949077b7d7350e : SUCCESS in 6m 25s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/821b04f781424f998105bdc99157010f : SUCCESS in 1h 31m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ba237dbef9ac454398fefadfc86c4101 : SUCCESS in 1h 45m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/42168de9d5a54dd5851d3b5f508ef4c5 : SUCCESS in 33m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8785d86c993748abb4fdef3be22b74c1 : SUCCESS in 38m 46s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/de8e8824c1de4122b9465177badf7839 : SUCCESS in 58m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/43bb40aa39ca49e0845e47ddb3fe2911 : SUCCESS in 31m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1a1a956ff1ba4ae6a0c8ec39ae6b3b48 : SUCCESS in 1h 39m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6c30f5276a9845adbf60705a8767f29c : TIMED_OUT in 3h 04m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2ca52262d2494f33bc6a10910a218c9b : SUCCESS in 1h 53m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/39bb9e7827e249b6828b94fa55f220ed : SUCCESS in 2h 05m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4f1be25a6eb64600ae6183ed04fec684 : SUCCESS in 1h 36m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/378d26c4c93a4d0eb988d885fcf9fa09 : TIMED_OUT in 2h 34m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"05f087230e5e6b21a8f3dd0113031384aa697226","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-02-08 10:44:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7896499456b745b0ecb79f078a38256400831de1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-08 10:48:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b002ba84e6b8a61addfc0097719b51bc6e4a1f79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 10:49:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7042e6f5e998deb1cd3f7bfd13f06afdab8e1442","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 12:53:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/086bdbfd6550427a8b51f8688ce41c1c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d01523aaa7d942a4a5330965631b7085 : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcb3ed7aeef24b6d8bb7a8a2e619071d : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de6eed473b2f403395c92daf3002b7b5 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3cffa7f58a345b9a8245e95984acaaa : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84423bc332e8478caa11792cc9ae489e : SUCCESS in 13m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/634d1892a57d4647a7779b4ffb52ae1d : SUCCESS in 4m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/edb251d0be7847a8ac3b4d91b8f389e6 : SUCCESS in 4m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dc26a909f58f499fadbd9c4576ec986d : SUCCESS in 8m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dc7bb2183cb84c10ab722db8091dbd08 : SUCCESS in 1h 58m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6d43918600ad41b98e0e1c54faccf6a5 : SUCCESS in 1h 50m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4a296f19d38d4e9583a5f47eb53d0803 : SUCCESS in 33m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/103f3a43e7ab40228dccab6f6eb4a1ec : SUCCESS in 36m 12s","accounts_in_message":[],"_revision_number":2},{"id":"fb9215a66c5ca2f204f9606b3d754522598fe938","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 12:53:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b07346dc175a77e210b97c2176157ad0ee9b470b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 12:55:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c7afab423cc4a368f9b61868eb15a82\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e32d360ea66145079f37e71ea10563e5 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/549bcece623a4b6baf1cbcf269b06fdb : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c87fbc07a3180201c6466f3571861ff1d63f7ce6","revisions":{"f5d7f1b5f70e512136ef8864b36166279fdeb787":{"kind":"REWORK","_number":1,"created":"2023-02-07 10:19:43.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/66/872866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/66/872866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/872866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/872866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/872866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/66/872866/1"}}},"commit":{"parents":[{"commit":"95bc6060e65879c6ebf00fe7510cd5b1cbbe79cc","subject":"Merge \"Fix PING health-monitor with recent Centos releases\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95bc6060e65879c6ebf00fe7510cd5b1cbbe79cc"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-06 20:36:29.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-07 10:19:36.000000000","tz":60},"subject":"Fix grenade job","message":"Fix grenade job\n\ndevstack_localrc is not used in grenade jobs, use\ngrenade_devstack_localrc instead.\n\nChange-Id: I012ac70d511906a7401f92a3112c931b271dbefa\n(cherry picked from commit a36e3ddab58d46991b71d8793061dbbd656d44fe)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5d7f1b5f70e512136ef8864b36166279fdeb787"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5d7f1b5f70e512136ef8864b36166279fdeb787"}]},"branch":"refs/heads/stable/zed"},"c87fbc07a3180201c6466f3571861ff1d63f7ce6":{"kind":"REWORK","_number":2,"created":"2023-02-08 07:21:41.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/66/872866/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/66/872866/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/872866/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/872866/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/872866/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/66/872866/2"}}},"commit":{"parents":[{"commit":"95bc6060e65879c6ebf00fe7510cd5b1cbbe79cc","subject":"Merge \"Fix PING health-monitor with recent Centos releases\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95bc6060e65879c6ebf00fe7510cd5b1cbbe79cc"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-06 20:36:29.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-08 07:21:08.000000000","tz":60},"subject":"Fix grenade job \u0026 pin pylint on stable/branches","message":"Fix grenade job \u0026 pin pylint on stable/branches\n\ndevstack_localrc is not used in grenade jobs, use\ngrenade_devstack_localrc instead.\n\npylint release 2.16.x adds new checkers, pin pylint to the previously\nused version.\n\nChange-Id: I012ac70d511906a7401f92a3112c931b271dbefa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c87fbc07a3180201c6466f3571861ff1d63f7ce6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c87fbc07a3180201c6466f3571861ff1d63f7ce6"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
