)]}'
{"id":"openstack%2Foctavia~872869","triplet_id":"openstack%2Foctavia~stable%2Fxena~I012ac70d511906a7401f92a3112c931b271dbefa","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-02-07 12:25:00.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I012ac70d511906a7401f92a3112c931b271dbefa","subject":"Fix grenade job","status":"MERGED","created":"2023-02-07 10:21:20.000000000","updated":"2023-02-07 15:18:32.000000000","submitted":"2023-02-07 15:16:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872869","meta_rev_id":"3c463f1fa4cb4caeb4cd093d9f278c91255fe7d1","_number":872869,"virtual_id_number":872869,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-07 15:16:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-02-07 12:25:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-07 10:40:01.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2023-02-07 12:25:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-07 10:40:01.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-02-07 13:05:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d64be21fd98bc4719a764c35bd30efff382f890f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-07 10:21:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c0e5055c863d08547ca647c1375fdd3dc6b0daa","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-02-07 10:40:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"828bd35eaabae18398061cf4c4614ca06affc40d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-07 12:25:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b45b68f9f62a897dbdc5c378d6c1ec9db0e15205","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 13:05:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10d4dfc028e04c4e98e98caa1f44ab40\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/831fa03d640242ff9f8f52935d93ed7c : SUCCESS in 11m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7663755927fe48b8a0360b9774e0c538 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8478db06af524cf59c7b30f70597ad35 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b5e8cbc8a084d4087e303b9053e8895 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dc2dead2e514fe899e0ce09c282c8ef : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d38d3b7566145da881b0429364dcd12 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f6c6b26b7cf4788ae1ec05bd48f5a4a : SUCCESS in 11m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0b138771ae504922b48301b3c82e5a4d : SUCCESS in 4m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c2b75d89735e4a7a9fe493ececa108f6 : SUCCESS in 10m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8467ebdbd3f245a783d1e3ab4d6ab920 : SUCCESS in 4m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e2e1f61e0dbe494c8b2276561eab3927 : SUCCESS in 7m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/165357295e49487eab5972890febd345 : SUCCESS in 1h 11m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5ad85300c6e4738bedececc84cede6e : SUCCESS in 1h 29m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b9332318ff73496dba78241e6d5ecaec : SUCCESS in 30m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/97a5ba5e22e2463199346fdf86cfd8f3 : SUCCESS in 40m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/57018598b55c4ea2a8f3ca6591c950c9 : FAILURE in 10m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b3fba2dbc07b45f585c698c29c27c309 : SUCCESS in 30m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6e114064fd4d496daa0753d5f0145fa5 : SUCCESS in 1h 35m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f3f5046bc6ad439482546d3827ff43ae : SUCCESS in 2h 08m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f451ec9372df4041afefd7c459330e73 : SUCCESS in 1h 45m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f8824166c714467090c401c38b91cbcd : SUCCESS in 1h 32m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4fae996c916a4b04a61e04a36660223d : TIMED_OUT in 2h 32m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2fa85d54625b4ce18a5c0d47665746cd : NODE_FAILURE Node request 300-0020381182 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"15ff642360100abfe14feeb2f05a9db4bf43cef0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 13:06:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac8057d0f1cc1270e891165c6bf0b3ca27a60715","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 15:16:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f2977e2c6c54ef2bd6d2bce5fffff19\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6227bc8fa2c48349aa995d67eb8f410 : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/694b77e840b149949f982dea1ceb7890 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/524efd94228742e7a0be0f6edba8cea0 : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9513d3153f344422a5ff49515febed09 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5773e9fe81f44677b385c7f50ff7255e : SUCCESS in 12m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a948fc2891c54d8f9f38ce7cdeca634c : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7f4a3544185844c3839ec2923f2d4379 : SUCCESS in 9m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7eeab0f0d1f84639b573f072f7a11dfc : SUCCESS in 1h 21m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a89455cd4ca94d649f760ae9a834e472 : SUCCESS in 2h 08m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4cbc0d77aaff426fa974536bd5af0001 : SUCCESS in 40m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/92ce0cdd8e344b128cf5229719337ca8 : SUCCESS in 40m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8b2f3eaf20aebf603ec076bca89b4c05d2ffb56a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 15:16:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3c463f1fa4cb4caeb4cd093d9f278c91255fe7d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 15:18:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67b1544099b7437db21f74a57d6b2be8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3088ee533954baeba8908a0b3534747 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d31b79dc8bad7b10e1b15cf9a6445fcb69ceee30","revisions":{"d31b79dc8bad7b10e1b15cf9a6445fcb69ceee30":{"kind":"REWORK","_number":1,"created":"2023-02-07 10:21:20.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/69/872869/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/872869/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/872869/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/872869/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/872869/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/872869/1"}}},"commit":{"parents":[{"commit":"cd677e1a8993bf29582a5deee74e2571c63b5b92","subject":"Merge \"Fix PING health-monitor with recent Centos releases\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cd677e1a8993bf29582a5deee74e2571c63b5b92"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-06 20:36:29.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-07 10:21:13.000000000","tz":60},"subject":"Fix grenade job","message":"Fix grenade job\n\ndevstack_localrc is not used in grenade jobs, use\ngrenade_devstack_localrc instead.\n\nChange-Id: I012ac70d511906a7401f92a3112c931b271dbefa\n(cherry picked from commit a36e3ddab58d46991b71d8793061dbbd656d44fe)\n(cherry picked from commit f5d7f1b5f70e512136ef8864b36166279fdeb787)\n(cherry picked from commit 4b588d4e37cea472ae70f44fe9e0591df83f609e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d31b79dc8bad7b10e1b15cf9a6445fcb69ceee30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d31b79dc8bad7b10e1b15cf9a6445fcb69ceee30"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
