)]}'
{"id":"openstack%2Foctavia~873489","triplet_id":"openstack%2Foctavia~master~I41f7e2341332b9cb74b4a59fedb6eed1af3c8062","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-09-20 18:27:36.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-09-20 16:25:03.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-05-23 08:00:49.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I41f7e2341332b9cb74b4a59fedb6eed1af3c8062","subject":"Add support for Rocky Linux","status":"MERGED","created":"2023-02-13 07:37:54.000000000","updated":"2023-09-20 18:30:42.000000000","submitted":"2023-09-20 18:27:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"873489","meta_rev_id":"53c15ff8e13518431d7ca4e5599cdbf79aa913a7","_number":873489,"virtual_id_number":873489,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-20 18:27:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2023-05-26 16:15:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2023-05-30 12:21:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-08-22 20:37:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-23 08:00:49.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2023-09-20 16:25:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 07:43:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-22 08:49:26.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"CC"},{"updated":"2023-05-22 09:14:30.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-05-26 16:15:59.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-05-30 12:21:07.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-08-22 20:37:21.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"569d9736266bcbafd6cc330305697c039c89b32b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-13 07:37:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11331ce12d40a33fdb7e22c725680c79f72bca56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 07:43:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2121c7d7d2b4f308faeb4c58cc73c70\n\n- octavia-amphora-image-build-live-rocky-9 https://zuul.opendev.org/t/openstack/build/e6d30d2e4e0e4aed9f60ba808e63835a : FAILURE in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"bc15eaf7afa58ce7abfbfb1f7124de4bfc32e1d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-13 07:46:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1013a282c59c4980b547769593e080a128a072d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 07:54:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc1ad0084b9d4bea8e68b543eba0242a\n\n- octavia-amphora-image-build-live-rocky-9 https://zuul.opendev.org/t/openstack/build/7c8816629e104e718c3a539ae2589153 : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":2},{"id":"b3e4fd3bcff8093a3ecaeb2cf790b011f4793c26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-13 08:40:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3ec7c0437fd467a9fa026c860eaa5c85c5cc2721","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 09:02:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34390d6f6b614cf089eb3af83ed2fbff\n\n- octavia-amphora-image-build-live-rocky-9 https://zuul.opendev.org/t/openstack/build/f7445755eea248169a8b353375ed04b5 : SUCCESS in 8m 36s","accounts_in_message":[],"_revision_number":3},{"id":"20b91d317420c21dc81f0cc87d0f616cef2537b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-13 09:52:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cff0da96b66256cfbf4194faf18c27daca579add","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 10:46:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6373334d1f5444f69f3398104ed81ca2\n\n- octavia-amphora-image-build-live-rocky-9 https://zuul.opendev.org/t/openstack/build/f2d2d1e4be3c43298fc7e3a4101629ea : SUCCESS in 10m 25s","accounts_in_message":[],"_revision_number":4},{"id":"4c590f4326c9483c3062a939c63d0fd1e46d01f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-13 20:09:05.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"af30474ffb87e711002b31efbfc13924142b9316","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 20:25:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ba915e0a5884166a6b2de74df79c542\n\n- octavia-amphora-image-build-live-rocky-9 https://zuul.opendev.org/t/openstack/build/7df55e2c6905497fab73eef416250c7c : SUCCESS in 11m 30s","accounts_in_message":[],"_revision_number":5},{"id":"cab45914ebb487b425c2cfcc7a801e1e99c7ed4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-14 06:22:15.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"733e145aa722c35386c8ff96adef38db6575a190","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 06:32:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e209ceae7d24447591f10a1e5c378873\n\n- octavia-amphora-image-build-live-rocky-9 https://zuul.opendev.org/t/openstack/build/c8e8632141154736941b0ac89110df68 : SUCCESS in 9m 23s","accounts_in_message":[],"_revision_number":6},{"id":"056b2b98bbd7f03fc6f569778236c816ff1d9729","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-15 09:41:18.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"b7ac5198a7ae70a9f1fa83c60a9ae7c6868b5fc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 09:57:41.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d6d7b451efe43a18b347fd0b1a4cf42\n\n- octavia-amphora-image-build-live-rocky-9 https://zuul.opendev.org/t/openstack/build/0ca33cfc94e245de9b318a17e18a4b09 : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":7},{"id":"07ec964179d33324a0ff05bf2e33482176df41ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-16 08:22:15.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"6cbc95a95db12a575c50b88539b40f0a1ca1a78b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 08:37:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbefc14953c24e9db296e3751106359b\n\n- octavia-amphora-image-build-live-rocky-9 https://zuul.opendev.org/t/openstack/build/ed7daf42fe4040a3a8bc41ee909a786b : SUCCESS in 10m 09s","accounts_in_message":[],"_revision_number":8},{"id":"ac041ba04b88c47a01735eea1d80ef6813bbf688","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-16 10:47:54.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"896bb15b5402401e08f6affece29ac8a8beabc0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 13:26:59.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d22ebc2f1a1b460f83773ac9e25f5e39\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c92691ac6b7a47ac84b52eefe6649393 : SUCCESS in 12m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dd269f2335a42fd829aa9ea58713ddf : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb9adc0195964b89ad840e3f61f6ffe0 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/876630ea563c415f8c2f49eebc7cb829 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/afbbd9e8484e4dddb8956a610512bb68 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af7c255662ac4c5e8d8fb112f064fc2f : SUCCESS in 11m 34s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d002ae12bef0423e8cb21d749b9edc91 : SUCCESS in 5m 07s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c317ef0bbc5a4781b3c480abb14a092e : SUCCESS in 7m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5761d47191d24f5eb8671ffb365b86fc : SUCCESS in 5m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3a2585cf2880447585cdcbd93f51ee2e : SUCCESS in 7m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cce8d145a1624d2181a5c2f69576c366 : SUCCESS in 7m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4e6ed13bde0d444c875401b4f4eaf7d7 : SUCCESS in 8m 18s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/76c425cd7e144638963ca81bb5cd2f20 : SUCCESS in 1h 55m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b67d9cbb904f481d95a1589a363062c7 : TIMED_OUT in 2h 32m 47s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/5d052e5e5ddb44c885e6af8845fddc8b : TIMED_OUT in 2h 32m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/716ad1cb68a64809b281d04692bb12ea : SUCCESS in 47m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e174cc162ed04225975ee545c4cbc5a9 : SUCCESS in 49m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4b04a97159ca4d9ca52a7f511a6b86b6 : SUCCESS in 1h 11m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5f764963219648638103108dd07987f8 : SUCCESS in 38m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9190efec4ea24eaf807d025a519b65de : TIMED_OUT in 2h 33m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cbe9b4f2dcdc4b83a4e90f32d7e7864c : SUCCESS in 2h 35m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/eaa941eab6084c1f9cfaa6a30a34fb0d : SUCCESS in 2h 00m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/fcf41a48bded434a833e5c531b2fc061 : TIMED_OUT in 2h 36m 05s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5fccf9972a9870a1d41172c5223f7217e35a4403","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-22 08:49:26.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9d6d7203fec101a1d7902ad8dd028e2190fb4671","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-22 09:03:33.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"17fb8587c51fcaf8e80e52a83afb6a0ef2f38c10","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-22 09:14:30.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f95b27e28514bba3095fe83395800257a7900347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 11:53:13.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0dffb8c830943d89633769b45e31b3c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff0b361c821947abab1611a264864a15 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcb34098e1664351ab1a1188abbeaa15 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe3f3c78b36f4f10ba22b35cdea06b5a : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb8ef141c2d04481a26280cefbe157f7 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ebd3858f2174b1d90015f65f295199c : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c653be15b043bbac2210d7e0288f98 : SUCCESS in 11m 07s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c5c2650b0b09470dbb37e087c5205900 : SUCCESS in 4m 58s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e2792b53ea394802adffa3383f3af5ec : SUCCESS in 8m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d08de9ad20274e0585ab539cc46664ee : SUCCESS in 4m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f653c076170141639bdb05dc7e80c874 : SUCCESS in 7m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/67deadde4b2243d5b7364d6f1b618bc6 : SUCCESS in 7m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6117b9f93d9f477098dd3edcac067112 : SUCCESS in 6m 50s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/254e810574d141f1bd573a504ac45755 : SUCCESS in 1h 40m 36s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b49eed815eca4c1f924d67d38f13bea6 : SUCCESS in 1h 59m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/47151e155b224541944074beac1af490 : SUCCESS in 1h 24m 00s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1f48fc78552d4cf0a297616d3985f1fb : SUCCESS in 1h 58m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8416df6fead8417fa896bbda38591e17 : SUCCESS in 1h 23m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/70bb62b86f80438284e11e5684508699 : SUCCESS in 40m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b724fe77bb2b48ab8caaa21d4e8071f2 : SUCCESS in 39m 58s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/47ed3bf729254842bd2654bb37bba9cb : FAILURE in 11m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6b2f3adc65e44b9db81818777c310cc3 : SUCCESS in 35m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/14d8c59eb79a45e6a7880f05589a07a4 : TIMED_OUT in 2h 32m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/904fe7495f874601be295ba921b5cd0c : SUCCESS in 3h 02m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4963de727a854042a7653911a829f10f : SUCCESS in 2h 13m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/6562a1a90a67495eac34f9e93a1d9128 : TIMED_OUT in 2h 33m 44s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"0b31e2318cea14d291b9b1512d38eaec5af92046","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-23 07:50:25.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"9e9b0997787d4904b76aaa3c7368601e91ba4f2d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-23 07:50:45.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"bbfd01729c4fc718dda4d9cf901763b152072f93","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-23 08:00:49.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"49b689eacaf12d218e3fc90516eef023fd23fef4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 10:54:54.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72404113e2b1448fba36c39c6310ff81\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5680fc05f9ff4fe1a7686066d4bb7a46 : SUCCESS in 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87de18c056b64559a3488435d4718c81 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c10b54fabb5047919e005b0a73e448e4 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1f09178f31446089833c8e3c2035ab0 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf2463d89aa7441fb44ba79dc74d0761 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cce2856af2c146919419717950cdb6d7 : SUCCESS in 11m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15b855b73ebf4828adf1cf59baf291a9 : SUCCESS in 4m 56s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f8dc52901ca34e638d8e0d64c80e9a20 : SUCCESS in 5m 00s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b7278f58807744e39e6123915eaaf0fe : SUCCESS in 10m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9e05c45e9ad84d97840662f25d3c2e19 : SUCCESS in 4m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/974e9d829b4545d68182149a5cd6602e : SUCCESS in 7m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6bde7e5d8f2f4d489ac63ce77b25e15d : SUCCESS in 7m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6a4594d60b7c4f4ba48030d058edac3d : SUCCESS in 7m 45s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/be9a6cb0d9b94a229edaad5809f99be8 : SUCCESS in 1h 19m 08s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2f6587b21c484d4dad588e631fc3c5ca : SUCCESS in 1h 50m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8bd7243683dc4c089321271be0d61e4c : SUCCESS in 1h 27m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/bedbecb187b0435d9a6143ce842606e6 : SUCCESS in 1h 46m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d8dba5231a9c49b4b989fae4779c53ac : SUCCESS in 1h 11m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d73aeb219a2e4f6e9dbd5a7d0b6b449e : SUCCESS in 40m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/af37cc01ed46423c8fd1397d11e0a303 : SUCCESS in 37m 48s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1165be31469f43658e3f1f26bc871539 : FAILURE in 15m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e899175b2ecb4018ac31e0f6c205b697 : SUCCESS in 32m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8b00fcf9658b472f9182f0747e7dedbf : SUCCESS in 2h 17m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/34cb0d9f5b284bd7890d25855427d553 : TIMED_OUT in 3h 03m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2b7b6a4a27d5461680cc250dc01d31a2 : SUCCESS in 2h 32m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/013cfd8595584624b3c13eb4b2145ce2 : TIMED_OUT in 2h 33m 37s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"127393894f048e5637f660e402306f7ea0de69c4","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-05-26 16:15:59.000000000","message":"Patch Set 10: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"5aac2b9443b07eb2371b7947b6562c5636085e0b","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-05-30 12:21:07.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"6f46a531d8b87680637aafa005bd51bc01b5cb3a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-22 20:37:21.000000000","message":"Patch Set 10: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"50d8d8d0151b55d77b1fa8d56d82d8dcac1dd7ee","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-08-24 12:09:49.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"5d997d498246eccfe92eee7259df96985c10aa8a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-08-24 12:13:51.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"df80accc928a86bf8601e403c8ac0d599ee44a72","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-09-20 16:25:03.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"badad4041a221fb8cc7308a2fea3d084be72bafc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 16:25:46.000000000","message":"Patch Set 10: -Verified\n\n(1 comment)\n\nStarting gate jobs.\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/octavia on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: octavia-tox-functional-py39-sqlalchemy-tips\n      parent: openstack-tox-functional-py39\n      description: |\n        Run tox python 3.9 functional against master of sqlalchemy and related projects\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n        - name: github.com/sqlalchemy/sqlalchemy\n          override-checkout: main\n        - name: github.com/sqlalchemy/alembic\n          override-checkout: main\n\n  in \"openstack/octavia/zuul.d/jobs.yaml@master\", line 161, column 3","accounts_in_message":[],"_revision_number":10},{"id":"1f1ff3ad3d23e3f6df847e1e6d9795b0cf1517cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 18:27:35.000000000","message":"Patch Set 10: Verified+2\n\n(1 comment)\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d201301b75f40a7b7a23b22606c28a3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/520d35f696fa4cca97fa1603e8399034 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45526b8dcd054550a5b5ac0a929cab13 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c34a64d5056c4eb3b42facc0b044437f : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12aada583c6942f1a82192169299a3c6 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aacc899fa9f4c198a64cd3782750f84 : SUCCESS in 9m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb48311c08fc44828a3d9e49e26b13c5 : SUCCESS in 5m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/58aa445c825a4b88b491b0d8fff431ac : SUCCESS in 4m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0e8c97927bdb4c1a98a867e1ccd42b93 : SUCCESS in 8m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/40005b2f87a3494eb6a4111e1fb78931 : SUCCESS in 1h 05m 01s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/def6ca82b6d14609ac33834005b7707e : SUCCESS in 2h 01m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c2cdbfb29ff747cba85600a62d6374ea : SUCCESS in 1h 18m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a4b3bf5a7f0747169097c9e26e85c923 : SUCCESS in 41m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bc136932ecaa4a6598e641d32196fa8d : SUCCESS in 40m 32s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/octavia on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: octavia-tox-functional-py39-sqlalchemy-tips\n      parent: openstack-tox-functional-py39\n      description: |\n        Run tox python 3.9 functional against master of sqlalchemy and related projects\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n        - name: github.com/sqlalchemy/sqlalchemy\n          override-checkout: main\n        - name: github.com/sqlalchemy/alembic\n          override-checkout: main\n\n  in \"openstack/octavia/zuul.d/jobs.yaml@master\", line 161, column 3","accounts_in_message":[],"_revision_number":10},{"id":"746815cd43be876cfb4c8ac80dd3ebfdb4d1a3cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 18:27:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"53c15ff8e13518431d7ca4e5599cdbf79aa913a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 18:30:42.000000000","message":"Patch Set 10:\n\n(1 comment)\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a6e853aa211490caf65b49838d0d5c2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e2f9c7e196840d29b3ffa6706982684 : SUCCESS in 1m 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f7a2e8a82eb44c7aab9949eae798e2e : SUCCESS in 1m 08s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/octavia on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: octavia-tox-functional-py39-sqlalchemy-tips\n      parent: openstack-tox-functional-py39\n      description: |\n        Run tox python 3.9 functional against master of sqlalchemy and related projects\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n        - name: github.com/sqlalchemy/sqlalchemy\n          override-checkout: main\n        - name: github.com/sqlalchemy/alembic\n          override-checkout: main\n\n  in \"openstack/octavia/zuul.d/jobs.yaml@master\", line 161, column 3","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"4e6e8f0c0e6db892f5f60437541531f84aabc7cb","revisions":{"cda2fc2c7650bca66eb9a906344b378b812d9388":{"kind":"REWORK","_number":1,"created":"2023-02-13 07:37:54.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/873489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/873489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/873489/1"}}},"commit":{"parents":[{"commit":"79a564775501a57fb1b635bafcb600ab9792a191","subject":"Merge \"fix haproxy_count to display the number of running processes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79a564775501a57fb1b635bafcb600ab9792a191"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-13 07:15:40.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-13 07:17:34.000000000","tz":-300},"subject":"WIP Add support for Rocky Linux amphora image","message":"WIP Add support for Rocky Linux amphora image\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/869424\n\nChange-Id: I41f7e2341332b9cb74b4a59fedb6eed1af3c8062\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cda2fc2c7650bca66eb9a906344b378b812d9388"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cda2fc2c7650bca66eb9a906344b378b812d9388"}]},"branch":"refs/heads/master"},"c44f159e9dc5b9010e8a58f299a59fb7060d4fa6":{"kind":"REWORK","_number":2,"created":"2023-02-13 07:46:35.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/873489/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/873489/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/873489/2"}}},"commit":{"parents":[{"commit":"79a564775501a57fb1b635bafcb600ab9792a191","subject":"Merge \"fix haproxy_count to display the number of running processes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79a564775501a57fb1b635bafcb600ab9792a191"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-13 07:15:40.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-13 07:26:25.000000000","tz":-300},"subject":"WIP Add support for Rocky Linux amphora image","message":"WIP Add support for Rocky Linux amphora image\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/869424\n\nChange-Id: I41f7e2341332b9cb74b4a59fedb6eed1af3c8062\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c44f159e9dc5b9010e8a58f299a59fb7060d4fa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c44f159e9dc5b9010e8a58f299a59fb7060d4fa6"}]},"branch":"refs/heads/master"},"2fe1ff5e436cb1d8c36d540a1ac60dcafdd69cf1":{"kind":"REWORK","_number":3,"created":"2023-02-13 08:40:51.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/873489/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/873489/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/873489/3"}}},"commit":{"parents":[{"commit":"79a564775501a57fb1b635bafcb600ab9792a191","subject":"Merge \"fix haproxy_count to display the number of running processes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79a564775501a57fb1b635bafcb600ab9792a191"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-13 07:15:40.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-13 08:20:42.000000000","tz":-300},"subject":"WIP Add support for Rocky Linux amphora image","message":"WIP Add support for Rocky Linux amphora image\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/869424\n\nChange-Id: I41f7e2341332b9cb74b4a59fedb6eed1af3c8062\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2fe1ff5e436cb1d8c36d540a1ac60dcafdd69cf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2fe1ff5e436cb1d8c36d540a1ac60dcafdd69cf1"}]},"branch":"refs/heads/master"},"96ed52b8e50c970c77012f7680eee7eea19654c6":{"kind":"REWORK","_number":4,"created":"2023-02-13 09:52:23.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/873489/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/873489/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/873489/4"}}},"commit":{"parents":[{"commit":"79a564775501a57fb1b635bafcb600ab9792a191","subject":"Merge \"fix haproxy_count to display the number of running processes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79a564775501a57fb1b635bafcb600ab9792a191"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-13 07:15:40.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-13 09:32:13.000000000","tz":-300},"subject":"WIP Add support for Rocky Linux amphora image","message":"WIP Add support for Rocky Linux amphora image\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/869424\n\nChange-Id: I41f7e2341332b9cb74b4a59fedb6eed1af3c8062\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96ed52b8e50c970c77012f7680eee7eea19654c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96ed52b8e50c970c77012f7680eee7eea19654c6"}]},"branch":"refs/heads/master"},"9cd2164893947bdd9362d338630f36c09cc41f3d":{"kind":"REWORK","_number":5,"created":"2023-02-13 20:09:05.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/873489/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/873489/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/873489/5"}}},"commit":{"parents":[{"commit":"79a564775501a57fb1b635bafcb600ab9792a191","subject":"Merge \"fix haproxy_count to display the number of running processes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79a564775501a57fb1b635bafcb600ab9792a191"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-13 07:15:40.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-13 19:48:53.000000000","tz":-300},"subject":"WIP Add support for Rocky Linux amphora image","message":"WIP Add support for Rocky Linux amphora image\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/869424\n\nChange-Id: I41f7e2341332b9cb74b4a59fedb6eed1af3c8062\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9cd2164893947bdd9362d338630f36c09cc41f3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9cd2164893947bdd9362d338630f36c09cc41f3d"}]},"branch":"refs/heads/master"},"0e8f64def39e47aa727517b6445b381d0f63ec2f":{"kind":"REWORK","_number":6,"created":"2023-02-14 06:22:15.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/873489/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/873489/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/873489/6"}}},"commit":{"parents":[{"commit":"79a564775501a57fb1b635bafcb600ab9792a191","subject":"Merge \"fix haproxy_count to display the number of running processes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/79a564775501a57fb1b635bafcb600ab9792a191"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-13 07:15:40.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-14 06:02:02.000000000","tz":-300},"subject":"WIP Add support for Rocky Linux amphora image","message":"WIP Add support for Rocky Linux amphora image\n\nDepends-On: https://review.opendev.org/c/openstack/diskimage-builder/+/869424\n\nChange-Id: I41f7e2341332b9cb74b4a59fedb6eed1af3c8062\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e8f64def39e47aa727517b6445b381d0f63ec2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e8f64def39e47aa727517b6445b381d0f63ec2f"}]},"branch":"refs/heads/master"},"6f6472c09b2106231d7e85f2f19d6adb2d5a02c4":{"kind":"REWORK","_number":7,"created":"2023-05-15 09:41:18.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/873489/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/873489/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/873489/7"}}},"commit":{"parents":[{"commit":"bb7c8ca2c9b0db7e65fdaecce05c694929299054","subject":"Merge \"Remove python-neutronclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb7c8ca2c9b0db7e65fdaecce05c694929299054"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-13 07:15:40.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-15 09:41:05.000000000","tz":120},"subject":"WIP Add support for Rocky Linux amphora image","message":"WIP Add support for Rocky Linux amphora image\n\nChange-Id: I41f7e2341332b9cb74b4a59fedb6eed1af3c8062\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f6472c09b2106231d7e85f2f19d6adb2d5a02c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f6472c09b2106231d7e85f2f19d6adb2d5a02c4"}]},"branch":"refs/heads/master"},"60275945d7dd60f8833b8ab0f79e76bcd3d7bda3":{"kind":"REWORK","_number":8,"created":"2023-05-16 08:22:15.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/873489/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/873489/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/873489/8"}}},"commit":{"parents":[{"commit":"bb7c8ca2c9b0db7e65fdaecce05c694929299054","subject":"Merge \"Remove python-neutronclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb7c8ca2c9b0db7e65fdaecce05c694929299054"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-13 07:15:40.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-16 08:21:30.000000000","tz":120},"subject":"WIP Add support for Rocky Linux amphora image","message":"WIP Add support for Rocky Linux amphora image\n\nChange-Id: I41f7e2341332b9cb74b4a59fedb6eed1af3c8062\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60275945d7dd60f8833b8ab0f79e76bcd3d7bda3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60275945d7dd60f8833b8ab0f79e76bcd3d7bda3"}]},"branch":"refs/heads/master"},"1bda41a8e86899debb89b74e91f4c8213978b74a":{"kind":"REWORK","_number":9,"created":"2023-05-16 10:47:54.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/873489/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/873489/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/873489/9"}}},"commit":{"parents":[{"commit":"bb7c8ca2c9b0db7e65fdaecce05c694929299054","subject":"Merge \"Remove python-neutronclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb7c8ca2c9b0db7e65fdaecce05c694929299054"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-13 07:15:40.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-16 10:47:36.000000000","tz":120},"subject":"Add support for Rocky Linux","message":"Add support for Rocky Linux\n\n* Added support for Rocky Linux in the amphora-agent\n* Amphora images for Rocky can be built when setting\n  OCTAVIA_AMP_BASE_OS\u003d\"rocky\"\n* Fixed the devstack plugin for Rocky Linux hosts\n\nChange-Id: I41f7e2341332b9cb74b4a59fedb6eed1af3c8062\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1bda41a8e86899debb89b74e91f4c8213978b74a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1bda41a8e86899debb89b74e91f4c8213978b74a"}]},"branch":"refs/heads/master"},"4e6e8f0c0e6db892f5f60437541531f84aabc7cb":{"kind":"REWORK","_number":10,"created":"2023-05-23 07:50:25.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/873489/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/873489/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/873489/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/873489/10"}}},"commit":{"parents":[{"commit":"bb7c8ca2c9b0db7e65fdaecce05c694929299054","subject":"Merge \"Remove python-neutronclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb7c8ca2c9b0db7e65fdaecce05c694929299054"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-13 07:15:40.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-23 07:49:25.000000000","tz":120},"subject":"Add support for Rocky Linux","message":"Add support for Rocky Linux\n\n* Added support for Rocky Linux in the amphora-agent\n* Amphora images for Rocky can be built when setting\n  OCTAVIA_AMP_BASE_OS\u003d\"rocky\"\n* Fixed the devstack plugin for Rocky Linux hosts\n\nChange-Id: I41f7e2341332b9cb74b4a59fedb6eed1af3c8062\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4e6e8f0c0e6db892f5f60437541531f84aabc7cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4e6e8f0c0e6db892f5f60437541531f84aabc7cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
