)]}'
{"id":"openstack%2Foctavia~873947","triplet_id":"openstack%2Foctavia~master~Idde359c55dbc87709e8891ab09d83deab049aff5","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2023-02-15 17:04:57.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-02-15 19:23:10.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idde359c55dbc87709e8891ab09d83deab049aff5","subject":"Fix unit tests for python-barbicanclient 5.5.0","status":"ABANDONED","created":"2023-02-15 16:48:52.000000000","updated":"2023-02-15 19:23:10.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0032a396a979b4a41e571097bd64ba003739514d","_number":873947,"virtual_id_number":873947,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-02-15 17:08:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-02-15 16:50:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-15 16:50:01.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-02-15 17:08:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a0276fe97f9b784377de28152c3d7bbb267e67e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-02-15 16:48:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac17515a2838030efb569687ac58936c60d41cb2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-15 16:50:01.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3530475b010b3b0084500491c54e5eaf721498d9","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-02-15 17:04:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a0f9fc8e305ddd9c642a25747aa2bea5f375d1ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 17:08:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48079a725e2845e69d2d254fc391909c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecf3cb7278d74052bf947c7ca8ca1c53 : SUCCESS in 9m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6546fb9a083436ea9efdc43ee984c6c : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/859f7780d32640b287fbbcb672d7fdfc : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0883d726fb7549f3bc0eacf287a69669 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09dec698a71f4e1e9edeef9f7ae6fc80 : SUCCESS in 12m 05s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ed1dd450c576441b98f42c3300fb8a6b : SUCCESS in 5m 06s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/48eb5f6ee4394bceb690e2fe13e6e3c2 : SUCCESS in 9m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b59b3a433ee7527cb0727cf28f8922691d895073","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-15 17:41:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0032a396a979b4a41e571097bd64ba003739514d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-02-15 19:23:10.000000000","message":"Abandoned\n\nhttps://review.opendev.org/c/openstack/octavia/+/873970 fixes the issue","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"317b1788706a8de5d39833c01cb0c2bcca1b1a34","revisions":{"317b1788706a8de5d39833c01cb0c2bcca1b1a34":{"kind":"REWORK","_number":1,"created":"2023-02-15 16:48:52.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/47/873947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/873947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/873947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/873947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/873947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/873947/1"}}},"commit":{"parents":[{"commit":"a2af707089b2b7c3e7597d8da55bd8ff4e0a01cd","subject":"Merge \"devstack: use iniset_rpc_backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a2af707089b2b7c3e7597d8da55bd8ff4e0a01cd"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2023-02-15 16:45:33.000000000","tz":60},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2023-02-15 16:48:50.000000000","tz":60},"subject":"Fix unit tests for python-barbicanclient 5.5.0","message":"Fix unit tests for python-barbicanclient 5.5.0\n\nVersion 5.5.0 of python-barbicanclient adds microversion support,\nwhich causes it to call keystoneauth1.discover.get_version_data\nto get the microversion information from the server. Adding\nadditional mock.patch for keystoneauth1.discover fixes the\nbreaking unit tests.\n\nChange-Id: Idde359c55dbc87709e8891ab09d83deab049aff5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/317b1788706a8de5d39833c01cb0c2bcca1b1a34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/317b1788706a8de5d39833c01cb0c2bcca1b1a34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
