)]}'
{"id":"openstack%2Foctavia~874193","triplet_id":"openstack%2Foctavia~master~Ie7bf7c1ecdbbf2c06fcfbbd65f59e6237be45a70","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ie7bf7c1ecdbbf2c06fcfbbd65f59e6237be45a70","subject":"DNM/WIP Debugging rocky linux nft/iptables","status":"ABANDONED","created":"2023-02-17 06:22:23.000000000","updated":"2023-05-16 11:49:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fc5f5feeecd505a58548fcf3fe400f92269104ad","_number":874193,"virtual_id_number":874193,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-15 19:18:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-17 06:33:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6efe345a0f2d182d9523c75910cd8f8bf715b9f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-17 06:22:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95b7f365795e96989a95d1651948981a798d60cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 06:33:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/592443f0147b4a2da466c6d54d7ff142\n\n- octavia-amphora-image-build-live-rocky-9 https://zuul.opendev.org/t/openstack/build/97413038db3446e2bb0e44251b6b184c : SUCCESS in 9m 41s","accounts_in_message":[],"_revision_number":1},{"id":"9fcf154a631e74568c75eb229d71422c019bce3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-15 09:42:05.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"97f510258ac0c4e47f2b33135706cb27d23c49ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 09:54:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98278c211b23495abd86b0ecfb597d99\n\n- octavia-amphora-image-build-live-rocky-9 https://zuul.opendev.org/t/openstack/build/398ff04ae36d40f284e57930d3dc21ae : SUCCESS in 10m 59s","accounts_in_message":[],"_revision_number":2},{"id":"286ac55b840301b1b3cce2b823a3811ba53b999c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-15 15:05:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"b35c0a47adffef915acd95c24f1c9064a1b9e6eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 15:15:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14552e924c424bdb94787233270539bd\n\n- octavia-amphora-image-build-live-rocky-9 https://zuul.opendev.org/t/openstack/build/375d20ba549f4a4491ee7398c0c2cb3a : SUCCESS in 8m 51s","accounts_in_message":[],"_revision_number":3},{"id":"c37a6eaf150ecb9657f984cb9c7d3ed5f2b7e56c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-15 19:03:10.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"150fb04972692ce7931a135ac8811c8790e4850e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 19:18:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba35e843e0134c588bdca9c875f83480\n\n- octavia-amphora-image-build-live-rocky-9 https://zuul.opendev.org/t/openstack/build/4f117265cc894ba882c69c6b64c99e81 : SUCCESS in 10m 09s","accounts_in_message":[],"_revision_number":4},{"id":"fc5f5feeecd505a58548fcf3fe400f92269104ad","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-16 11:49:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2908fc264a6d7efc7ab91a3658debe180f2e4fce","revisions":{"520443ebc028480e107acbe8c929bcdf45b10711":{"kind":"REWORK","_number":1,"created":"2023-02-17 06:22:23.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/93/874193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/874193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/874193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/874193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/874193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/874193/1"}}},"commit":{"parents":[{"commit":"0e8f64def39e47aa727517b6445b381d0f63ec2f","subject":"WIP Add support for Rocky Linux amphora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e8f64def39e47aa727517b6445b381d0f63ec2f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-17 06:04:30.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-17 06:04:30.000000000","tz":60},"subject":"DNM/WIP Debugging rocky linux nft/iptables","message":"DNM/WIP Debugging rocky linux nft/iptables\n\nChange-Id: Ie7bf7c1ecdbbf2c06fcfbbd65f59e6237be45a70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/520443ebc028480e107acbe8c929bcdf45b10711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/520443ebc028480e107acbe8c929bcdf45b10711"}]},"branch":"refs/heads/master"},"4c33c35e552d7b14c7b20c778ca3fc9fc06f8b7f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-05-15 09:42:05.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/93/874193/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/874193/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/874193/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/874193/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/874193/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/874193/2"}}},"commit":{"parents":[{"commit":"6f6472c09b2106231d7e85f2f19d6adb2d5a02c4","subject":"WIP Add support for Rocky Linux amphora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f6472c09b2106231d7e85f2f19d6adb2d5a02c4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-17 06:04:30.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-15 09:42:05.000000000","tz":0},"subject":"DNM/WIP Debugging rocky linux nft/iptables","message":"DNM/WIP Debugging rocky linux nft/iptables\n\nChange-Id: Ie7bf7c1ecdbbf2c06fcfbbd65f59e6237be45a70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c33c35e552d7b14c7b20c778ca3fc9fc06f8b7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c33c35e552d7b14c7b20c778ca3fc9fc06f8b7f"}]},"branch":"refs/heads/master","description":"Rebase"},"61574acd6a795d0e0094088b12689efe459a8357":{"kind":"REWORK","_number":3,"created":"2023-05-15 15:05:04.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/93/874193/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/874193/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/874193/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/874193/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/874193/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/874193/3"}}},"commit":{"parents":[{"commit":"6f6472c09b2106231d7e85f2f19d6adb2d5a02c4","subject":"WIP Add support for Rocky Linux amphora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f6472c09b2106231d7e85f2f19d6adb2d5a02c4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-17 06:04:30.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-15 15:04:53.000000000","tz":120},"subject":"DNM/WIP Debugging rocky linux nft/iptables","message":"DNM/WIP Debugging rocky linux nft/iptables\n\nChange-Id: Ie7bf7c1ecdbbf2c06fcfbbd65f59e6237be45a70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61574acd6a795d0e0094088b12689efe459a8357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61574acd6a795d0e0094088b12689efe459a8357"}]},"branch":"refs/heads/master"},"2908fc264a6d7efc7ab91a3658debe180f2e4fce":{"kind":"REWORK","_number":4,"created":"2023-05-15 19:03:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/93/874193/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/874193/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/874193/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/874193/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/874193/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/874193/4"}}},"commit":{"parents":[{"commit":"6f6472c09b2106231d7e85f2f19d6adb2d5a02c4","subject":"WIP Add support for Rocky Linux amphora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f6472c09b2106231d7e85f2f19d6adb2d5a02c4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-17 06:04:30.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-15 19:02:55.000000000","tz":120},"subject":"DNM/WIP Debugging rocky linux nft/iptables","message":"DNM/WIP Debugging rocky linux nft/iptables\n\nChange-Id: Ie7bf7c1ecdbbf2c06fcfbbd65f59e6237be45a70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2908fc264a6d7efc7ab91a3658debe180f2e4fce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2908fc264a6d7efc7ab91a3658debe180f2e4fce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
