)]}'
{"id":"openstack%2Foctavia~874951","triplet_id":"openstack%2Foctavia~master~I914aaa2eb57cf4c7206909e8ea17af1033c54965","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-03-07 16:24:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I914aaa2eb57cf4c7206909e8ea17af1033c54965","subject":"Fix example policy file system-reader role","status":"MERGED","created":"2023-02-23 18:24:25.000000000","updated":"2023-03-07 16:26:13.000000000","submitted":"2023-03-07 16:24:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874951","meta_rev_id":"c40b7afdc56f28036280bfa795ea93cadae9dbdf","_number":874951,"virtual_id_number":874951,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-07 16:24:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-03-07 15:29:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-07 15:32:18.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-07 15:32:18.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-03-07 15:29:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-23 21:02:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-07 15:29:20.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-03-07 15:32:18.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"792036a9c50e58b407e1aaac7f470a0589a08dec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-23 18:24:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b1c858799913da79434187efb342d9d14d60014","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 21:02:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75dca6eebb9c45788b81ee07b32527e3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7641e6e0833404ab16ccfa7e8af154d : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d07bda3183984f8ca55cc704715b5c91 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/743dabd13f3d4376859f3fd1b0fe3229 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/481b476017b84a05a7d2ea4834db64be : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/034614a0544c45bf8458c633f541510d : SUCCESS in 13m 23s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/76451f39f9d34ad18a6d11f5a55a3f67 : SUCCESS in 4m 43s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5d4284566d1649cba4ac88cbb1819810 : SUCCESS in 7m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d446796437424e5eabf6565ad2ddcd97 : SUCCESS in 1h 26m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/25a5bcea6d2c456781ef7b3e9e2b4262 : SUCCESS in 1h 54m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f454c5b13f0f4db9b21944b616ae3ef0 : SUCCESS in 38m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/995aa8acd7d84c8a895bf344bbef3f34 : SUCCESS in 40m 30s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e1fa231c164c41fe9d453ce019f641a3 : SUCCESS in 1h 09m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/946c8737d03d4155915e096bc939a8d8 : SUCCESS in 30m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d602f59933e442f2b1927408327dc967 : SUCCESS in 2h 00m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/542a511ee6da4105bed24188fe9f160b : FAILURE in 2h 10m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9348845f493e46f2be889894a60f4270 : SUCCESS in 1h 52m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/add70dd852b645469c5adac6de4bf373 : SUCCESS in 2h 00m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/db0242e9efa94f82b9933dc120b9d7bf : SUCCESS in 1h 53m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0870c0239ca7403082e16deb1c5683d5 : TIMED_OUT in 2h 34m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5fdb41c049c52b3fd96d1d58c47f40b017cb9c6e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-07 15:29:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b767e06d3e48d6470b8e7e8dddba534089d46763","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-07 15:29:30.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"0d58ba5e41e87b149a86743b5c0fa9e3a149ef17","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-07 15:32:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fb24a3d5a0a5a5e5d37c9cb239096f73ca52df63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 15:32:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b54d75e8cab596322e328876715839226915d68b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 16:24:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f6633aa580d46ba8d0fbadd070a66a8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/babec5129759417c93daa9bb891bb3b0 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2dcdc40ba8d4c62a1706d9967f78cb1 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f76e3819a1124804885df44251593b58 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d288d82285848c2bc6e0fb40db151d7 : SUCCESS in 11m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7dfec8548c344971a6a1f2b94264ab16 : SUCCESS in 49m 19s","accounts_in_message":[],"_revision_number":1},{"id":"2d5ad225c8e3c84c6d00225cf1554a262df1d08d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 16:24:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c40b7afdc56f28036280bfa795ea93cadae9dbdf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 16:26:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08eb3f52e5a54d86aa99534ef25cf8f1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83ef96408007429f951b4018be4daf57 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32988c0ae82ac8b52a95d5bf5b2f087d12230642","revisions":{"32988c0ae82ac8b52a95d5bf5b2f087d12230642":{"kind":"REWORK","_number":1,"created":"2023-02-23 18:24:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/51/874951/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/874951/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/874951/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/874951/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/874951/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/874951/1"}}},"commit":{"parents":[{"commit":"b4b6e07fe07f2b1103ebe733db3f76f05d7acf0c","subject":"Merge \"Fix typo in lvs-masquerade.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b4b6e07fe07f2b1103ebe733db3f76f05d7acf0c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-02-23 18:22:52.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-02-23 18:24:23.000000000","tz":0},"subject":"Fix example policy file system-reader role","message":"Fix example policy file system-reader role\n\nThis patch fixes the sample keystone_default_roles-policy.yaml file to use admin instead of reader for the system-reader role.\n\nChange-Id: I914aaa2eb57cf4c7206909e8ea17af1033c54965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/32988c0ae82ac8b52a95d5bf5b2f087d12230642"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/32988c0ae82ac8b52a95d5bf5b2f087d12230642"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[]}
