)]}'
{"id":"openstack%2Foctavia~875367","triplet_id":"openstack%2Foctavia~master~I1a9a9177cafb8c52f13a646b12cff1c35afd8679","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-03-02 08:37:45.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-03-01 16:17:21.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I1a9a9177cafb8c52f13a646b12cff1c35afd8679","subject":"Fix SQLAlchemy warning about conflict relationship with Tags","status":"MERGED","created":"2023-02-27 08:05:15.000000000","updated":"2023-03-02 13:02:03.000000000","submitted":"2023-03-02 13:00:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875367","meta_rev_id":"e4dc8a3cbb19cade1c684f3df54c7b2eb44b33d0","_number":875367,"virtual_id_number":875367,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-02 13:00:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-01 16:17:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-03-01 12:57:11.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-01 16:17:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-27 11:04:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-27 16:17:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-03-01 12:57:11.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"7e55b927885a385585fc2b4c14bee46c372c351a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-27 08:05:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60f2646effb3a528129487d5447cf60330e0bbc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 11:04:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c84b3c8cbd2a4f48b5b5abbf1599ae5a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d99ecfde0bbd43f0ab4a00873ffd49fd : SUCCESS in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3737c37214b4f6a99f95be64e8e0ff1 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adf8744c32ca4204be2954b5ee53f0f0 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97c1a9a5cfb540a6aa0cc15af166e9f2 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa5b9d5cab2c46188256e993005c5fa6 : SUCCESS in 10m 55s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c62e47a8f8344334803a4b30c5ab872d : SUCCESS in 6m 00s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/99c56629f5394a23bd98202f430cf6f3 : SUCCESS in 8m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fef9994db3484b57a224c99cb4d7fd87 : RETRY_LIMIT in 3m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/408775374a784da4970025c66fad42fb : SUCCESS in 6m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9e03cffc58654e3b91e244d021cddb5e : SUCCESS in 6m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/54768d6df9df4dd1a04ff01858a95f23 : SUCCESS in 7m 18s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d19020d0851141908a18cd1b624663ef : FAILURE in 5m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/29ef63bea899417f9f1a1864fa6c64e1 : SUCCESS in 1h 43m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f595d1c7c6a9445090dc7121654cf058 : SUCCESS in 38m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b6129d2edf7e44809313b7333744b51f : SUCCESS in 49m 05s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ea60a64bea794e978c5c215d09d404ca : SUCCESS in 1h 13m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a76f366792b4700bb339bfcc7945a1c : SUCCESS in 33m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/852488ad9ed749dc97a4ce78e4e651e1 : SUCCESS in 2h 32m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5e0771fe2c554522997c7c7f5e7bdd0a : SUCCESS in 2h 55m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3ad31caa1bb34cddb6b22b87d07ce90d : SUCCESS in 2h 19m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f45119f876454af59ea1bb96ad049ecd : TIMED_OUT in 2h 33m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2340da30a2554decb7e0c1560e081644 : SUCCESS in 2h 02m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a12cc5b7f1a467ebdd6c1dc513c31cd : SUCCESS in 2h 19m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96177379debdf4bb3f71f6e4a82e1637f05134c1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-02-27 16:17:53.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"04c5a16e26b0c9725c42784cce4b9140c63f01b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-27 16:23:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"225b7f1413208ebbbc6ebbed8ba3642ee49fe58c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-27 16:24:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6db24882f540016b52342dce5680184feba288df","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-01 12:57:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e92f4f548b9ff4a101e21e0334f0ec6c7bf80131","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-01 16:17:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"768487dfd03a32b18fbc91dc277e232b4f5ca6a6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-02 08:37:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"58d6d7907c497960ef7346fabc6c90cdd66dcd34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 11:08:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/757ef87b9aef4d24bcd380c26e05eb06\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/227fedc372e34776b6002cc692d74117 : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/995b17b588784aa19bc7455ade2f17c2 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fcfd04a74de4a1f8f531e47b61dab4d : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8456005c5ac642349966667cc268aae2 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e5532bb85fa4e229bd727d9ae295024 : SUCCESS in 10m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7820c301f550457fb2862c3a34f5cef4 : SUCCESS in 4m 26s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9fd238f67b434d37814b56a27c0fc7b5 : SUCCESS in 4m 56s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5b0c68376096490694b1d2f7a1078161 : SUCCESS in 6m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/29c04c5e7f1e435dbdefd9bcb201588b : SUCCESS in 4m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/63a9d516cfe14545b44db12829cea54b : SUCCESS in 6m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fe772b584e7f4d3da302a26cdb1d0888 : SUCCESS in 11m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d6eb09f0a47d412b9fcb0b260363a581 : SUCCESS in 6m 46s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/83520fc428404beb91589044b63de5d8 : SUCCESS in 1h 57m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2da5521faf5043caadb4d689da629c35 : SUCCESS in 1h 43m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2ce22f8604cb46b89561b1d846a0733d : SUCCESS in 33m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0c5bb5d52c6a4bf89449ef9415412614 : SUCCESS in 37m 37s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/33cdcf2c43f0458b913e07ee42c2b3b0 : SUCCESS in 1h 10m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ef2488c882764e92b4c86a151760903e : SUCCESS in 31m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/68cb5aff29964554a24d39b32e5cf573 : SUCCESS in 1h 48m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/700743fb9a4f4257a4b422d899d5a652 : SUCCESS in 2h 03m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bf2819a9eda8447ab9a67554fdbe63c1 : SUCCESS in 1h 44m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b5a53d0e2cda43389585f1c4b2e9e913 : SUCCESS in 2h 05m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2bb5a6c1493248aca6e299e0e13b730e : SUCCESS in 2h 03m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/73911dbc5bd240d6b6692e48431979f9 : SUCCESS in 2h 23m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5f7a4bc04fd251e048516508e25164014a763baf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 11:09:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2ec9ac259f52f8e9c615364300de734d272021f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 13:00:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9742d3e8c81b428d93d5e911dd9ab21f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee8d4068bfa34a609aaa4141f5080857 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1370d8628df4d08a1d4c08addaa5bb9 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6581c54cf9249eb93eea5ab4f130d57 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fa07f264a3141dc8a3e9452637486f0 : SUCCESS in 15m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/963feb88f40c42cb96c57f6ca553bdc7 : SUCCESS in 4m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f7396e0e3a9142e2b678952964138e00 : SUCCESS in 4m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0e37a5144a7449acb677c70dfafcce2a : SUCCESS in 6m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d62a4373831943ea9051704ecb36e190 : SUCCESS in 1h 30m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/58c735076524474ebae28f8655ff4dbc : SUCCESS in 1h 40m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f1fdfa7576904eba83d22204daf62742 : SUCCESS in 44m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/267b34301ad34cab9610c75c6785716f : SUCCESS in 52m 14s","accounts_in_message":[],"_revision_number":2},{"id":"94bcee43cf2c48efeab47b47fa893831c07dc998","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 13:00:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e4dc8a3cbb19cade1c684f3df54c7b2eb44b33d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 13:02:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/380528259a004a13bd29c45a9ee4e1ee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3621afa5eeb4638b90614a49f7b6b76 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ef5839451fe481e88127b193c95f08a : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8409c06cda0bb3e3c0795b14ec3917ca2b5afe7b","revisions":{"e30c1da4b5f9af597bf383ec1423b071a1f2840c":{"kind":"REWORK","_number":1,"created":"2023-02-27 08:05:15.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/67/875367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/875367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/875367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/875367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/875367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/875367/1"}}},"commit":{"parents":[{"commit":"b4b6e07fe07f2b1103ebe733db3f76f05d7acf0c","subject":"Merge \"Fix typo in lvs-masquerade.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b4b6e07fe07f2b1103ebe733db3f76f05d7acf0c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-23 16:19:38.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-27 07:40:26.000000000","tz":-300},"subject":"Fix SQLAlchemy warning about conflict relationship with Tags","message":"Fix SQLAlchemy warning about conflict relationship with Tags\n\nSQLAlchemy shows a warning about a conflict with a relationship with the\nTags object:\n\n  SAWarning: relationship \u0027\u003cObj1\u003e._tags\u0027 will copy column \u003cobj1\u003e.id to\n  column tags.resource_id, which conflicts with relationship(s):\n  \u0027\u003cObj2\u003e._tags\u0027\n\nRemove this warning by using the overlaps [0] keyword in the definition\nof the relationship.\n\n[0] https://docs.sqlalchemy.org/en/20/orm/relationship_api.html#\\\n    sqlalchemy.orm.relationship.params.overlaps\n\nChange-Id: I1a9a9177cafb8c52f13a646b12cff1c35afd8679\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e30c1da4b5f9af597bf383ec1423b071a1f2840c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e30c1da4b5f9af597bf383ec1423b071a1f2840c"}]},"branch":"refs/heads/master"},"8409c06cda0bb3e3c0795b14ec3917ca2b5afe7b":{"kind":"REWORK","_number":2,"created":"2023-02-27 16:23:55.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/67/875367/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/875367/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/875367/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/875367/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/875367/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/875367/2"}}},"commit":{"parents":[{"commit":"b4b6e07fe07f2b1103ebe733db3f76f05d7acf0c","subject":"Merge \"Fix typo in lvs-masquerade.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b4b6e07fe07f2b1103ebe733db3f76f05d7acf0c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-23 16:19:38.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-27 16:23:37.000000000","tz":60},"subject":"Fix SQLAlchemy warning about conflict relationship with Tags","message":"Fix SQLAlchemy warning about conflict relationship with Tags\n\nSQLAlchemy shows a warning about a conflict with a relationship with the\nTags object:\n\n  SAWarning: relationship \u0027\u003cObj1\u003e._tags\u0027 will copy column \u003cobj1\u003e.id to\n  column tags.resource_id, which conflicts with relationship(s):\n  \u0027\u003cObj2\u003e._tags\u0027\n\nRemove this warning by using the overlaps [0] keyword in the definition\nof the relationship.\n\n[0] https://docs.sqlalchemy.org/en/20/orm/relationship_api.html#\\\n    sqlalchemy.orm.relationship.params.overlaps\n\nChange-Id: I1a9a9177cafb8c52f13a646b12cff1c35afd8679\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8409c06cda0bb3e3c0795b14ec3917ca2b5afe7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8409c06cda0bb3e3c0795b14ec3917ca2b5afe7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
