)]}'
{"id":"openstack%2Foctavia~875382","triplet_id":"openstack%2Foctavia~master~Icac1764e818b9a600bc939cc5883c64f5adaed36","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-07-31 17:30:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icac1764e818b9a600bc939cc5883c64f5adaed36","subject":"Fixed VRRP nopreempt option","status":"MERGED","created":"2023-02-27 10:49:52.000000000","updated":"2024-07-31 17:32:00.000000000","submitted":"2024-07-31 17:30:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875382","meta_rev_id":"7ff6469a8199951b6d82119c89757ad081d37e40","_number":875382,"virtual_id_number":875382,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-31 17:30:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2024-06-13 17:16:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2024-07-03 22:00:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-31 15:30:30.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-31 15:30:30.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-27 13:20:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-13 17:16:18.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-07-03 22:00:57.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-07-31 15:30:30.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"722eb40a733299e8b2cb729ffddae1e3c4ba6578","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-27 10:49:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aed47985b910d0e6954d5ea44e40ba415f39c339","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 13:20:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ba08c8366454076a49dffae9c46c1f5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba842e7e7eab45418745672ac3bac82b : RETRY_LIMIT in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b49ad37a180f4fcfb4289b6c2e914746 : RETRY_LIMIT in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6f1d72781154b359b8d27777c99e2cf : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b43b7f50f57c417aa763f2798871f1c4 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42f4b37394b04b8d935eb733cd11a6ee : SUCCESS in 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6d8be87b98a4e6e92b85ea340ce1619 : SUCCESS in 5m 15s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/fb07a3f3fb7a49c2935113debbbb55eb : SUCCESS in 5m 21s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7d9dae8bf54e4a7eb467126d47f11f24 : SUCCESS in 7m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ca399b165b784817b38a6f8bb0e7991d : SUCCESS in 4m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3a191135858e48c9bffcc3cc124933ba : SUCCESS in 8m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c70a4a949ee34c9cb3d671b766b82d1e : SUCCESS in 8m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/afe8631f3adf46eb81e56a6f360a4875 : SUCCESS in 6m 49s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/82e96e08f1a346f5837a5120afb6fe25 : SUCCESS in 1h 34m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75f2d11984a44af1a0921d7264403c85 : SUCCESS in 1h 48m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f391536d5d0e467fa9bf0ed5f46b88af : SUCCESS in 41m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/61bbc166ba0d49a5b43b387b03c02d0a : SUCCESS in 37m 45s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5987d98a647a431b971d8214327bbb9a : SUCCESS in 1h 13m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e6a3fa88f3bf451abb212b2a69020478 : SUCCESS in 32m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3442e6060e6442c09eb5155860fb3404 : SUCCESS in 1h 58m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8374f4f238b944a79a7ee6794e808e6a : SUCCESS in 2h 27m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1caa70765fad4d5cbe448ed1af818256 : SUCCESS in 1h 57m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c14bcd99ab1b427f9f297b523dd5718d : SUCCESS in 2h 14m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9471e79ad9554408b0b5262a5d9a0271 : SUCCESS in 1h 58m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2583eaebdbe4e2da2b57a6eb7ad73b3 : SUCCESS in 2h 26m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a13214c0f9d3bd3c9bb62bfd1f54fc33eeb508b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-31 05:27:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9227e8d473b5a50ee056e467c26f005810bd0267","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 08:10:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/200ca60413cf4b12b82d0b5adcfea5df\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bde7aee2902b4fe2a1a63fd53eabb9d8 : SUCCESS in 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/161239b1c1ef4c3089badc17b88ab982 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de310560144d46e8afd447d29f82e8d3 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1127c6ac49094935a15a52e3722ee76c : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91a066188c0c45dba77a8ab152583adc : SUCCESS in 12m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8ddd90e08f343ab9c771f5b0a9351c7 : SUCCESS in 6m 32s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/2e1c48514ab04b45b8c5d89acfd6ce82 : SUCCESS in 4m 22s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/6bbf4be3610d4f3bbfd3011892d79601 : SUCCESS in 8m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/29c6f4e8920a4d3db69ed9dfaaffcaa3 : SUCCESS in 4m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/18857cb0e83b4d91806109a457af1193 : SUCCESS in 8m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/602bb6e76d9048af905d9d110ce8cf50 : SUCCESS in 7m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/03c74a0d747f43eaa8d50c884f3639be : SUCCESS in 8m 21s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ca7303c7b15947ef91865c867c4519ec : SUCCESS in 1h 17m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d1c117018ddf4560b656e2e94c12e49e : SUCCESS in 1h 56m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1aa8d435e3514e739082389020cdddb7 : SUCCESS in 38m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/294b27eb48fd404b9632ebb6c00db92c : SUCCESS in 43m 13s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f2cc264581754f1595d92d0e8b769b3e : SUCCESS in 58m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3ea3a245d94e444096b1b17a2d63a48b : SUCCESS in 36m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d57fd6808004460db8b34258f4debe68 : SUCCESS in 2h 14m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c5cc3d1024f64c119d1a6e7ea86f3fa2 : SUCCESS in 2h 21m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f83478f3a21b4200900a754c3b8e1177 : SUCCESS in 2h 11m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f9ed0377a85b49e6af6b5beedddc721c : SUCCESS in 2h 10m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c0b903481bfb4145a0c5b9d79d2050c2 : SUCCESS in 1h 58m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/df7286c874ea41f0bd5ff7e3684dc3fe : TIMED_OUT in 2h 34m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ecde72c58b0494b3856c1a3e4c642592f176025","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-06-13 17:16:18.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a10bf458a3960d96f24fadaea118cd14bd20bf7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-07-03 22:00:57.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f16fedff547886cbd9cb14cb8cfc55c6f6ea4331","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-07-31 15:30:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b25a6fdf6ba89b393d78ec70bd23ad21ebd00501","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 15:31:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4235281a9e61150cb3f4238fa83e5048c14a6594","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 17:30:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58d4ef8a3629410ead04942823c75cb1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0b51dfa72d8415787cb683441f208e5 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fcf11c5af77f45239935b868a0d45080 : SUCCESS in 5m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ed72ab3e54c4aac89607cd75a4d7c27 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22c26bf3ea3b4886bba2e64a91398cdf : SUCCESS in 11m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c731b5d14fe4a538d020720f18daa26 : SUCCESS in 4m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/398d36a651b849b8b87746ed6082a73c : SUCCESS in 4m 28s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f3fc658323324704930b9a43e4e10630 : SUCCESS in 7m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/14d1b10bdf0c41beae8b1ebdcc699524 : SUCCESS in 1h 49m 32s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/503b767f13bb4a2e88c98d9778b66c8b : SUCCESS in 1h 56m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/94039c70ec714587ae638a2bc77163bd : SUCCESS in 1h 17m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ba7fcc1057c64b7baa85f7f24f14e929 : SUCCESS in 42m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f607deef27c84fdaae20ea9066f80662 : SUCCESS in 1h 03m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b7e76c3269beea424eb624d233d3cbb98442633d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 17:30:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7ff6469a8199951b6d82119c89757ad081d37e40","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 17:32:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc0cd569840c44bab40e4e3b0772eb5d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26f2e8cb8a784009bf370840e2fefe02 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4e307c0e5c24ef1915cc3a484e7e811 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"911faf26239840fd654c741a673c8bd657ba2c46","revisions":{"911faf26239840fd654c741a673c8bd657ba2c46":{"kind":"REWORK","_number":1,"created":"2023-02-27 10:49:52.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/82/875382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/875382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/875382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/875382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/875382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/875382/1"}}},"commit":{"parents":[{"commit":"b4b6e07fe07f2b1103ebe733db3f76f05d7acf0c","subject":"Merge \"Fix typo in lvs-masquerade.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b4b6e07fe07f2b1103ebe733db3f76f05d7acf0c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-27 10:26:17.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-27 10:27:32.000000000","tz":-300},"subject":"Fixed VRRP nopreempt option","message":"Fixed VRRP nopreempt option\n\nThe nopreempt option from keepalived didn\u0027t work as expected because the\nrole of the MASTER amphora was set in the configuration file [0].\nRemoving the default roles fixes this issue.\nNow the newly created amphora after a failover doesn\u0027t preempt the\nMASTER role from the other amphora.\n\n[0] https://github.com/acassen/keepalived/blob/master/doc/man/man5/\\\n        keepalived.conf.5.in#L2011-L2017\n\nStory 2010595\nTask 47457\n\nChange-Id: Icac1764e818b9a600bc939cc5883c64f5adaed36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/911faf26239840fd654c741a673c8bd657ba2c46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/911faf26239840fd654c741a673c8bd657ba2c46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
