)]}'
{"id":"openstack%2Foctavia~875532","triplet_id":"openstack%2Foctavia~master~Iec80510ed6977c2cff3c6ec8fc5948c9fc44e410","project":"openstack/octavia","branch":"master","topic":"sqlalchemy2","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-06-06 10:03:47.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iec80510ed6977c2cff3c6ec8fc5948c9fc44e410","subject":"WIP Adding a functional job with SQLAlchemy master","status":"ABANDONED","created":"2023-02-27 16:35:25.000000000","updated":"2024-06-06 10:03:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b817e27dc581add91d5f2ed8ed4f8999b52ce0e3","_number":875532,"virtual_id_number":875532,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-05-03 07:18:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-27 16:36:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0c009427f433a45ac3c013220f60709a9e1a45df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-27 16:35:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"440102b5a949cba5d5c251271f4768bf20baa0ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 16:36:22.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch master.  The error was:\n\n  Job openstack-tox-functional-sqlalchemy-master not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates: []\n      check:\n        jobs:\n          - openstack-tox-functional-sqlalchemy-master\n      queue: octavia\n      gate:\n        fail-fast: true\n        jobs:\n          - openstack-tox-pip-check-reqs:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^octavia/tests/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py38:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-scenario\n          - octavia-v2-dsvm-tls-barbican\n          - octavia-grenade\n      periodic:\n        jobs:\n          - publish-openstack-octavia-amphora-image-focal:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-centos-9-stream:\n              branches: ^(?!stable/.*).*$\n          - octavia-amphora-image-build\n          - octavia-grenade-ffu\n  # Putting octavia-v2-dsvm-scenario-fips in periodic as centos 8 is too slow\n          - octavia-v2-dsvm-scenario-fips:\n              voting: false\n      experimental:\n        jobs:\n          - octavia-v2-dsvm-scenario-nftables\n          - octavia-v2-dsvm-noop-api-amphora-v2\n          - octavia-v2-dsvm-tls-barbican-amphora-v2\n          - octavia-v2-act-stdby-dsvm-scenario-amphora-v2\n          - octavia-grenade-amphora-v2\n          - octavia-v2-dsvm-cinder-amphora-v2\n          - octavia-v2-dsvm-scenario-two-node-amphora-v2\n\n  in \"openstack/octavia/zuul.d/projects.yaml@master\", line 3, column 3","accounts_in_message":[],"_revision_number":1},{"id":"af7d150f2e56deae0af9b741f7995abae5382814","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-27 16:36:47.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"43eb4754285040f28f7b8050564bc1b2073af9fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 16:37:46.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/octavia on branch master.  The error was:\n\n  Job openstack-tox-functional-sqlalchemy-master not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates: []\n      check:\n        jobs:\n          - openstack-tox-functional-sqlalchemy-master\n      queue: octavia\n      gate:\n        fail-fast: true\n        jobs:\n          - openstack-tox-pip-check-reqs:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^octavia/tests/.*$\n                - ^releasenotes/.*$\n          - openstack-tox-functional-py38:\n              irrelevant-files:\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^doc/.*$\n                - ^etc/.*$\n                - ^octavia/tests/unit/.*$\n                - ^releasenotes/.*$\n          - octavia-v2-dsvm-noop-api\n          - octavia-v2-dsvm-scenario\n          - octavia-v2-dsvm-tls-barbican\n          - octavia-grenade\n      periodic:\n        jobs:\n          - publish-openstack-octavia-amphora-image-focal:\n              branches: ^(?!stable/.*).*$\n          - publish-openstack-octavia-amphora-image-centos-9-stream:\n              branches: ^(?!stable/.*).*$\n          - octavia-amphora-image-build\n          - octavia-grenade-ffu\n  # Putting octavia-v2-dsvm-scenario-fips in periodic as centos 8 is too slow\n          - octavia-v2-dsvm-scenario-fips:\n              voting: false\n      experimental:\n        jobs:\n          - octavia-v2-dsvm-scenario-nftables\n          - octavia-v2-dsvm-noop-api-amphora-v2\n          - octavia-v2-dsvm-tls-barbican-amphora-v2\n          - octavia-v2-act-stdby-dsvm-scenario-amphora-v2\n          - octavia-grenade-amphora-v2\n          - octavia-v2-dsvm-cinder-amphora-v2\n          - octavia-v2-dsvm-scenario-two-node-amphora-v2\n\n  in \"openstack/octavia/zuul.d/projects.yaml@master\", line 3, column 3","accounts_in_message":[],"_revision_number":2},{"id":"dd9253addb54c02f3b49a29edfecc0ec4ed2bc0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-27 16:39:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ee7181232eeed8bcf974a0d3ed28427bb847d021","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 16:47:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8a58adc5d264ba99cdc87fbc8e78a32\n\n- octavia-tox-functional-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a69c3281f8de4595a51168a3e5315a55 : FAILURE in 6m 25s","accounts_in_message":[],"_revision_number":3},{"id":"8002fa612670f40d1829bbbb5fbf584d46e2b9f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-02-28 09:18:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b443bfe8dd9115b912ee716f266a35db6e2db343","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 09:30:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bc32fa1e81349dfb732cf35a5a79354\n\n- octavia-tox-functional-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9193a55fba6e4e8bb160617e86ce098f : FAILURE in 5m 25s","accounts_in_message":[],"_revision_number":4},{"id":"a205ad43dc171ff249d351ad1a669880981f9cc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-01 12:33:46.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"355d22394d03ffbeb73507fa3c4e68c949289763","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 12:51:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6e96aa8e19b450aaf427e2e9d26527c\n\n- octavia-tox-functional-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/24347bdbef58411e959dfe0d3cc20521 : FAILURE in 6m 16s","accounts_in_message":[],"_revision_number":5},{"id":"c394553e5cd22de11ac899dd22a4af0575491bc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-29 08:44:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8a3d877159c46ea563e90487ffc310e537309980","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 09:02:38.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/51998e03f609488e91f5bf214238a4d7\n\n- octavia-tox-functional-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/fa60e239d7524d41b983d148f1be580a : FAILURE in 5m 44s","accounts_in_message":[],"_revision_number":6},{"id":"1326f74b59976b210fe00ff4f40feed37a86a50b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-29 12:50:13.000000000","message":"Uploaded patch set 7: New patch set was added with same tree, parent tree, and commit message as Patch Set 6.","accounts_in_message":[],"_revision_number":7},{"id":"6629c9c1f07bcfe171986e8d7396bdfe227b5f86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 13:05:58.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2fc0c51c2144ce99ddb2358e86b379c\n\n- octavia-tox-functional-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/12210967bed14aef8dce091a746b1536 : FAILURE in 6m 34s","accounts_in_message":[],"_revision_number":7},{"id":"c4c9d2bbf280add712f6b3f4e6b45805b13cf035","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-24 06:50:24.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"b4392bac3d8a0c47e0f1c4c897190d444c85aee7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 07:02:07.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8a6e3d9f855488c94a47016c26f45a3\n\n- octavia-tox-functional-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/b19a576bc1ff48a28997a94e1a04e323 : FAILURE in 5m 36s","accounts_in_message":[],"_revision_number":8},{"id":"4049b1ac2f01627ad48349ad53e0730b3a279542","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-03 07:07:19.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"6da4b87965509e26e34c83b69c0a52de469b29ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 07:18:54.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/442bdbda70f04e44ba801775f6819dc1\n\n- octavia-tox-functional-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/251b5e53fba4488cbda5a4beef790bdb : FAILURE in 5m 18s","accounts_in_message":[],"_revision_number":9},{"id":"b817e27dc581add91d5f2ed8ed4f8999b52ce0e3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-06-06 10:03:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"11f74f4727fefba416701066b173c65334c97e63","revisions":{"4a6e0d656b9570bdfd21b5bc634990bc8953838f":{"kind":"REWORK","_number":1,"created":"2023-02-27 16:35:25.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/32/875532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/875532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/875532/1"}}},"commit":{"parents":[{"commit":"8e5614d087b37d54dd11d68ed649f8e8bbeb1eec","subject":"[sqlalchemy2] Disabling \u0027create_engine.convert_unicode\u0027 filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8e5614d087b37d54dd11d68ed649f8e8bbeb1eec"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-27 16:34:45.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-27 16:34:45.000000000","tz":60},"subject":"WIP Adding a functional job with SQLAlchemy master","message":"WIP Adding a functional job with SQLAlchemy master\n\nChange-Id: Iec80510ed6977c2cff3c6ec8fc5948c9fc44e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a6e0d656b9570bdfd21b5bc634990bc8953838f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a6e0d656b9570bdfd21b5bc634990bc8953838f"}]},"branch":"refs/heads/master"},"cfccbf0da7c9ebe2cf143097597f155e98c497ee":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-27 16:36:47.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/32/875532/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/875532/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/875532/2"}}},"commit":{"parents":[{"commit":"8e5614d087b37d54dd11d68ed649f8e8bbeb1eec","subject":"[sqlalchemy2] Disabling \u0027create_engine.convert_unicode\u0027 filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8e5614d087b37d54dd11d68ed649f8e8bbeb1eec"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-27 16:34:45.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-27 16:36:23.000000000","tz":60},"subject":"WIP Adding a functional job with SQLAlchemy master","message":"WIP Adding a functional job with SQLAlchemy master\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.db/+/874858\n\nChange-Id: Iec80510ed6977c2cff3c6ec8fc5948c9fc44e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cfccbf0da7c9ebe2cf143097597f155e98c497ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cfccbf0da7c9ebe2cf143097597f155e98c497ee"}]},"branch":"refs/heads/master"},"6d348a021534ee825acaab524d2d8ba901d85397":{"kind":"REWORK","_number":3,"created":"2023-02-27 16:39:33.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/32/875532/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/875532/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/875532/3"}}},"commit":{"parents":[{"commit":"8e5614d087b37d54dd11d68ed649f8e8bbeb1eec","subject":"[sqlalchemy2] Disabling \u0027create_engine.convert_unicode\u0027 filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8e5614d087b37d54dd11d68ed649f8e8bbeb1eec"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-27 16:34:45.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-27 16:39:20.000000000","tz":60},"subject":"WIP Adding a functional job with SQLAlchemy master","message":"WIP Adding a functional job with SQLAlchemy master\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.db/+/874858\n\nChange-Id: Iec80510ed6977c2cff3c6ec8fc5948c9fc44e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d348a021534ee825acaab524d2d8ba901d85397"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d348a021534ee825acaab524d2d8ba901d85397"}]},"branch":"refs/heads/master"},"8cb72ccf5ce5ad44c398f6d24aa9b1df4485103f":{"kind":"REWORK","_number":4,"created":"2023-02-28 09:18:08.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/32/875532/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/875532/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/875532/4"}}},"commit":{"parents":[{"commit":"8e5614d087b37d54dd11d68ed649f8e8bbeb1eec","subject":"[sqlalchemy2] Disabling \u0027create_engine.convert_unicode\u0027 filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8e5614d087b37d54dd11d68ed649f8e8bbeb1eec"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-27 16:34:45.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-28 09:17:53.000000000","tz":60},"subject":"WIP Adding a functional job with SQLAlchemy master","message":"WIP Adding a functional job with SQLAlchemy master\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.db/+/874858\n\nChange-Id: Iec80510ed6977c2cff3c6ec8fc5948c9fc44e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8cb72ccf5ce5ad44c398f6d24aa9b1df4485103f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8cb72ccf5ce5ad44c398f6d24aa9b1df4485103f"}]},"branch":"refs/heads/master"},"1ab47ee1bc548ce2f1eeea48ed2f11c660a76562":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-03-01 12:33:46.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/32/875532/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/875532/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/875532/5"}}},"commit":{"parents":[{"commit":"079d6a5c8a5628a5bffb5cf2d87f64cdf0333383","subject":"Fix sqlalchemy declarative_base import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/079d6a5c8a5628a5bffb5cf2d87f64cdf0333383"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-27 16:34:45.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 12:12:43.000000000","tz":-300},"subject":"WIP Adding a functional job with SQLAlchemy master","message":"WIP Adding a functional job with SQLAlchemy master\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.db/+/874858\n\nChange-Id: Iec80510ed6977c2cff3c6ec8fc5948c9fc44e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1ab47ee1bc548ce2f1eeea48ed2f11c660a76562"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1ab47ee1bc548ce2f1eeea48ed2f11c660a76562"}]},"branch":"refs/heads/master"},"cbf140c87861ed69d1678bb1ab1034478f01d75d":{"kind":"REWORK","_number":6,"created":"2023-03-29 08:44:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/32/875532/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/875532/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/875532/6"}}},"commit":{"parents":[{"commit":"f4077e2d39e68f185c5bb1fcdfa15e00f18a8bfd","subject":"Fix sqlalchemy declarative_base import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4077e2d39e68f185c5bb1fcdfa15e00f18a8bfd"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-27 16:34:45.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-29 08:12:57.000000000","tz":-240},"subject":"WIP Adding a functional job with SQLAlchemy master","message":"WIP Adding a functional job with SQLAlchemy master\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.db/+/874858\n\nChange-Id: Iec80510ed6977c2cff3c6ec8fc5948c9fc44e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cbf140c87861ed69d1678bb1ab1034478f01d75d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cbf140c87861ed69d1678bb1ab1034478f01d75d"}]},"branch":"refs/heads/master"},"75a7cddc85d4d5af59776c06707b60f97490402c":{"kind":"NO_CHANGE","_number":7,"created":"2023-03-29 12:50:13.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/32/875532/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/875532/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/875532/7"}}},"commit":{"parents":[{"commit":"1cb0be1902cfe3c6ff8c495dff7a39fbcbbe44aa","subject":"Fix sqlalchemy declarative_base import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1cb0be1902cfe3c6ff8c495dff7a39fbcbbe44aa"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-27 16:34:45.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-29 12:48:47.000000000","tz":-240},"subject":"WIP Adding a functional job with SQLAlchemy master","message":"WIP Adding a functional job with SQLAlchemy master\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.db/+/874858\n\nChange-Id: Iec80510ed6977c2cff3c6ec8fc5948c9fc44e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/75a7cddc85d4d5af59776c06707b60f97490402c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/75a7cddc85d4d5af59776c06707b60f97490402c"}]},"branch":"refs/heads/master"},"93e32fe2085edf24d074f76b5066ca0192645ecd":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-04-24 06:50:24.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/32/875532/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/875532/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/875532/8"}}},"commit":{"parents":[{"commit":"5a4ba1943f75574ce6e2f8cbc399e1364958f861","subject":"Fix sqlalchemy declarative_base import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a4ba1943f75574ce6e2f8cbc399e1364958f861"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-27 16:34:45.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-24 06:44:18.000000000","tz":-240},"subject":"WIP Adding a functional job with SQLAlchemy master","message":"WIP Adding a functional job with SQLAlchemy master\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.db/+/874858\n\nChange-Id: Iec80510ed6977c2cff3c6ec8fc5948c9fc44e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93e32fe2085edf24d074f76b5066ca0192645ecd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93e32fe2085edf24d074f76b5066ca0192645ecd"}]},"branch":"refs/heads/master"},"11f74f4727fefba416701066b173c65334c97e63":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-05-03 07:07:19.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/32/875532/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/875532/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/875532/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/875532/9"}}},"commit":{"parents":[{"commit":"6794cf09e92544d895ee4d9b5b0607f967f4b382","subject":"Fix pool_timeout for tests with sqlite file backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6794cf09e92544d895ee4d9b5b0607f967f4b382"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-27 16:34:45.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-03 07:03:49.000000000","tz":-240},"subject":"WIP Adding a functional job with SQLAlchemy master","message":"WIP Adding a functional job with SQLAlchemy master\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.db/+/874858\n\nChange-Id: Iec80510ed6977c2cff3c6ec8fc5948c9fc44e410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/11f74f4727fefba416701066b173c65334c97e63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/11f74f4727fefba416701066b173c65334c97e63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
