)]}'
{"id":"openstack%2Foctavia~875892","triplet_id":"openstack%2Foctavia~master~Ic8d1f38f3891ca55fcb00eb6b3b5487b94dcd72c","project":"openstack/octavia","branch":"master","topic":"sqlalchemy2","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-08-22 14:02:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic8d1f38f3891ca55fcb00eb6b3b5487b94dcd72c","subject":"Fix sqlalchemy declarative_base import","status":"MERGED","created":"2023-03-01 12:33:46.000000000","updated":"2023-08-22 14:02:39.000000000","submitted":"2023-08-22 14:02:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875892-sqlalchemy2","meta_rev_id":"c7d7553a9159d249df69f293b732bd9468907a16","_number":875892,"virtual_id_number":875892,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-22 14:02:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-08-16 20:47:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-22 12:17:44.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-22 12:17:44.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-01 15:05:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-16 20:47:42.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-08-22 12:17:44.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"09ca46db0cfdcbb98a43860011a2c21b4c3fe703","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-01 12:33:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9e9c7eeac71660d433bb4d1bb14e5879d113b37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 15:05:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2832e130c28b4a1babebc199eb648ff3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6bd6486caaf4927a7a29655cb21ec26 : FAILURE in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/115a11d1895c412d97c3aa0f2a57ff60 : FAILURE in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39a3c6c680344eb39b6e181dca19dea2 : FAILURE in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59abd0b6cfff4d5297190ce41caf12da : FAILURE in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4281fd608db24fe29ab871767d6ebeff : SUCCESS in 11m 24s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/01c707357f9c45a09ea64d99f3d8d16a : FAILURE in 4m 56s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f4d6efc647ff4ded8132a2c6e5efda22 : SUCCESS in 8m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3c60c386d7a2467290945711b3d3001d : SUCCESS in 5m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/003a7485c6ca483ba5b3040a6da1a50f : SUCCESS in 7m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7e5b575dc59e4145819415847391237b : SUCCESS in 7m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d4e0b7c4bd7141f2854da2e3bb1b920c : SUCCESS in 7m 01s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/178cb2973d2b42f89d8917fd264a22ad : SUCCESS in 1h 10m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3791b6e70a16434e948d5b4c74d3e7c5 : SUCCESS in 1h 44m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9af0d05e4a124774a5c06e9a8b1c7a44 : SUCCESS in 40m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f7d739de36e14f73b25decf54517970c : SUCCESS in 37m 42s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/873263d4529b4ec0bc81454006b6ec7a : SUCCESS in 1h 10m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2bcd13d6439c439297fd258f5a1724e7 : SUCCESS in 32m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1290eb0ff6fb4172a6952b18fbd5180a : SUCCESS in 1h 55m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5ba7495fe4c341b286c658869d2c5f5b : SUCCESS in 2h 04m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/de0cdb76bf924264b8d71601a17b8212 : SUCCESS in 1h 55m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d6fc9941009b462ba34b7eb4f01b9bef : SUCCESS in 2h 01m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/19b65c8fc7424b18a45cc22ee56130c6 : SUCCESS in 1h 33m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/692e086d8cd04d88bffc19b893fe9a59 : SUCCESS in 2h 24m 37s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py","accounts_in_message":[],"_revision_number":1},{"id":"adafd5e7fbea8bdeb149acd4486343d596545b82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-29 08:44:58.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"bf4932351dde205df57d78d7fa849537b50b4f17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 11:38:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cadd741a3a714c8b990a4e4c7f22aee8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf3666531a9a46ba85cea498bfeafb2b : FAILURE in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f955e4e3a9e4430491a432b8d37cd79a : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5b5f57c876d4b7285181cd1eba70575 : FAILURE in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02665b15584145dab748e6ed7dbcc040 : FAILURE in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c1599b03d6744d491cf4f0e3f57d16a : SUCCESS in 11m 48s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ab855e283f4d428f988c86efc88c9991 : FAILURE in 4m 57s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/13ff2fb32bfe4ac7984189feac576195 : SUCCESS in 7m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/16424577bc7d4eb3b6c0b0c33f9ef52a : SUCCESS in 5m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2264045917bd40868c4acbb521805f5a : SUCCESS in 7m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e807045abdeb42a9b6bc163adb3bec58 : SUCCESS in 7m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4104c1d8cbeb430cb107daa0a7b579c6 : SUCCESS in 8m 21s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/83fd1eb85f274f3da48d856426e775e1 : SUCCESS in 1h 18m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5d04dc32a99f4cc486646d1df7251722 : SUCCESS in 1h 34m 44s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/4bbbb641059f494c848840f4d2cda5cb : TIMED_OUT in 2h 34m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5a5289156d044773bb24f2c0dd7f3439 : SUCCESS in 43m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9ab83c6b287c48e883f6a3fcc5001a19 : SUCCESS in 36m 44s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c6f5168a44ee4e408e7c992534201b40 : SUCCESS in 1h 04m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6669dc85ea884e56bcf23280e6cabee3 : SUCCESS in 41m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a1467487005d4643a5f59f3334aca310 : SUCCESS in 2h 30m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9fc25aab66ef4e309201ba2915b6b57d : SUCCESS in 2h 41m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0522e303b7224ac2b01368413f4ffedb : SUCCESS in 2h 12m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/24b4545f322a472689a318f9cf09f4be : SUCCESS in 2h 10m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aa0ebcc1f62d43cc9253f46ac8d6c15572e77489","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-29 12:50:13.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"15bde2eafd889cc0c3dcbf4ffde41ffd20f887b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 15:43:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf30d7a676f44bc286e2cc16c0dc8cce\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f41edfe53ab24750835db6466fdb7f38 : FAILURE in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97f7e0f830e44a3fad731626824a85ac : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b78cfd9e039e4b5ea82e07875a8c6c13 : FAILURE in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a87d5e1278147b6ba30779a1594c0ba : FAILURE in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c3da8e8e3a84827bb0a01e9d3b286db : SUCCESS in 13m 03s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/13949ab050e24077865e842e8864c809 : FAILURE in 5m 20s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d85377183d4e43d6ba5bd6c005bb1a8c : SUCCESS in 8m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dc05380b352840dcb44b9df6386f106a : SUCCESS in 5m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3ce9f012cd114fbf8839196c3559bdec : SUCCESS in 8m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5342fc71ce5a47828fe83915ae5091fc : SUCCESS in 7m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/62b52475fb5741878c751fd02a8e7ce6 : SUCCESS in 7m 43s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/55eeef2fb7a64b0592a829affda886b5 : SUCCESS in 1h 25m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1793a2a768fc40719f3e109d62784da3 : SUCCESS in 1h 44m 36s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/b33152445a1f455192f28c71510ccd83 : SUCCESS in 1h 50m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5c7d29b430e94ed9a35d0aa7755eecd8 : SUCCESS in 34m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1cdcdab33b394abab7cd2c310066275d : SUCCESS in 37m 19s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e26b503859604c2d84d578215b8feb36 : SUCCESS in 1h 11m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/69940ef66c4144dc9ff76662663bdd91 : SUCCESS in 31m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e72d5efbdfbf4482a5e8a7e1f37f4d28 : SUCCESS in 2h 02m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e99875db9bce432fbba32c2cb2f55914 : SUCCESS in 2h 34m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/74ff58fc9cbf4f64a75e8b5dc79978a0 : SUCCESS in 1h 37m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c3fa9a2bad544fb93e1dda2506c1302 : TIMED_OUT in 2h 35m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5eb8355ccc31dc9586d56259b5fc0ff00000d740","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-24 06:50:24.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"e276ae7e88bd9a5b0979a1c3a081ed20bb416626","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 09:23:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/37ad57e349454d99a502fd70563c24a6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a25d95e6012041a79bc4d3674db8e899 : FAILURE in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08684559c8184838a8a29240889aea7b : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c59b3612c26e4829abba3255abbb7873 : FAILURE in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4fb03984190648d894c52c8e0fea3d62 : FAILURE in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7cdd1ead1934dfab47edfa17e4ad6bc : SUCCESS in 12m 25s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4118b79d27494e81b7738f782c83e0f2 : FAILURE in 4m 56s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ae25c878421940f4a00d1f68a9ba1221 : FAILURE in 6m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/06e91791132f418a8d711de6aab92b47 : SUCCESS in 4m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/10f589a774dd4988bfb62b47623b6aaf : FAILURE in 8m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/877041e5469e4280b481ccd78bb3f209 : FAILURE in 8m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9b9960da750145e7916a8f05a311d9cb : FAILURE in 8m 16s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9c002d5844864166a0770bf0301c17be : SUCCESS in 1h 40m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8b5aa7b16f5d4879959533e5c0b0b52f : FAILURE in 9m 54s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/4c22c410f88441f9942a96162022f57d : FAILURE in 5m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/970aa5437e004f04a5d5a77cd1b9e913 : NODE_FAILURE Node request 300-0021043611 failed in 0s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b8da17107dc14dfa8a0df901cf4eeac5 : FAILURE in 41m 30s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9325a56ed1e845ae96ae8bff5bae0276 : SUCCESS in 1h 06m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/412f9240f2824a9fbe6e97ce6542f0f9 : FAILURE in 9m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2aad12adc69845e0ad4be3db7bfa8ab8 : FAILURE in 10m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ff0d1889ad3846e1984fffd4f4ee5b1d : FAILURE in 13m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9e99a5ff006249b9af8b662754dca815 : FAILURE in 10m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/52d1f781910b4c56b624f0591e9bb81b : SUCCESS in 2h 23m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4df9e64c5c093e7df1fbfd20748d3be0bd8685c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-05-03 07:07:19.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"cb4f67e4b69b36f24ae285aa735af2cf0611b1a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 11:01:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6c7491c84fb49c2baf5dd173edd0cb5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e1ecbbdfc1f48c8a61f5b3f8691ab47 : FAILURE in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de3e76b2e18f4f8488a3ae8ceeef3064 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/701d26bc46ff4c4982f7872848f55f40 : FAILURE in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1786942a575949d2bd8651dc3bbe77f5 : FAILURE in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d23da1b7a8f44c9bb026c26cc6d9f755 : SUCCESS in 11m 56s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/449b617117844534bb28541a3fe422f4 : FAILURE in 6m 12s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/feee77e3afd048988c80132b05a3d09a : SUCCESS in 7m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/19f96458e41e40238396d6048fbd764d : SUCCESS in 4m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fb3f9e2c87e54a9b807dd5e63e4b7036 : SUCCESS in 8m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/72322e27b6574670bdff5a2846ae316b : SUCCESS in 7m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a9eba1e7235445e995746eecca2b2a6c : SUCCESS in 7m 20s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a5f5e9ae20ac4c8ea17e8a0cb180a59b : SUCCESS in 1h 32m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc6c93a176f24abd8d30028ab16df721 : SUCCESS in 2h 11m 46s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/f482fa33f52b48b9820de22e40beb152 : SUCCESS in 2h 12m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/010de5e4436947a3a76ddf8fa8c89f85 : SUCCESS in 47m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1426d6c9a96b4fa2a8e673189229066b : SUCCESS in 52m 08s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5184c4082523436fa269670681cff9ca : SUCCESS in 1h 13m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/894f5160a3cd48bd80a7d8c9c31c88f0 : SUCCESS in 43m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1c290e5062bc45498b623dfa039520f5 : TIMED_OUT in 2h 36m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4dce08ef24b74c04bb0ff12d6d95aa23 : SUCCESS in 2h 47m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fa6c1821f4a44e109fb1c3e484c624ea : SUCCESS in 2h 20m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e5597e665914bcebd43f42e7a86ab25 : TIMED_OUT in 2h 34m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6bec698779a0959789a8b603935e2aead2929848","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-28 07:24:53.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"8c834deef416734e781656c1ae1790cd9bc26040","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-28 09:33:27.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"42670ec971c5e5aed2dda329a93e9551b85e5f0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 12:39:02.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f197b794447f43c481b13ca3886802fd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/986123f44d0c40598b6693fad2e150b1 : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ebc3f17ec87460eb53c3cfa97491f05 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b6c7b5cf2a3426a909ca5e2fcd60c88 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23b4cf3b65704282bcb08079e6a58ba2 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d47858a38394c12ab6efb8cc6b36650 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d8e298279f341168b2a3379d279ee52 : SUCCESS in 13m 05s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/715849147eff4682bcfa0eb2ccbeac85 : SUCCESS in 6m 14s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fc814e58dbc944e0b4674fc0c99b7b9f : SUCCESS in 8m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/15afde3e9d4d4cd096b54f9b9c4fec2d : SUCCESS in 8m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e8c45b3715304290ad896d56e6a9f223 : SUCCESS in 6m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a4bbbab41ef242238168ee3c25feae9c : SUCCESS in 8m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d5382c3329f04252baf93cf8bd8eec68 : SUCCESS in 9m 33s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1b7881ed6b654245912c74b5d6034614 : SUCCESS in 1h 26m 10s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7110f1d259db4a88a983e12448b5aa22 : SUCCESS in 1h 40m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/71e103b81c5d4ad2a62a40c0a2dd79bb : SUCCESS in 1h 23m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/85f4ae3235764c8baf5a8cfcc5931b9d : SUCCESS in 1h 48m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/83549be2632148739928af9cae16c4a4 : SUCCESS in 1h 28m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ca01cc4581ee4112aa5462abc1893148 : SUCCESS in 46m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2f52c1b0351a4bbfa7d98d8ced4cc7d0 : SUCCESS in 44m 18s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/adb4a93b7e2c49bcb4616a2ed35b9f6a : SUCCESS in 1h 08m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/43a2feb937a54299933a21e25dca9388 : SUCCESS in 39m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/36c793d843894b6289e31d9d5e860da2 : TIMED_OUT in 2h 33m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/88fc4b60e20e4a1382aa5c5cefdb8a50 : TIMED_OUT in 3h 03m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/df71c78e2fcf4fcc90263e63a789acb5 : SUCCESS in 2h 01m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/87841c065e5a4450ab7e545c7e897b1e : SUCCESS in 2h 25m 06s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8f5d09d9a7f37a076f653845065c5b67c3fe0b2c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-16 20:47:42.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4b5ec13a959f3443760c18b4198fb517e36f5cc5","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-08-22 12:17:44.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"4f40677a776edcfa1fb7142deea3c552eef79622","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 12:18:26.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"868ad30cee47321b74b871442c30ee16ca888e30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 14:02:38.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1fc79f74a0e47db83d50a98d457b229\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/316ec3e2120a4d35abee8952069ae7fb : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16616fd5099d4546a11baba64f46fb18 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a4f9f01a543411c8480f151f40511d1 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18acb04bff5e47d2b6b44e1ee9e92a5b : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c498b1c10ef4041b12168cf110ec6be : SUCCESS in 12m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5f34a0b4b87e49fbb4864dd816ec3978 : SUCCESS in 5m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c8d36673d3db461ab23688082b15aada : SUCCESS in 8m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d5988dc3ea0c4c1ea3f01d23fd1e0cc0 : SUCCESS in 1h 16m 02s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/29715d76cf19453e914a1f7e20513850 : SUCCESS in 1h 42m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e9e4089c725b4f0994147ce2667502f5 : SUCCESS in 1h 16m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/25b6197186b84268a6048e1355a75040 : SUCCESS in 43m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/32075d3cc25b44f1be67be7d25de9d2b : SUCCESS in 54m 50s","accounts_in_message":[],"_revision_number":7},{"id":"c7d7553a9159d249df69f293b732bd9468907a16","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 14:02:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"e59a555becbece5c776b0876cacb0e120fcd174f","revisions":{"079d6a5c8a5628a5bffb5cf2d87f64cdf0333383":{"kind":"REWORK","_number":1,"created":"2023-03-01 12:33:46.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/92/875892/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/875892/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/875892/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/875892/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/875892/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/875892/1"}}},"commit":{"parents":[{"commit":"aaca5cb20b6605ffe3f99a011d8543d8d6ea0096","subject":"[sqlalchemy2] Disabling \u0027create_engine.convert_unicode\u0027 filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aaca5cb20b6605ffe3f99a011d8543d8d6ea0096"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 12:03:15.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 12:12:43.000000000","tz":-300},"subject":"Fix sqlalchemy declarative_base import","message":"Fix sqlalchemy declarative_base import\n\nSQLAlchemy reports:\nMovedIn20Warning: The declarative_base() function is now available as\nsqlalchemy.orm.declarative_base()\n\nChange-Id: Ic8d1f38f3891ca55fcb00eb6b3b5487b94dcd72c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/079d6a5c8a5628a5bffb5cf2d87f64cdf0333383"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/079d6a5c8a5628a5bffb5cf2d87f64cdf0333383"}]},"branch":"refs/heads/master"},"f4077e2d39e68f185c5bb1fcdfa15e00f18a8bfd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-03-29 08:44:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/92/875892/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/875892/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/875892/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/875892/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/875892/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/875892/2"}}},"commit":{"parents":[{"commit":"334a9fe0496c87eaf77c968f8ed4896ca7653102","subject":"[sqlalchemy2] Disabling \u0027create_engine.convert_unicode\u0027 filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/334a9fe0496c87eaf77c968f8ed4896ca7653102"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 12:03:15.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-29 08:12:57.000000000","tz":-240},"subject":"Fix sqlalchemy declarative_base import","message":"Fix sqlalchemy declarative_base import\n\nSQLAlchemy reports:\nMovedIn20Warning: The declarative_base() function is now available as\nsqlalchemy.orm.declarative_base()\n\nChange-Id: Ic8d1f38f3891ca55fcb00eb6b3b5487b94dcd72c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4077e2d39e68f185c5bb1fcdfa15e00f18a8bfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f4077e2d39e68f185c5bb1fcdfa15e00f18a8bfd"}]},"branch":"refs/heads/master"},"1cb0be1902cfe3c6ff8c495dff7a39fbcbbe44aa":{"kind":"NO_CHANGE","_number":3,"created":"2023-03-29 12:50:13.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/92/875892/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/875892/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/875892/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/875892/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/875892/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/875892/3"}}},"commit":{"parents":[{"commit":"bcf8b32abdbf2f4bcc121771c8dad419f2dd0d44","subject":"[sqlalchemy2] Disabling \u0027create_engine.convert_unicode\u0027 filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bcf8b32abdbf2f4bcc121771c8dad419f2dd0d44"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 12:03:15.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-29 12:48:47.000000000","tz":-240},"subject":"Fix sqlalchemy declarative_base import","message":"Fix sqlalchemy declarative_base import\n\nSQLAlchemy reports:\nMovedIn20Warning: The declarative_base() function is now available as\nsqlalchemy.orm.declarative_base()\n\nChange-Id: Ic8d1f38f3891ca55fcb00eb6b3b5487b94dcd72c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1cb0be1902cfe3c6ff8c495dff7a39fbcbbe44aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1cb0be1902cfe3c6ff8c495dff7a39fbcbbe44aa"}]},"branch":"refs/heads/master"},"5a4ba1943f75574ce6e2f8cbc399e1364958f861":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-04-24 06:50:24.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/92/875892/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/875892/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/875892/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/875892/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/875892/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/875892/4"}}},"commit":{"parents":[{"commit":"3f3223d38ae0ae1f6cfc9bb11e836b107edf0ac0","subject":"[sqlalchemy2] Disabling \u0027create_engine.convert_unicode\u0027 filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f3223d38ae0ae1f6cfc9bb11e836b107edf0ac0"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 12:03:15.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-24 06:44:18.000000000","tz":-240},"subject":"Fix sqlalchemy declarative_base import","message":"Fix sqlalchemy declarative_base import\n\nSQLAlchemy reports:\nMovedIn20Warning: The declarative_base() function is now available as\nsqlalchemy.orm.declarative_base()\n\nChange-Id: Ic8d1f38f3891ca55fcb00eb6b3b5487b94dcd72c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a4ba1943f75574ce6e2f8cbc399e1364958f861"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a4ba1943f75574ce6e2f8cbc399e1364958f861"}]},"branch":"refs/heads/master"},"fd233a40561c9d7de29c7e84426fc9fa084ab5f7":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-05-03 07:07:19.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/92/875892/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/875892/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/875892/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/875892/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/875892/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/875892/5"}}},"commit":{"parents":[{"commit":"08120206ea4f8e9e96eb00109ff5981f02e7190f","subject":"[sqlalchemy2] Disabling \u0027create_engine.convert_unicode\u0027 filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/08120206ea4f8e9e96eb00109ff5981f02e7190f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 12:03:15.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-05-03 07:03:37.000000000","tz":-240},"subject":"Fix sqlalchemy declarative_base import","message":"Fix sqlalchemy declarative_base import\n\nSQLAlchemy reports:\nMovedIn20Warning: The declarative_base() function is now available as\nsqlalchemy.orm.declarative_base()\n\nChange-Id: Ic8d1f38f3891ca55fcb00eb6b3b5487b94dcd72c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd233a40561c9d7de29c7e84426fc9fa084ab5f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd233a40561c9d7de29c7e84426fc9fa084ab5f7"}]},"branch":"refs/heads/master"},"258ef4b2b4d62e2d1b27584b94d61fc636864722":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-07-28 07:24:53.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/92/875892/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/875892/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/875892/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/875892/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/875892/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/875892/6"}}},"commit":{"parents":[{"commit":"14cd10794b0a9e55fa41684070ff7e905f57a2a5","subject":"[sqlalchemy2] Disabling \u0027create_engine.convert_unicode\u0027 filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/14cd10794b0a9e55fa41684070ff7e905f57a2a5"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 12:03:15.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-28 07:24:53.000000000","tz":0},"subject":"Fix sqlalchemy declarative_base import","message":"Fix sqlalchemy declarative_base import\n\nSQLAlchemy reports:\nMovedIn20Warning: The declarative_base() function is now available as\nsqlalchemy.orm.declarative_base()\n\nChange-Id: Ic8d1f38f3891ca55fcb00eb6b3b5487b94dcd72c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/258ef4b2b4d62e2d1b27584b94d61fc636864722"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/258ef4b2b4d62e2d1b27584b94d61fc636864722"}]},"branch":"refs/heads/master","description":"Rebase"},"e59a555becbece5c776b0876cacb0e120fcd174f":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-07-28 09:33:27.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/92/875892/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/875892/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/875892/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/875892/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/875892/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/875892/7"}}},"commit":{"parents":[{"commit":"dac994e3f5b2f85ef0f7c0cfcb02c1bc0907dd80","subject":"[sqlalchemy2] Disabling \u0027create_engine.convert_unicode\u0027 filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dac994e3f5b2f85ef0f7c0cfcb02c1bc0907dd80"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 12:03:15.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-28 09:33:27.000000000","tz":0},"subject":"Fix sqlalchemy declarative_base import","message":"Fix sqlalchemy declarative_base import\n\nSQLAlchemy reports:\nMovedIn20Warning: The declarative_base() function is now available as\nsqlalchemy.orm.declarative_base()\n\nChange-Id: Ic8d1f38f3891ca55fcb00eb6b3b5487b94dcd72c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e59a555becbece5c776b0876cacb0e120fcd174f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e59a555becbece5c776b0876cacb0e120fcd174f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
