)]}'
{"id":"openstack%2Foctavia~876034","triplet_id":"openstack%2Foctavia~master~I2ecfc0f40549d80b3058b76c619ff4ef35aadb97","project":"openstack/octavia","branch":"master","topic":"amphorav1-removal","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-04-05 20:21:18.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-04-05 15:04:15.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-04-05 08:36:04.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I2ecfc0f40549d80b3058b76c619ff4ef35aadb97","subject":"amphorav1 removal","status":"MERGED","created":"2023-03-02 06:55:26.000000000","updated":"2023-04-05 20:22:53.000000000","submitted":"2023-04-05 20:21:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876034-amphorav1-removal","meta_rev_id":"6c94097771dd219bf8fecd5666f14800fa174a88","_number":876034,"virtual_id_number":876034,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-05 20:21:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-04-05 15:04:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-05 08:36:04.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-04-05 15:12:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2023-04-05 08:37:29.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-02 09:38:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-28 14:45:47.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-04-04 10:47:49.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"4f7dcdaabd964cf815d5f40fdf0ce08f652f635e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-02 06:55:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"675d60685064babc342095979b843b14ffb405aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 09:38:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cf7ace7529c4aa48b5b6124185f302d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ca6a4850488f49faa5d7936e99b49da0 : SUCCESS in 2m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60d82d4c2c334758bd6b5ae72da7a032 : SUCCESS in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c660ff4fb4bf4b41b7b732aa2ee302b3 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/546ee2ee6ffb490ba5d5e4fc1ed76346 : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/275cbe491b7c46fba4a1068665d44926 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d81815210f2e48158d01bfa722bf0344 : SUCCESS in 12m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42336c32363246a6816189b18f4601fc : SUCCESS in 4m 56s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/08c046bf5ab8411e9e33c8d8d108fcca : SUCCESS in 5m 48s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ea0e586977bd4542ac5b27c169728750 : SUCCESS in 9m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/28b439f0443245efbc6005d24cf7905b : SUCCESS in 5m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0f231d7c3ff54bfea2c12e8c91d8b3de : SUCCESS in 6m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/37dd30fa02dc49f3a55a437d55089494 : SUCCESS in 7m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fa2d660bdb134b5d9340f8174adbcf24 : SUCCESS in 7m 59s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/465d07c26af940f3aa77a7ffe830270d : SUCCESS in 1h 11m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c6088b84a2f64996a99949a7e169f6fe : SUCCESS in 2h 25m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/10b82a91ba3c449a85e4d4c2c4a952ff : SUCCESS in 37m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/142379a1ef3943adaf282a49fc43f39b : SUCCESS in 52m 03s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/39d8e151c51e4e7d8f600dfda3d9d911 : SUCCESS in 1h 18m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4ef3e73205df4a749f24db3da75443b4 : SUCCESS in 36m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/011cfc18a16d4366862ff1fa4df7344a : TIMED_OUT in 2h 34m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/efc8b62023314a12812eef55e7491460 : SUCCESS in 2h 36m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/da833f2d38884eeaa95074c4e62e088c : SUCCESS in 1h 45m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/710aa6065c9c4aab83829bce5b4623b1 : TIMED_OUT in 2h 33m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5baed2a399ad43a29eb0e9c401059f1d : SUCCESS in 2h 29m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/56ac9820fd9c4454b41b01781efcd4a4 : TIMED_OUT in 2h 34m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50a27b91598608b20c3cf42f7cac8e11b4f67bce","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-23 08:25:17.000000000","message":"Topic set to amphorav1-removal","accounts_in_message":[],"_revision_number":1},{"id":"179ed3c0fcc97d60870fa047b5ffe9ee223e52b7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-28 14:45:47.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae02812e7f4944dcdc631b8cfc9bc449534cd696","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-29 07:28:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"43325db0b30f24037f041a410722ff5c8b6d1421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 10:09:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/516cf9974c8d40d58971eb687d40a1ae\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e21b74d3dbc2491f8d0fc6aedacafc69 : SUCCESS in 2m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8aeef150cea5419ca1e9db4f0e24b42d : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7145c261d804bdbaf909b3be2522c8e : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c251b1fa2f947bcb77279485e969d3a : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe2d072339404a2b97d7e11e0f059685 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d62da0147dc14db59c1aba00d742edb2 : SUCCESS in 11m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81af3e2a4ebf4f96975d720435fa47e4 : SUCCESS in 5m 10s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0f827f8613b343bdb9de512d144fd23d : SUCCESS in 5m 24s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/9c9ce5801dbf4a929730876cce9fe5fd : SUCCESS in 8m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8c10062d3e1f440ca6387be3b210410f : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6dd75adbee254c6789bb083896e03b47 : SUCCESS in 7m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/07391831b6f34078b5256e8d085e6136 : SUCCESS in 8m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/73326e8e51b845baa99f35fbf32c537a : SUCCESS in 8m 20s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/483012dd9b9248238e6adfc8ca7bf925 : SUCCESS in 1h 31m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/81d298744fe449458f1d7033c5d0f5d4 : SUCCESS in 1h 55m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/013c2301b49c4d9cb314e57dee3f88f9 : SUCCESS in 44m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/687e119c52434a09993abdcf78bdc9c3 : SUCCESS in 48m 27s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/67b1cde27e244195baf659d4dfb7f606 : SUCCESS in 57m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c84128271e574762a26c935fc9a5cfea : SUCCESS in 42m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c64bf9e369ec40eeaf099f6ff56d7a67 : SUCCESS in 1h 57m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/65b0d68a7ff441f386bf1eac67700521 : SUCCESS in 2h 12m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9282d69931aa422ea1554d09487eb1b9 : SUCCESS in 2h 09m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/95c13e11526d4561ae5b4e17246d3ccb : SUCCESS in 2h 01m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/55aecc343cec44ab8ab8b71f97d375b7 : SUCCESS in 2h 12m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/94e9da200920443283495bfe44cce669 : TIMED_OUT in 2h 33m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"83eb41b17a9dba9a88514627025b3638892ef4d5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-03 20:48:03.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"da769a79f844eb24199bb9b1614e71be73eead33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-03 20:55:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e2ecd1227a75c781f1f085efc1d134d2a8b478a3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-03 20:55:44.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"883bb140eb56134b8f569a658e4067c72c6eab51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 23:12:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c129de3f22048748d4fd089f899adf2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/41350f6d8dfa4d8180d652ef57e7f5b0 : SUCCESS in 2m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5902c3bba5a4db19d76fccf0ffdac72 : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89f15882acc94ddfadc76cf7f5383cef : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/694dc6b4313546dfb08c11be8d5c81eb : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ecb0a5042e74d64a230f4af649c7f4a : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51a6547fb4434cb69202f2e5c6c69b3e : SUCCESS in 10m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/441ac8b8a7f543379aae4c617b75da5b : SUCCESS in 5m 11s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5f707bbc4b714df0a2a0e5f75413018b : SUCCESS in 4m 52s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ae94f38c42e9462997aa288b6d19e072 : SUCCESS in 9m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2991253f63e0463cb397c8cb7378be8c : SUCCESS in 4m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1e26ed21b4bb4bb0a37a9c26530d4636 : SUCCESS in 6m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/36b8546d99094a648a3d26ded3fee957 : SUCCESS in 7m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8f9c5923b272467db7e04940c0684d6b : SUCCESS in 6m 35s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9f62d4dea8a5400ebe30b02c238bcc00 : SUCCESS in 1h 22m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2bce294e809c492ea2bc7976b008bea1 : SUCCESS in 2h 05m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/22f00c2a449641649c6d5826eeb62feb : SUCCESS in 35m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a9b8ed080a7e40eeab55aa3624a07474 : SUCCESS in 38m 28s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fd7d2ae7f3ca4e94a1953e463cced505 : SUCCESS in 59m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c4fc1ec69a2f4c1589bd09c89398d350 : SUCCESS in 33m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6d12000f0abf4385bc206e608ef01c56 : SUCCESS in 1h 44m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f3faef36608349c1950376fadd44c197 : SUCCESS in 2h 15m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e3181fd5ab0a4067b0a2a5a972970de4 : SUCCESS in 1h 37m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fbb8e500f18a48c2aa228176769ef4bc : SUCCESS in 1h 48m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0a251ae6aee948d8b26de93a02811910 : SUCCESS in 1h 41m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c54814cb86f4ed78e529b41590936a3 : SUCCESS in 1h 55m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"410fa948989ca8ffb48b8a234fde36cb0b4d2301","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-04 10:47:49.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"955f52344d899d210dee5992a694cd1207066dd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-05 07:23:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"34e54af104e6f5d30f47a2effe423a54074135e3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-05 07:23:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ed889b3770f1fc4f9647cc190d24b9b127bfd6a3","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-05 08:36:04.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3c85b342e5ca608c0458eafde947e8ee272bdece","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-05 08:36:30.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"ce6cce0cca902afe9efa74ac3d1eba5e5cb2e552","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-05 08:37:29.000000000","message":"Patch Set 4: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7e22827cbbabd9e6843d6509f93b197111dda8a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 09:59:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b2664a414ced4fa2bebe630d93a5677e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1318c202df4c40b4a4172415dc3b4123 : SUCCESS in 1m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1db971eb16944d6a8ece0fc0c0e27352 : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee0aee528eab4193b22b790d01e2d023 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83654ea766eb469c81fa381e869519de : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c723731a85c4b149c5d5dbef3626271 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66e9e901f0cc4fbb87798b6226909788 : SUCCESS in 11m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/592fd7a1d4dd42e9835ed333e389ffe5 : SUCCESS in 5m 18s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a5da5a2de08c4f8da8885dbf2acc6e71 : SUCCESS in 4m 51s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/6a75e2d1f7f64a2bb5bfe90acd0773ed : SUCCESS in 7m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fba0a66c11d549aeb8bbe769127fb834 : SUCCESS in 4m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e2818fe33c28492e9fcd418c28d3180b : SUCCESS in 7m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dd0fc2a4ca8c4e5bbcd9fe649c8f6a68 : SUCCESS in 8m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f1822116d457454588a0c6a521ca7e8b : SUCCESS in 7m 12s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f462b2cfb545435da7c7c507c74d2663 : SUCCESS in 1h 49m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b46abca1f4c843eea0447c6f37b3ffc7 : SUCCESS in 2h 26m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/76cfa6b0ba8a427892a9a5ade5eb4d14 : SUCCESS in 38m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5456b9482c1c44caa8405b54bd38eba3 : FAILURE in 7m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/db9654832c88455c9cdc23c149b5109e : FAILURE in 34m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9c556ad731b04781889b9855b350cd18 : SUCCESS in 34m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/209beaae90a948a8b39be4a659207be0 : SUCCESS in 2h 31m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fed07dfe8ef0426785c67430ed4d118f : SUCCESS in 2h 30m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fda56ede73634e39ad5a8578a4e5fc68 : SUCCESS in 2h 03m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/585afa968604446ea9b54865bcce90a7 : SUCCESS in 1h 48m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/15625c88c3ad484ba7cbac674f72b9b6 : SUCCESS in 1h 41m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/6eea3cde784d48808569a3eda4f16235 : FAILURE in 25m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"771f642040496ca11f154b8f24f400dece7be1aa","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-05 15:04:15.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a2793db3004cd09acc2195a12c2d817842a672c8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-05 15:12:49.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"691aed7e9c368e916c0a3c340922e6d304380c49","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-05 15:12:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"24dac695449b6103465984d9bdfdbd6988b1b0b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 17:59:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e741a15578c3473bac92233b96676df0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0aa5d91e105b4fe08ec5aab92b6353dd : SUCCESS in 1m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1098e26d51e4c5ea9ac93e39cdc890d : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acbc581a1f0e43d8a7feb70182ed52ea : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc9d0d98b6ad4c12b0995ce71aedab33 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4fa456d26c941cf855a08f6c1ed03f8 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e30b728325948a7b351abac52453e31 : SUCCESS in 10m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55f825fe89bf4698b5d0eb823caaaa79 : SUCCESS in 4m 39s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/416b6aac6b084bdc9cda4d400940d802 : SUCCESS in 5m 32s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4b977d5be6a3497f94370d0ac2d6b930 : SUCCESS in 7m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/394ac0af792c4ac0a192e62d21af79dc : SUCCESS in 5m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5c658d785bf44c06ae528ed6cfc2f6bd : SUCCESS in 6m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/15e1b29825ba44a4843b7f1757b5989c : SUCCESS in 6m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/26d22a09f51346b6b321346dc489705c : SUCCESS in 7m 01s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/56388a6913b045af8a68244fb578760b : SUCCESS in 1h 36m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6d53b5878c5c4a9b9cfda4e68ec46ed9 : SUCCESS in 2h 33m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c68ea9321d7741d883a0bb85a87d4310 : SUCCESS in 46m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ad08d38b24ad4f888d4c1a8dbf805b8e : SUCCESS in 53m 17s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/dd0932140efa46f1baf95c3babffabea : FAILURE in 38m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84579f08370b4bf7ac9e2e2121a29154 : FAILURE in 5m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/59e2a36f2e334a87a0321ace169fd6f8 : FAILURE in 6m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/101269605a764d59b2c75b0d6a6d4ed5 : SUCCESS in 2h 41m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/48d425ed6de448fba49eb840a7af182f : TIMED_OUT in 2h 34m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f816e3f2977d430c942b9d6f5b26174a : FAILURE in 7m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/99a5ad29532642f69dbd36532eb9e756 : SUCCESS in 1h 43m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/22c66c7d31ed41469e5c28f206a0aa3c : FAILURE in 24m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9c7d673eb6962e65e7558ab18f2dadc93bac1c4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 18:00:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d47fe43c4578e4234acd6091ebd2b85f5144e633","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 20:21:17.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a73cb2c95144375b3e83f59e1162f1c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b56bf66258084fc4be1d2d299341640b : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9f90f0d0b944b388dac5f9201b8e061 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/076fe2dcdff546a28e6ea85aaf51f8db : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e58d444180d40a78a471507804aa043 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d5b544fbf9845d9b8c9af779ad1f6a4 : SUCCESS in 11m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9902d09fc9de4fb2894d3f19455204cf : SUCCESS in 5m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/723caa15a589412ebbe85930e60762a3 : SUCCESS in 4m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7f035bc6073a4ff4a2110b34d01ac8d9 : SUCCESS in 7m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b50d79270f9846b19563c4beb7c835c0 : SUCCESS in 1h 13m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce7ca4b7c01543349673f20065f1ef7b : SUCCESS in 2h 19m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4b294b35129741598e8249492fbf0267 : SUCCESS in 43m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/676f01e1ee8449e7ae7ec3b0ca12614c : SUCCESS in 47m 48s","accounts_in_message":[],"_revision_number":4},{"id":"0037004218624c46bd86c1784c0d107e673c1cfe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 20:21:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"6c94097771dd219bf8fecd5666f14800fa174a88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 20:22:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f52ee82f918544879b0c2c20d6b3b6f4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a59624dba44f4aac92f15bb4c82d2569 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a32a2bb5d224ddb9fb0070977d5b74a : SUCCESS in 50s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6c0515c9880b5310dd0647a5881233f07898d812","revisions":{"d209249239953af20d6c0925802481260ce8623f":{"kind":"REWORK","_number":1,"created":"2023-03-02 06:55:26.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/876034/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/876034/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/876034/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/876034/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/876034/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/876034/1"}}},"commit":{"parents":[{"commit":"3475a4b3c2b0614d4a6f1e51d6b08d3abcc5f7ca","subject":"Merge \"Deprecate user_data_config_drive option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3475a4b3c2b0614d4a6f1e51d6b08d3abcc5f7ca"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 14:46:38.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-02 06:03:06.000000000","tz":60},"subject":"amphorav1 removal","message":"amphorav1 removal\n\nRemoving the amphorav1 provider, it was deprecated in Zed and can now be\nremoved in Bobcat 2023.2.\n\nChange-Id: I2ecfc0f40549d80b3058b76c619ff4ef35aadb97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d209249239953af20d6c0925802481260ce8623f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d209249239953af20d6c0925802481260ce8623f"}]},"branch":"refs/heads/master"},"aff294d8b5fedd30b54db7b2ec7bb5fe1a37fc8e":{"kind":"REWORK","_number":2,"created":"2023-03-29 07:28:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/876034/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/876034/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/876034/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/876034/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/876034/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/876034/2"}}},"commit":{"parents":[{"commit":"1794a61899dc5fb65d6a36640c41028eaf0d5ff5","subject":"Merge \"Fix pool creation with single LB create call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1794a61899dc5fb65d6a36640c41028eaf0d5ff5"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 14:46:38.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-29 07:25:01.000000000","tz":-240},"subject":"amphorav1 removal","message":"amphorav1 removal\n\nRemoving the amphorav1 provider, it was deprecated in Zed and can now be\nremoved in Bobcat 2023.2.\n\nChange-Id: I2ecfc0f40549d80b3058b76c619ff4ef35aadb97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aff294d8b5fedd30b54db7b2ec7bb5fe1a37fc8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aff294d8b5fedd30b54db7b2ec7bb5fe1a37fc8e"}]},"branch":"refs/heads/master"},"bba759577ea5f92abb36172911cb06a348330437":{"kind":"REWORK","_number":3,"created":"2023-04-03 20:55:32.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/34/876034/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/876034/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/876034/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/876034/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/876034/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/876034/3"}}},"commit":{"parents":[{"commit":"1794a61899dc5fb65d6a36640c41028eaf0d5ff5","subject":"Merge \"Fix pool creation with single LB create call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1794a61899dc5fb65d6a36640c41028eaf0d5ff5"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 14:46:38.000000000","tz":60},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-04-03 20:55:31.000000000","tz":0},"subject":"amphorav1 removal","message":"amphorav1 removal\n\nRemoving the amphorav1 provider, it was deprecated in Zed and can now be\nremoved in Bobcat 2023.2.\n\nChange-Id: I2ecfc0f40549d80b3058b76c619ff4ef35aadb97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bba759577ea5f92abb36172911cb06a348330437"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bba759577ea5f92abb36172911cb06a348330437"}]},"branch":"refs/heads/master"},"6c0515c9880b5310dd0647a5881233f07898d812":{"kind":"REWORK","_number":4,"created":"2023-04-05 07:23:14.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/876034/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/876034/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/876034/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/876034/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/876034/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/876034/4"}}},"commit":{"parents":[{"commit":"1794a61899dc5fb65d6a36640c41028eaf0d5ff5","subject":"Merge \"Fix pool creation with single LB create call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1794a61899dc5fb65d6a36640c41028eaf0d5ff5"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 14:46:38.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-05 07:23:06.000000000","tz":120},"subject":"amphorav1 removal","message":"amphorav1 removal\n\nRemoving the amphorav1 provider, it was deprecated in Zed and can now be\nremoved in Bobcat 2023.2.\n\nChange-Id: I2ecfc0f40549d80b3058b76c619ff4ef35aadb97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c0515c9880b5310dd0647a5881233f07898d812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c0515c9880b5310dd0647a5881233f07898d812"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
