)]}'
{"id":"openstack%2Foctavia~876035","triplet_id":"openstack%2Foctavia~master~Ibeb9132e65706b50706249df3e8b6c455863513c","project":"openstack/octavia","branch":"master","topic":"amphorav1-removal","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-04-18 15:47:09.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-04-13 18:03:04.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"Ibeb9132e65706b50706249df3e8b6c455863513c","subject":"Remove specific handling of amphorav1 parameters","status":"MERGED","created":"2023-03-02 06:55:26.000000000","updated":"2023-04-18 20:39:01.000000000","submitted":"2023-04-18 20:37:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"876035-amphorav1-removal","meta_rev_id":"70d3897bbe0b1e84cb89fc463d70672911ae48c2","_number":876035,"virtual_id_number":876035,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-18 20:37:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-13 18:03:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-04-17 12:19:41.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-04-17 12:19:41.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":-2,"date":"2023-04-17 12:19:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-02 08:22:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-05 18:52:23.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-04-12 17:45:57.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"daa2a46d92b77ab50971ebaae4a1a189262c87e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-02 06:55:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65c32fb503637c4a1a19c4fe856ac7528bc0ce9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 08:22:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b2c36f8b5c5c4ed7a18732209365037e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c4316e85d29410ba586d2cec72c066e : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae9969bce65f4fa49f9ca06d5960bbd2 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27adde77627a454d8d3556c5c8e03741 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a60094ff52441baa62a3e6b4f55b4b0 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d49155ae9fee4eeb877e3e5e10c82693 : SUCCESS in 11m 27s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8b891977398e40b9b63f724ce0587bca : SUCCESS in 5m 11s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0e9894b255944477833c171b2aba27fd : SUCCESS in 7m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9de228a645cc44b1a335e37f09cb1e20 : SUCCESS in 4m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/63e3e38d84474aae91ba69c9009c8dbc : SUCCESS in 7m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/757f64d5e244463fad9f3e17bbb8b79f : SUCCESS in 7m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a1f5575018ef4b2e9a6f93aa1cf62beb : SUCCESS in 7m 13s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ecc50d978b9846b6a36c557d65365bc0 : SUCCESS in 1h 24m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c6a12586f9054adf95243cd44e2f77c2 : FAILURE in 40m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9afbf4ae9a584127b383f2325a3c3627 : FAILURE in 29m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8ba96cb08aed468f8a7dd342f2e1a843 : FAILURE in 42m 31s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6b7f5944502544d59d53cf5b89cc7c93 : FAILURE in 57m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/634cb388a99f40c19904d59ccadc1815 : FAILURE in 30m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/10b95e2abf404aa793ba25eabb1dc593 : FAILURE in 37m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e72d4afb27f048b08e11b29b5a244152 : FAILURE in 44m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/764e5d2d4efb4399914d7fa8fc7554e7 : FAILURE in 45m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f429b05714be4331817b73e5b908c8b4 : FAILURE in 36m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d3a543d6f55a4a889ce69b6712daeab5 : FAILURE in 48m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9ee9e12f9864a5b878a89892f82b48d : FAILURE in 46m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7b1f8b435403b0e604aa9ea898b407ee5de64b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-02 17:05:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e0d7dac854faf0fda681811beb052f39ae6df730","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 20:11:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79e39d401b704222a45bede3dc9e11de\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0bdfdd64df448929419ff8a77191cf3 : SUCCESS in 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00add10c05864e9f81709bd15058bc83 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ab223555cb649c59685b6e4e8f01a54 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03649d8837c74960b5cda20a5b839693 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/907aaff2836746bb9325f91a3fd9a7c4 : SUCCESS in 11m 39s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/55f34eec29ab4a7b87f945801049bf71 : SUCCESS in 4m 59s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4f7a394814654301bc175261772e946d : SUCCESS in 7m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4d8f64657f0846979b898cdd60c6a48c : SUCCESS in 4m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5e74ad7a525a4dc9a71e3dc86b8a6e95 : SUCCESS in 7m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4c3eb4a1ae1645a98a63b31e436eb503 : SUCCESS in 7m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/051ee93f0f72493ca86b034f335047b3 : SUCCESS in 6m 25s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b7baaac80eed4c99a2346d1edca24a12 : SUCCESS in 1h 37m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/92cf46c4270d4ae3941d2e18ab8a06a2 : SUCCESS in 1h 42m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/459268f052bc4af2a563384c664b696f : SUCCESS in 50m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3694a180a8594dfe9cef4b3f446c9d75 : SUCCESS in 37m 43s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7ba0bdc35a4245279f3e1de986545e51 : SUCCESS in 1h 15m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6bbfa83decd642639ced48e2385a56cf : SUCCESS in 33m 08s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c23ce68df9e34874813109509424812b : SUCCESS in 2h 04m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f2f16b1e38f34da3a1fd6ab32df6a979 : SUCCESS in 2h 58m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2f121b21915d4a0fa9d97077a6a83157 : TIMED_OUT in 2h 33m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e0daec0bd9a34882b48a7b7c02941830 : SUCCESS in 2h 27m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e900b322c41a477b95c6782abac9df81 : TIMED_OUT in 2h 33m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e383f80555f54905a623cbfc35ae8f8e : TIMED_OUT in 2h 34m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"295de2cb3be73a0a3b886b79c3079459f800fa92","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-23 08:25:33.000000000","message":"Topic set to amphorav1-removal","accounts_in_message":[],"_revision_number":2},{"id":"0f12fb6b96c1cd7b825052854f073a4ad716ee47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-29 07:28:58.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"d87408698f572cd67486ec70591490fa2e56c420","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 09:34:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8989bf5c1da244cb85f923f19c34c41a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c029a4aa4f90498d8a5ba1624af5e734 : SUCCESS in 11m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d571014c7fa43f9873e20adf8dbc618 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/105bc86c0f524b37b6b3ba33f4ef29f2 : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e18d7c033e1f490cbf55ddb8e1728e3c : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a2d807449044d1f8a7dbf3093b5ced2 : SUCCESS in 13m 19s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f1fb03eb7d0a4cb3b7daad4df1f07e82 : SUCCESS in 4m 17s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/59ee726a6b7b4f6d9e8dd9f17b2f695d : SUCCESS in 9m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ecef69059e6b48b7b04a48fedf30490b : SUCCESS in 4m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/84d5f0bf090340868600a598c9eb081a : SUCCESS in 7m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/067fbd78284e4e56b33381f6343a3e61 : SUCCESS in 9m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dbd5dabb73604932acb0f0776adb9b23 : SUCCESS in 6m 16s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1eab7342aa014cd5a8934c87b076af3d : SUCCESS in 1h 22m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ced1bae06c1c4a96861cf093cf7d0725 : SUCCESS in 1h 32m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6ceee7160b31473587e791eaed5db9b2 : SUCCESS in 33m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b0eea82f850b4d52b1cec6781c4af70f : SUCCESS in 39m 42s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3d060db51f6142f1a173f6bb3001f792 : SUCCESS in 1h 17m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/87d9e7f54d544ee0a0c1e878a9680b7f : SUCCESS in 32m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7e0a61de7ca04fcdb21b303b38b145d1 : SUCCESS in 1h 47m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8cb84c6fcf8e4b78988a0c910dd6c44a : SUCCESS in 2h 03m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/71c8d579ab074a39abd09a7497b2a5e6 : SUCCESS in 1h 51m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/84b4fc6dd9ed4f5abb03a30961c45430 : SUCCESS in 1h 49m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/54b2283d9edb423f8ab6bdf84b63703b : SUCCESS in 1h 53m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3be606010e4e46e6bc06c751be4a4601 : NODE_FAILURE Node request 300-0020855898 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"07899e3aedcf7f89dca67b7c300ca8d6804d6ca5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-05 15:03:49.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"31511e2ffcd5d49147e82903e7bec44a029538fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 17:17:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdaf6fb826454b9c9adf42128f423775\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1966523c0e5453f9f9526473ac9c546 : SUCCESS in 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cde283e958a4b618fcc6a15e8e1bad5 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70a022ab3ebb4620ab47a4d101e66b05 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5309034713c947c58a4bf537ad08284e : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac36d2a3fb044af7956c36e06f3cfa25 : SUCCESS in 12m 18s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d4d7c3681dbf4bc78f652c52d8338922 : SUCCESS in 5m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/2b9104e81d69465fa4d516ae450f57b0 : SUCCESS in 8m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/10ac182410a341148394057fa1127ce4 : SUCCESS in 4m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e32229bb917842ecb4c48ec139b9160d : SUCCESS in 8m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6caa2f00755948fd9f3570ae101e7cae : SUCCESS in 7m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5f3cc8d47e6a4113995bcc725b2c57b6 : SUCCESS in 7m 11s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7f80324ff7754fa0955e5582de65acce : SUCCESS in 1h 24m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf81a3ab33db47d8becc62bb38656ed8 : SUCCESS in 1h 50m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d055a403cd7b49c38854308676116f8e : SUCCESS in 39m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1fc00bd80fac47f4af93776a6d21f490 : SUCCESS in 40m 35s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0604ccfb477d4c5bb724a0165b786502 : FAILURE in 39m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/150a69f6bc434bbda571cf3a2b0a5b16 : SUCCESS in 35m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3fa75296a03146318df6173023672a96 : SUCCESS in 1h 53m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/23dbc823feb0450a90b8e6a962f360ca : SUCCESS in 2h 10m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5561f8f38a3e45ad818ad7cd904d181c : SUCCESS in 1h 41m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0f9588788886491a899f24cf34ddcff9 : SUCCESS in 1h 44m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cedfa636d7c549a8bbdceeff75d7a7f8 : SUCCESS in 1h 51m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/705e024cedc949a3abc731081b4f8396 : FAILURE in 29m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6f61e23f347905172a886d03b53d166f828dadfd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-05 18:52:23.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d652bfb4dcba3c37473d2bdc97033e280418af40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-07 08:28:32.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:NO_CHANGE OR changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"changekind:NO_CHANGE\")\n","accounts_in_message":[],"_revision_number":5},{"id":"93d2a07d3084e0d6abfc0ed0cce7cf716eee77c7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-07 08:29:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1c3834e185601a0460bb1f523f0f57c99b8fbc37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 11:33:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e36d9b7825c48a1b241116dfe4d1c14\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3ec44c8b6d14aac84f032ed4986c3a6 : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dd180bd3b384b18b50f0160a2eea16c : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/417d8a792b344af293e49e46d9bfad06 : SUCCESS in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87c2513176b74284b4c8847e9e59ecab : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09d7498310114be6ac57eaac91f2a373 : SUCCESS in 10m 38s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ed452c1f62954593bc9b4b6891e11ee7 : SUCCESS in 6m 18s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/6b402d0b6bdc4c04b9797bc2a7304773 : SUCCESS in 8m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ac2315c134da49db9bf2c3bca028de45 : SUCCESS in 5m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f6e44a418bed4bedb228a1b5e4ef6627 : SUCCESS in 7m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/70237bd3456f41e69b42574dade865c5 : SUCCESS in 6m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/26aa5385637a40e9b8a5176ec265c2b0 : SUCCESS in 9m 23s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/432a6bc2724b46f998ccd605201af3d7 : SUCCESS in 1h 25m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/48b41f2aab9e4aff94c6eb5cd0c2d53f : SUCCESS in 2h 30m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/84dd2fc86a1a403a8e1c0e41a8914e38 : SUCCESS in 51m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9eb4905f59a54a67956e52094575baae : SUCCESS in 53m 37s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/24fe859b5a0b4fa8ac2bfa37a42bd6cf : FAILURE in 38m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/230c382de8bb472b8c98958842fefb9f : SUCCESS in 35m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4188ecc3181b45b8852dd46761bb6a37 : SUCCESS in 2h 15m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6ba01c0990e242a5ba3ba9b61a68858c : SUCCESS in 2h 49m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/55902b263d3c48a381812c513e8a81d6 : SUCCESS in 2h 08m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3d7c4a14adc643599b856c8bd6d16e65 : FAILURE in 7m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0e361347f9ee4709a6c424b373b28551 : SUCCESS in 2h 07m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/6de0144a6feb498ca9b883a6c20ce904 : FAILURE in 24m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"23411930c66bc34d1c1a6c1b538b15ea591b8858","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-12 17:45:57.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"869ba96588823dd549176d167bd0c57afdf10649","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-13 18:03:04.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a06d0383151be0e0c6bd8a94ff4e9756cd26d1a1","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-17 12:19:41.000000000","message":"Patch Set 5: Backport-Candidate-2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c6bb172ab80039388bc95af4b6d3a815f93de6af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 12:20:28.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9a27874dd711065dce02a61e0eba39d205054bfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 12:27:02.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b491b6587c9c41f29666e7cf84d7edff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9a092995b6e433ba6f95a77e6b4dd59 : CANCELED\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/806f5d023fad4d55babb0ab05f8ec602 : CANCELED\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82f0cf888105475aa56709492af87aac : CANCELED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee61a344bfb142d183e584c6a621bca0 : CANCELED\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/67b90bb6f8f14bb2a665b792161f3099 : CANCELED\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3799f5bbd3fe44a88a53d2572e713011 : CANCELED\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/987814224a16442090a248ccd377164f : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4fad8c40dfa24d0bb750c8a7d59f9d8d : FAILURE in 5m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3efb547806ca4421bc6ab7c3d6afd7da : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/693d79cc59524dd4bf2de4880b357658 : CANCELED","accounts_in_message":[],"_revision_number":5},{"id":"7eed8150ba3f630c857268fda426eafa88107a42","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-18 13:04:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cfe4df6117b45443da35d23d7b272df711a591fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 15:45:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/568724b8a51a457f892fa3305aff46db\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb9749ed35ca4be3b23a5f4815485ea7 : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ed0d5438c4e4874b9a011d7402db2e6 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11a2084ed56640cd9f64706b658e332a : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12bcb07f18e74cbdb6aca8e369e4638e : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cf93d21f47747da8ef1429971c9b2f4 : SUCCESS in 12m 51s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1ab6c162ee7a439e80a86219bbfc0dcc : SUCCESS in 5m 11s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/691d6ebac3e54b32a5572cbb3c3433f8 : SUCCESS in 7m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/13a9235a349f4a1c9a1fb197a31ddbda : SUCCESS in 4m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ff51503610e441d580d89d80d3799e26 : SUCCESS in 7m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0dfc29337fc4430daf8445d4f6dbf7f1 : SUCCESS in 7m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4f9c7ea9fd4c4c578f1beb235247c289 : SUCCESS in 7m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7e11891ed8894358b8b2370d99da8caa : SUCCESS in 1h 34m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11c28eced2b1455888f958487c74e155 : TIMED_OUT in 2h 33m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6ce2fa7681434d21bdfa6d498e653025 : SUCCESS in 49m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7e0b1934f19e4d37a4221a72ce3e3be4 : SUCCESS in 51m 09s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/48308978483e466b8c0bede7407b24b5 : SUCCESS in 1h 11m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/49e4628adec94a88bb342bab3dcf7efb : SUCCESS in 40m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c27bb810646e4d8f925720bc666429eb : SUCCESS in 1h 52m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/25408fb28c9e44cdb88f3b6bb7082f64 : SUCCESS in 2h 21m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d1c5d0f28ec54e9988e02fa54a01bde6 : SUCCESS in 2h 21m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f7dac64993c4418eb06c83af6c736521 : SUCCESS in 1h 59m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b85672d27eef4661948c717453b096a5 : SUCCESS in 2h 19m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/84370db2897e42e0b24d7535ec48297c : TIMED_OUT in 2h 34m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6adbab7a303ff18d0837f402da28244177b534a1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-18 15:47:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f3b2903e5a263e2b6b57bd7f8f73cc3d48f6c246","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 18:22:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/292b07f7e2ac47959d1b3d96fa096d12\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f95436d0711f4e9998ed25544da261a1 : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea091d6f894545289d5c6e7ce63eada8 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04ffe8f2c58a45c692b7193d99e72186 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9c2570e8f2d4f7381d863528f3f90b4 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c95816efce54ec79996022883e9c7ce : SUCCESS in 11m 45s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1a47e3d7851445619ef12604af458f39 : SUCCESS in 4m 50s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5afdb653d8ad4c21aa0135412445ba54 : SUCCESS in 9m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d4b37aa001644db9ad9504743a1fe2eb : SUCCESS in 4m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fd54a7c490fe45948e723ee8e73b3eef : SUCCESS in 7m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e94ea24901b7407ebcc5db8eb725fda3 : SUCCESS in 6m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9e46adcf8e30460f950b7315b3c40739 : SUCCESS in 8m 44s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7170fe2c9d1349c6bcc99924ddb0982e : SUCCESS in 1h 38m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d75f1e016ab40489c0f9f1bef5eb7c5 : SUCCESS in 1h 49m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c0e0e08cd3ef40aea201d010e049d9ea : SUCCESS in 36m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/edf527c228014dcdb59cce589056851d : SUCCESS in 45m 08s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4bb147d69fd243fdbbffc87112e56ab6 : SUCCESS in 58m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/25f29535f5ac4c778223471e1772214b : SUCCESS in 35m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dbc051eddb864e91bf55d334e95afaf8 : SUCCESS in 2h 15m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0cc42be57d7344c280280af2742704a8 : SUCCESS in 2h 30m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9c8ed52c7b734bb887fb35f8f7f760fd : SUCCESS in 2h 32m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6d8880d8b7094a1492caa161a6cf1647 : SUCCESS in 1h 51m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b0f588357787478fadc662b0aa86ecfa : SUCCESS in 2h 17m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f91dbeb376449979bcb3ee19da6a591 : TIMED_OUT in 2h 33m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f0b535ea90ca548b88fef9fa41391a59749dbaf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 18:23:13.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c265fd6fe5fb3be389d2bfada15cf779afd9fb85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 20:37:36.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac4a6383b35c4b3689855bab528e3c67\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2bcfe5159e54c8c9b73d6117d935131 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ecf500d6dca6492287ea31458fe909a8 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ae468a5846a4874b1d32f9e832536e0 : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a92b493ecb744fc9a57e930b6181bf67 : SUCCESS in 11m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b3fac6456af64d6091cbe3883cd9aa72 : SUCCESS in 4m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/71d70a78f70740049cc2eb822201e56e : SUCCESS in 7m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/af2dd519c3a747a2b2623dcc99e717bb : SUCCESS in 1h 24m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f9b80342a66447e7a37b145a042d548f : SUCCESS in 2h 13m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2b4fa61d919f4988bb1c5df85047b6de : SUCCESS in 39m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e7972e6c816d4a59895013c104f8511d : SUCCESS in 42m 56s","accounts_in_message":[],"_revision_number":5},{"id":"5d7c6eb7d16502c20077d7071a27f2af143149c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 20:37:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"70d3897bbe0b1e84cb89fc463d70672911ae48c2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 20:39:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f8e6666d6854d9ba2eabda3183b90ca\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdb47d80a5074ac88a51a401cd57f79c : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"324c3528ec7726fb6c69f70473b75b14357ff14a","revisions":{"da21de58f59c99137118932644ed97a1ae83f2d1":{"kind":"REWORK","_number":1,"created":"2023-03-02 06:55:26.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/35/876035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/35/876035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/876035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/876035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/876035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/35/876035/1"}}},"commit":{"parents":[{"commit":"d209249239953af20d6c0925802481260ce8623f","subject":"amphorav1 removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d209249239953af20d6c0925802481260ce8623f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 20:51:19.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-02 06:26:38.000000000","tz":60},"subject":"Remove specific handling of amphorav1 parameters","message":"Remove specific handling of amphorav1 parameters\n\nSome functions in the drivers accepted 2 different types for the same\nparameters: provider dict or provider data models.\nNow that amphorav1 is removed, we can remove the cases that handle both\ntypes.\n\nChange-Id: Ibeb9132e65706b50706249df3e8b6c455863513c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/da21de58f59c99137118932644ed97a1ae83f2d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/da21de58f59c99137118932644ed97a1ae83f2d1"}]},"branch":"refs/heads/master"},"c84e1a89622b63af6d9d227e0226be5f8d66d7cc":{"kind":"REWORK","_number":2,"created":"2023-03-02 17:05:43.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/35/876035/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/35/876035/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/876035/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/876035/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/876035/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/35/876035/2"}}},"commit":{"parents":[{"commit":"d209249239953af20d6c0925802481260ce8623f","subject":"amphorav1 removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d209249239953af20d6c0925802481260ce8623f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 20:51:19.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-02 15:59:15.000000000","tz":-300},"subject":"Remove specific handling of amphorav1 parameters","message":"Remove specific handling of amphorav1 parameters\n\nSome functions in the drivers accepted 2 different types for the same\nparameters: provider dict or provider data models.\nNow that amphorav1 is removed, we can remove the cases that handle both\ntypes.\n\nChange-Id: Ibeb9132e65706b50706249df3e8b6c455863513c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c84e1a89622b63af6d9d227e0226be5f8d66d7cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c84e1a89622b63af6d9d227e0226be5f8d66d7cc"}]},"branch":"refs/heads/master"},"43d43860201702f8cb2ac2cdb76aca857edcc0d1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-03-29 07:28:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/35/876035/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/35/876035/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/876035/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/876035/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/876035/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/35/876035/3"}}},"commit":{"parents":[{"commit":"aff294d8b5fedd30b54db7b2ec7bb5fe1a37fc8e","subject":"amphorav1 removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aff294d8b5fedd30b54db7b2ec7bb5fe1a37fc8e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 20:51:19.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-29 07:25:03.000000000","tz":-240},"subject":"Remove specific handling of amphorav1 parameters","message":"Remove specific handling of amphorav1 parameters\n\nSome functions in the drivers accepted 2 different types for the same\nparameters: provider dict or provider data models.\nNow that amphorav1 is removed, we can remove the cases that handle both\ntypes.\n\nChange-Id: Ibeb9132e65706b50706249df3e8b6c455863513c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43d43860201702f8cb2ac2cdb76aca857edcc0d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43d43860201702f8cb2ac2cdb76aca857edcc0d1"}]},"branch":"refs/heads/master"},"78ef74155bc0451770500340adea42eef6fe3198":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-04-05 15:03:49.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/35/876035/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/35/876035/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/876035/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/876035/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/876035/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/35/876035/4"}}},"commit":{"parents":[{"commit":"6c0515c9880b5310dd0647a5881233f07898d812","subject":"amphorav1 removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c0515c9880b5310dd0647a5881233f07898d812"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 20:51:19.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-05 15:03:49.000000000","tz":0},"subject":"Remove specific handling of amphorav1 parameters","message":"Remove specific handling of amphorav1 parameters\n\nSome functions in the drivers accepted 2 different types for the same\nparameters: provider dict or provider data models.\nNow that amphorav1 is removed, we can remove the cases that handle both\ntypes.\n\nChange-Id: Ibeb9132e65706b50706249df3e8b6c455863513c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78ef74155bc0451770500340adea42eef6fe3198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78ef74155bc0451770500340adea42eef6fe3198"}]},"branch":"refs/heads/master","description":"Rebase"},"324c3528ec7726fb6c69f70473b75b14357ff14a":{"kind":"REWORK","_number":5,"created":"2023-04-07 08:28:32.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/35/876035/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/35/876035/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/876035/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/876035/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/876035/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/35/876035/5"}}},"commit":{"parents":[{"commit":"6c0515c9880b5310dd0647a5881233f07898d812","subject":"amphorav1 removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6c0515c9880b5310dd0647a5881233f07898d812"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 20:51:19.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-07 08:28:24.000000000","tz":120},"subject":"Remove specific handling of amphorav1 parameters","message":"Remove specific handling of amphorav1 parameters\n\nSome functions in the drivers accepted 2 different types for the same\nparameters: provider dict or provider data models.\nNow that amphorav1 is removed, we can remove the cases that handle both\ntypes.\n\nChange-Id: Ibeb9132e65706b50706249df3e8b6c455863513c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/324c3528ec7726fb6c69f70473b75b14357ff14a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/324c3528ec7726fb6c69f70473b75b14357ff14a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
