)]}'
{"id":"openstack%2Foctavia~876036","triplet_id":"openstack%2Foctavia~master~Ica68aa13b7499bc752c9bbcff1fd713ca1081b05","project":"openstack/octavia","branch":"master","topic":"amphorav1-removal","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-04-20 18:36:26.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-04-19 08:35:42.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Ica68aa13b7499bc752c9bbcff1fd713ca1081b05","subject":"Merge amphorav2 provider doc into amphora doc","status":"MERGED","created":"2023-03-02 06:55:26.000000000","updated":"2023-04-20 18:37:36.000000000","submitted":"2023-04-20 18:36:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"876036-amphorav1-removal","meta_rev_id":"755ca03cce3a1fbe00566b53c86374b449eeaa1d","_number":876036,"virtual_id_number":876036,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 18:36:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2023-04-07 08:29:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-07 08:29:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-04-19 08:35:42.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-04-19 08:35:42.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-02 07:07:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-28 14:46:21.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-03-29 12:05:22.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-04-03 17:02:11.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"d611db6ef062d58d850b25dcccf30153b2fe65e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-02 06:55:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b40cda4f80bddaad5f2bc150e1de3054169ee259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 07:07:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1202f09f90f4ffe9f0f15b46e453a8b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9506fe0ebecd48e0b09e174359533ac3 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f329fc69890c4a20b71f9d09f719de66 : SUCCESS in 10m 56s","accounts_in_message":[],"_revision_number":1},{"id":"586f02d7c4ddbc0612fa9afb62db1c90389101a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-02 17:05:43.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"a852e2daa9621e85e4192617854295f958e77762","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 17:40:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b75ef6e0544421d928245fc834bce29\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2462b9b673ed4d20995983ee79a749e0 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe0b06100a0b4a049790800ab24f19bc : SUCCESS in 11m 50s","accounts_in_message":[],"_revision_number":2},{"id":"fde36493bd44def37619b3c1a388ddb1121bc119","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-23 08:25:37.000000000","message":"Topic set to amphorav1-removal","accounts_in_message":[],"_revision_number":2},{"id":"6e70a36f4b4a181e5cb7cc0851a80e279088fb62","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-28 14:46:21.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca4ac57c83cdd878e399a12b0961d028b7e02ea5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-29 07:28:58.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"6ea6a0777cd4831e5ffcb3e032c0a6f8526e5a1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 07:47:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd21049b709446fb946a40d227feac61\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07f59760b5a246d784c3a96246615b19 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0223c4d0d2ef4a0ab10745363a251cda : SUCCESS in 11m 07s","accounts_in_message":[],"_revision_number":3},{"id":"0bdf143ae5cb1338630964163f0dcac83d48a2b1","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-03-29 12:05:22.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6cdbbe0f528a6ae40e30f07ed62acc0c33d07321","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-03 17:02:11.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0eb13bf522e96d710ee58bb0275ca7e1ac58190f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-03 17:09:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7ddeeb33fa5d201b3f113dc7aa51a41143939358","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-04 10:28:50.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"fb4ce8416250a970fb60a9d8414cd4508572a4a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-05 15:03:55.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c1501c674675ccc2ed682f3c2dbcf8d0249065cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 15:22:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cc0168c804a42058af241a319210656\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/969a2f6e3d53421b8a90230ca2797d3c : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f92d0c7f64f4ed19cc3ac8209c45cd9 : SUCCESS in 11m 16s","accounts_in_message":[],"_revision_number":4},{"id":"88707eb6daab203f8ea813464805db835ee7c7d9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-05 18:54:13.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d42b180a84991caa459419e79d309dafde04eb3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-07 08:29:31.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:NO_CHANGE OR changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"changekind:NO_CHANGE\")\n","accounts_in_message":[],"_revision_number":5},{"id":"fd68e432e3f5268203f285feecfbc3be5380b420","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 08:42:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51a7d5f392ed48ac9e67940265e11d5e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f29bc94013cd4268ae34497efea92b88 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd171e968a464acfae57dbcd64a5c9fd : SUCCESS in 11m 57s","accounts_in_message":[],"_revision_number":5},{"id":"3e324175824a06c63d164931f19e179d7048fdec","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-19 08:35:42.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"0b0f42bd8d1d664492bd2560896bbaf5af66753d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 08:36:13.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5d4f4ad378c72524a78e94bbd6691e7c798a3e65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 08:47:23.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/43cac5853aaf4b74b33be471e7ba60a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/046817c8aadb40d9837f00e8d1c81095 : FAILURE in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcacdff9542a45d09c6e836007ad6bb0 : CANCELED\n\nWarning:\n  Comments left for invalid file octavia/hacking/checks.py","accounts_in_message":[],"_revision_number":5},{"id":"89b082dd5fa370f34797ef85902f232ec1c16301","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-20 18:14:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a4ba03a86c89c2e9e58d530361911c82b62bdedb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 18:25:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6596cbd4c1ef446184c1b76d30527837\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ebf6f939a214793a874704a4ce12023 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddeb0812084c4f78b83bc68932fba715 : SUCCESS in 9m 26s","accounts_in_message":[],"_revision_number":5},{"id":"26de0d2c8c75ea33a138d18e14f8c16c1896317b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 18:25:37.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f82540e0880f977b00b2e817216c746e542ec679","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 18:36:26.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4908e87019e8467fa8af2837698eb89d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f7b772e9de04dd3a10d52e016746cfe : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1642cb08d344f219a3e2c3959141988 : SUCCESS in 9m 43s","accounts_in_message":[],"_revision_number":5},{"id":"7ad7214a255d083197a99dfa7136c32c03d0e852","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 18:36:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"755ca03cce3a1fbe00566b53c86374b449eeaa1d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 18:37:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e69aa8b12e0749308df5cdb3be5045b9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a20af7010dc24984a6e868ee0700c78c : SUCCESS in 52s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3b4ef33f14e11de6868b683a3a42d99f62824993","revisions":{"97ec843ec7a5fe9d1f287570110c207df2db60c9":{"kind":"REWORK","_number":1,"created":"2023-03-02 06:55:26.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/36/876036/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/876036/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/876036/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/876036/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/876036/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/876036/1"}}},"commit":{"parents":[{"commit":"da21de58f59c99137118932644ed97a1ae83f2d1","subject":"Remove specific handling of amphorav1 parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/da21de58f59c99137118932644ed97a1ae83f2d1"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 20:54:10.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-02 06:28:23.000000000","tz":60},"subject":"Merge amphorav2 provider doc into amphora doc","message":"Merge amphorav2 provider doc into amphora doc\n\nChange-Id: Ica68aa13b7499bc752c9bbcff1fd713ca1081b05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97ec843ec7a5fe9d1f287570110c207df2db60c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97ec843ec7a5fe9d1f287570110c207df2db60c9"}]},"branch":"refs/heads/master"},"c6a7a5add827a723f6984d8a9c12bc90563647a9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-03-02 17:05:43.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/36/876036/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/876036/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/876036/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/876036/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/876036/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/876036/2"}}},"commit":{"parents":[{"commit":"c84e1a89622b63af6d9d227e0226be5f8d66d7cc","subject":"Remove specific handling of amphorav1 parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c84e1a89622b63af6d9d227e0226be5f8d66d7cc"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 20:54:10.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-02 15:59:15.000000000","tz":-300},"subject":"Merge amphorav2 provider doc into amphora doc","message":"Merge amphorav2 provider doc into amphora doc\n\nChange-Id: Ica68aa13b7499bc752c9bbcff1fd713ca1081b05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c6a7a5add827a723f6984d8a9c12bc90563647a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c6a7a5add827a723f6984d8a9c12bc90563647a9"}]},"branch":"refs/heads/master"},"3479bce008d4ef2cb184ae2291fbaa2f3ece9f39":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-03-29 07:28:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/36/876036/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/876036/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/876036/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/876036/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/876036/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/876036/3"}}},"commit":{"parents":[{"commit":"43d43860201702f8cb2ac2cdb76aca857edcc0d1","subject":"Remove specific handling of amphorav1 parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43d43860201702f8cb2ac2cdb76aca857edcc0d1"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 20:54:10.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-29 07:25:03.000000000","tz":-240},"subject":"Merge amphorav2 provider doc into amphora doc","message":"Merge amphorav2 provider doc into amphora doc\n\nChange-Id: Ica68aa13b7499bc752c9bbcff1fd713ca1081b05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3479bce008d4ef2cb184ae2291fbaa2f3ece9f39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3479bce008d4ef2cb184ae2291fbaa2f3ece9f39"}]},"branch":"refs/heads/master"},"ecb6aac2151a134f79946bc9154b910129832818":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-04-05 15:03:55.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/36/876036/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/876036/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/876036/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/876036/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/876036/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/876036/4"}}},"commit":{"parents":[{"commit":"78ef74155bc0451770500340adea42eef6fe3198","subject":"Remove specific handling of amphorav1 parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78ef74155bc0451770500340adea42eef6fe3198"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 20:54:10.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-05 15:03:55.000000000","tz":0},"subject":"Merge amphorav2 provider doc into amphora doc","message":"Merge amphorav2 provider doc into amphora doc\n\nChange-Id: Ica68aa13b7499bc752c9bbcff1fd713ca1081b05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ecb6aac2151a134f79946bc9154b910129832818"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ecb6aac2151a134f79946bc9154b910129832818"}]},"branch":"refs/heads/master","description":"Rebase"},"3b4ef33f14e11de6868b683a3a42d99f62824993":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-04-07 08:29:31.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/36/876036/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/876036/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/876036/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/876036/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/876036/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/876036/5"}}},"commit":{"parents":[{"commit":"324c3528ec7726fb6c69f70473b75b14357ff14a","subject":"Remove specific handling of amphorav1 parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/324c3528ec7726fb6c69f70473b75b14357ff14a"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 20:54:10.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-07 08:29:31.000000000","tz":0},"subject":"Merge amphorav2 provider doc into amphora doc","message":"Merge amphorav2 provider doc into amphora doc\n\nChange-Id: Ica68aa13b7499bc752c9bbcff1fd713ca1081b05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b4ef33f14e11de6868b683a3a42d99f62824993"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b4ef33f14e11de6868b683a3a42d99f62824993"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
