)]}'
{"id":"openstack%2Foctavia~876037","triplet_id":"openstack%2Foctavia~master~Ia1b1df9f3e7074dd71c7e4d1a94b307f7d878bee","project":"openstack/octavia","branch":"master","topic":"amphorav1-removal","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-04-20 20:44:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia1b1df9f3e7074dd71c7e4d1a94b307f7d878bee","subject":"Update amphorav2 jobs","status":"MERGED","created":"2023-03-02 06:55:26.000000000","updated":"2023-04-20 20:47:06.000000000","submitted":"2023-04-20 20:44:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876037-amphorav1-removal","meta_rev_id":"e8219995c3eb46110bb1cad08dad455382902381","_number":876037,"virtual_id_number":876037,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 20:44:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-07 08:29:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-04-20 18:15:27.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-04-20 18:15:27.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-02 08:23:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-05 18:55:13.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-04-20 18:15:27.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"6bb58819dab37b6d00ee8cfcff4d76d2ba3424fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-02 06:55:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbf9b12c2c260cb8f6e14744d1bf0ae769ff7fff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 08:23:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/050df7f876fd49eba334d63007bf37f9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35c69ecbf14b49d1962a6f5a317b07fe : SUCCESS in 12m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed3591864c6543e0b4933f623f4b50cf : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21bac29e3ea548c398bcf4e3244dd3d4 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df3f030834cd40f2bd636e275ecd8eb5 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d723fa6ede744ceca288252b048f897b : SUCCESS in 10m 43s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/cc1aeda74a574241b75ccfe452df13bc : SUCCESS in 5m 25s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7bd0aca3aaa3477a915e69019de92000 : SUCCESS in 6m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fca7ee83103d4cd89146699bd3a650ab : SUCCESS in 4m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/787b57c66bc1427993f04b745e9fd0de : SUCCESS in 6m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0f1b049b5685484d938ac8962a785d11 : SUCCESS in 7m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/784ccb591353491a89d2b70bf4746212 : SUCCESS in 7m 04s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c521fdd822a94fd2b6d61ca317781e1d : SUCCESS in 1h 09m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7353a3a3706d4c1092a2a0257b3e48db : FAILURE in 37m 00s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/867e949be6424c3da0584955caf9d122 : FAILURE in 46m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fcc1ffe46e594b84bdd10041c9d6234e : FAILURE in 35m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/37c2161e75f04179bd4a8004fabc6e47 : FAILURE in 37m 38s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/59ec4d3403574c279e89b6c0232245eb : FAILURE in 1h 06m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8eab02578d5546bca80d067eaa2b3e70 : FAILURE in 29m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/45dd3caa6c164543bcfc0ebccbb0d52e : FAILURE in 38m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0b94f0270d2f470aaa45b00dd4f07416 : FAILURE in 58m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/31309c5dc8f648688865671fa2e1f2f4 : FAILURE in 34m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/80b307d222304627a60516acdabab914 : FAILURE in 51m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58da641312c73daf10e3b1370d3d7dbabd1c8d9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-02 17:05:43.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"f3bd4acc17404544d8846a7e3c0ea68d3c7ebbed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 19:53:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a868ff4fea8146ed8201ce9724047781\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d6387e9ecc749bd8463ce0958e15bb7 : SUCCESS in 13m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b5349bd8ada4372b751dfd5b0e3186c : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e244ec7b256546b98835215cfc59cf5e : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b0cf56289a8464e89ed7c72164d5f0a : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc1ef2d41d9043e99f3bc5e1e9f297a0 : SUCCESS in 10m 38s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/986b008413fb45ef904b10bebab76012 : SUCCESS in 5m 06s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/9c2800f442b6422b91bbfd7b8471d262 : SUCCESS in 7m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4ca6a143ff7a4faf89dcadb1e954503a : SUCCESS in 5m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0d06fa6edb8346bcac8e66cde0125cc6 : SUCCESS in 7m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1996dcfa5c37486fa4dc258efd0423f8 : SUCCESS in 6m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0a8b770284a243c99a1f6cece476ccd9 : SUCCESS in 7m 09s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/140a9deec4444172ae0ccb076475d4e8 : SUCCESS in 1h 29m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c63493cd0a342208aa608bf34dfe74f : SUCCESS in 1h 50m 50s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/c7288a79fbad48749703e89b6ff8e6c2 : TIMED_OUT in 2h 33m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f4f3d3d7161a4708b0995927f837f556 : SUCCESS in 43m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/905bb537c7c74be98ab2f5d55f2e78f0 : SUCCESS in 1h 08m 34s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/db31fa8bf5af497da59fedc11e639981 : SUCCESS in 1h 24m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c6e2da981c6240418480259d75ec4bc5 : SUCCESS in 39m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/801fc5b9abe24d968fd5dfaf2a15c4d1 : SUCCESS in 2h 31m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/90686fffec41400ebe7a6984e7f6f088 : SUCCESS in 2h 37m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/57a3e341e08b438d8c00d92c7f24aa6b : SUCCESS in 2h 31m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/bde8d61d013e46bf8a679e88cd50558c : SUCCESS in 2h 28m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"38faeede4881e6699c1033a8135a533146afe482","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-23 08:25:40.000000000","message":"Topic set to amphorav1-removal","accounts_in_message":[],"_revision_number":2},{"id":"9e070fd17f280d30de8cfd0f74df80de4f6ec209","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-23 08:25:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0a9c2c726eeb846863e3989244fc198d7cab7082","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 11:41:03.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a6bb974833f43d4b8820a3c1597fd83\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1f15965c8404aa69f5e2552366651f9 : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3e9a8fb9a844de78d9567d7dc5632e6 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd49a45ab5804c3f9634ee69d899b4ca : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/713bb75a7ac7450c98aff37efa6878a2 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5314886e3d9b4fffb5fc2a5805720d31 : SUCCESS in 13m 46s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/105e9e1a3d6a4c1196809a150a44a759 : SUCCESS in 5m 08s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/45570589537b4838982653d06b69e56e : SUCCESS in 7m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4fcf551d6ef84d36a4bf3997904aaeed : SUCCESS in 4m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/010d0561709e445abf9fc6bdeff9e4fc : SUCCESS in 7m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/be30fbd6047f4148afc64edf44b45282 : SUCCESS in 7m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d3de9e67145b428ea822bea43c069d27 : SUCCESS in 6m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3e5fe5d511a448eca2c5586c43e0d0c7 : SUCCESS in 1h 31m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8e1c01f3f584494a9db8299a97632dd0 : SUCCESS in 2h 26m 49s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/e5fd844d4c1e4a20b08e8fcdf1f76032 : SUCCESS in 1h 59m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2b7a3aa11973478d9e93921931414384 : SUCCESS in 31m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d56d754e6e4b440a939c0677a5b2fb33 : FAILURE in 29m 17s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/efc604c146014096afba0e2c7f31bc9d : SUCCESS in 1h 09m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7282fdf2911e4de9954a347f6cb4acea : SUCCESS in 46m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4e002bcbe9a94f8d817cedf9686f7147 : SUCCESS in 2h 24m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/439c4aa27b464974abbecb0808bd714c : TIMED_OUT in 3h 03m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/585dde6b1f3e4f8b8bd102bbb0f905d4 : SUCCESS in 2h 04m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a0f0992f9b345f5901c30651c04eec2 : SUCCESS in 2h 28m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"364f4427dd312aeca1a9f0f4da69a4ed03ad7b0c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-28 19:25:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d29b10f0d4d8a64dc699f20079165db6481d44d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 22:11:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a064af2d817348bdb796d12f20cf1eae\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a1d5955ff1b4acca8e4f84e123069e4 : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cd2f4c07d3845ba9b1850144fbbab1b : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90d5ff17fd2440f2bf7258fe70caa7b3 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4857ff4e49f425ab40503bbc9b1e311 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a90da58ff2384b06bddbff3bfb01fe2b : POST_FAILURE in 10m 52s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c70cace3925240e5abb11de0e6f1e916 : SUCCESS in 5m 01s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/21c25fd2f0934099b3427e36d9214a12 : SUCCESS in 12m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/782200b9090f49b68c5e4455ab29cab0 : SUCCESS in 4m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1c0e80cce534459cab98a5096c15c1f9 : SUCCESS in 6m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fed4f776859a4e48bd21b826c66429e5 : SUCCESS in 8m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/94ef503d0e0643e5a056a8e46158e6ef : SUCCESS in 7m 44s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0faeece104f04d239f31869ce9815c93 : SUCCESS in 1h 06m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27fd25b39c5c483e916996907273c32a : SUCCESS in 2h 17m 22s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/c08882ea5cf146679c17cc492a8ba556 : SUCCESS in 2h 14m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7aa08744d8d64197b3590df6e0d2266a : SUCCESS in 38m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8c1ebcd24f644f909d42b79a7aeeb6e4 : SUCCESS in 41m 46s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/28ee39c86dd84536afcb59f39886b2e8 : SUCCESS in 53m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b4a15b1148b54fd4b6f85e85bbae65ba : SUCCESS in 34m 08s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ae13d927ca054635b5595235389be6df : TIMED_OUT in 2h 34m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5b08d147c20440bc9a675c52dd2c569e : SUCCESS in 2h 14m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1ed7c3ccda194e77a25370702ebf3cb0 : SUCCESS in 2h 00m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3c420a8ae5a474fa631cde62e7175b1 : SUCCESS in 2h 14m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"53c002b9bae77aeb1143dfc7c6fb41f9cce27008","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-29 07:28:58.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"9c69bda6784b1957e6ea158f0e88ca73c674f882","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-29 08:13:34.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.","accounts_in_message":[],"_revision_number":4},{"id":"a98afac44cf9a142f22c3046b645f9f96913f85d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 10:34:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8709bc6629a94a0081bd1d50e702a15a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0e5e177c8924d31bb03296a67a71cc6 : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15bb477ae17f4677b038ba8f929d80ea : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61a0455394d245aea1355a9577dd039a : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/332aacd3d33d4cd197152c9bbd0515ab : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17d480ff17a640e08544060f704a8e74 : SUCCESS in 12m 12s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9e3b153d20b34f73b82db869fa7b9f3a : SUCCESS in 6m 10s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4f0f97c48087410a8a52df5d321de1e3 : SUCCESS in 8m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/75be979e3faf43348d6944b1e1245aec : SUCCESS in 5m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0cbb4981e0ef4e958f6b474799530160 : SUCCESS in 7m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a6094584dbeb4589ac4c96cf10fe8048 : SUCCESS in 8m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bef3e6fa25024644a4960c3720c1f783 : SUCCESS in 8m 07s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5c052251fb2a4872b45e1a044b0e6e6c : SUCCESS in 1h 30m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/65619ecb6a104f8fb0cb78e8ebb98758 : SUCCESS in 1h 28m 09s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/ac4e69d26d344142911725ed99419d44 : SUCCESS in 1h 49m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3a8ec5d5f7fb4f82ace109587906595e : SUCCESS in 33m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/85cbe7258ee94bcfbbc838aa1bf83d87 : SUCCESS in 35m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c02dc22d99964aab90c0247b9f2b4de7 : SUCCESS in 1h 09m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/857fdd17e48c45829f00cd085569a900 : SUCCESS in 32m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e3a61e2f77954024b2ce1a0d0ae59b13 : SUCCESS in 1h 31m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/da6dda775bf0432f8fe7fef81dcde7a9 : SUCCESS in 1h 59m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f0a265834f1f44a1be7e8d4971ff9f8d : SUCCESS in 1h 47m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/de9f0bc0ec3343d087d7738b0c97f583 : SUCCESS in 2h 16m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"69ddc96a59e0569d5eb312775d4f55d41a937b08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-05 15:04:01.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"d674e063c44f37303671d43064b58142d71a87e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 17:50:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e803d0e4b894b798711c1302bf33f41\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54efa6c4965b41ba8789e1c72f16e4ad : SUCCESS in 11m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79767bc142574ba5b78c406f718f8e8f : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ada15467f6d8430eb436c62f72e78543 : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efb1efd3cbff41e5b082fd51573b1bda : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3181e56aa594456ea8b510bb6161c82c : SUCCESS in 10m 16s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d3e1d7bf31d84936ac062645b1cc326a : SUCCESS in 4m 29s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/dec07c9cdec04c6192903c660a80a9a2 : SUCCESS in 8m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a070c9dbf8994868a32c09080b438109 : SUCCESS in 4m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eb8811470be04f1ebe312afa72d6df73 : SUCCESS in 7m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/651e5de9c26945eb840cca3ce975b1f0 : SUCCESS in 7m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b029845c79264d2d93ecaa342c78622d : SUCCESS in 6m 52s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0323d93ec1dd48f387ccf6bfc92b1f0f : SUCCESS in 1h 31m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1b3112d532354f298348559695c9c393 : SUCCESS in 2h 00m 39s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/45a4d47dbe524990a4f7d9dff6e33bb3 : SUCCESS in 2h 18m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9c196fc0b4fc40789ee53b94c07e7d97 : SUCCESS in 37m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fac475c9551340b8aa5a6f1b0f1e9901 : SUCCESS in 57m 12s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b6835bbba15f4806a19f31243f7ae865 : FAILURE in 32m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd74bf48f3c44c91959f86991fdcae36 : SUCCESS in 43m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c8f5c7f596904bdfb0b3fe56cb3f4baf : SUCCESS in 1h 50m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ba971770016944cbaa127d284794dc6d : SUCCESS in 2h 42m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/00a8a0899e0f47fba3fa4906da667ff4 : SUCCESS in 2h 17m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d466fa583f4544c98e1722c54dec49dc : FAILURE in 25m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7dd18e8f17445ad30760a43264e28654d8b967d3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-05 18:55:13.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e08e4a5bbd12f3df6fb4c121cd2f46dd73a748bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-07 08:29:50.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:NO_CHANGE OR changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"changekind:NO_CHANGE\")\n","accounts_in_message":[],"_revision_number":6},{"id":"8cfa7c761908bd060f6838bce378c4a476080068","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 11:17:07.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b14f89c26ff4c15ac603a881227b893\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3522830c7424529a481aab1a36a64ae : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20c0bfa800534838940fe96d2d02230d : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2f18443f8254980b73e9caec84d59a8 : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e226f30d26a346afb1af1602a53999ee : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/628e9ed73a584c1ba41d2378d582e634 : SUCCESS in 11m 15s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a7de99ba54454299996ce7fb88b282fe : SUCCESS in 4m 52s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/380726e045814ecf93fb03e455b43c33 : SUCCESS in 8m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a74b198b5c53406e9d77ac8bd9094a2a : SUCCESS in 4m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/58c7f9856ea842c5839651a569bda8b1 : SUCCESS in 6m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/11eadfc9b6634e6c901b07da1d914d7c : SUCCESS in 7m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0a0bf0c5dbd747ed8a5cd63d8086ce07 : SUCCESS in 7m 23s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/53ea04d27d7c4890a1dfdcc4721a4476 : SUCCESS in 1h 29m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4b2d0cfdedd94871bc172b3206f4e1dc : SUCCESS in 2h 00m 36s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/bcae6c86bc72430fb34105bc23e147a9 : TIMED_OUT in 2h 34m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fe87bf9267734e46ad670bd6e99473a0 : SUCCESS in 35m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/10794eb441014cc9b7a280a1338b306c : SUCCESS in 57m 54s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5e21c39ee4384fefbca507ae920b7494 : FAILURE in 29m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ef8e6b3e4fa47b9b71a3a6fafa0e75d : SUCCESS in 30m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c86b8a052d9b4bae9ce40a818a778c70 : SUCCESS in 2h 02m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/df1ccc9f3c2e49e299c5ada6ef653367 : SUCCESS in 2h 27m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3b773f49b5ca451f8769b30cb72f9b9c : SUCCESS in 1h 43m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/92542fe5a1374299ae30a7487bcf3441 : FAILURE in 30m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a36fc6c06c19e41fd98a2801f116a898ae5b7e70","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-18 13:15:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cddcca62705b0f9d758f3e4a7aeef9688f5da4c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 15:53:18.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cc036b77a7b418fbc6a36edcf1d6332\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/581ff5a5095c419d86d6e56acbe96baf : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08c285698d1d4142b24657286e9a4aff : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14c8b417b45f4a478b508fd8c9094bd5 : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e5220460eef43ccb592c37aefecbed6 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0602b8eaa3443d09a5e243be8afac63 : SUCCESS in 10m 29s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1cc3f79cd3fb42d9ad1b5a3bb5882567 : SUCCESS in 5m 18s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5a283b5c4cec4a02b87432c860c40d7e : SUCCESS in 9m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/74481f0a71e347e6b8891a8b0ff08ed3 : SUCCESS in 4m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/19ec8d8920c741d8aad2d0ae7a14c2fd : SUCCESS in 8m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/daf2189071ae4d42b28b8880297a75bd : SUCCESS in 8m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/23f1c70956de4dcd8c70ce30bc901c0f : SUCCESS in 7m 07s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9fdd87f2c75a4df1a9ba5cd5ecac2c25 : SUCCESS in 1h 33m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95f021edf5864c4ba97f131e45ba0f34 : SUCCESS in 1h 44m 10s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/db083d0d60ed41eeb286f3e841a4c67e : SUCCESS in 1h 37m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7facf550dfcc4c5c860c9884c63e0c2b : SUCCESS in 34m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f578618823e24e068f4e8df6b564a670 : SUCCESS in 53m 00s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/544db551a37e46c0b56b45abd32900f7 : SUCCESS in 1h 04m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5f03f489c2544e3ba32357849b4b8312 : SUCCESS in 29m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a234ade633a54f1794fb2ad5ee90e5f9 : TIMED_OUT in 2h 33m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3793536cf40e48ea84b4299b54f2e339 : SUCCESS in 2h 33m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e178112de59849b3b26096b04d12eb53 : SUCCESS in 2h 20m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b45f206e8d144218916a04b94358a87 : TIMED_OUT in 2h 35m 00s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e5915421da6161f35aa3b10e9966e7bc5c58bd09","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-20 18:15:27.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"5ff17272e26060e3df4bded41210a114ff944aea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 18:25:56.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9ebe1593d7f82e82907adf31124b8e1140a285fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 20:44:55.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf74c5c7f38b4626bbf5b34da4d45ec1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cf2d2ec483d4c0290a956ec2889779f : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37a359e454064e69b7f91189b66d8bd5 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3365e76c982f4303bfc81e5e0ae36978 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a67582fef3f147f893ca9f42657c9059 : SUCCESS in 12m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b39580bad81d4e6286bc0ae53286495e : SUCCESS in 4m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/197ceead9b1b411699973571ab160a50 : SUCCESS in 7m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fcad884e54ff43729bf1808947112b92 : SUCCESS in 1h 24m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3fb041a617f44da6b9aa0f6c0d20aa62 : SUCCESS in 2h 17m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e1c55e250a484c84b70cb76d12251307 : SUCCESS in 43m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c83f088543644e259f0222f2a594a4cf : SUCCESS in 42m 39s","accounts_in_message":[],"_revision_number":6},{"id":"2024164f30c887e585e607bf5a0ffbfa51f96926","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 20:44:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"e8219995c3eb46110bb1cad08dad455382902381","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 20:47:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99330099aff44419ab3058883366140a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1533cac600e47a882c528507914e26a : SUCCESS in 59s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6dbe7b8d1c2da4633323d61b9d9d290eafb20e24","revisions":{"b3f908cbc397e8adb511e556e60511b7eb93f66d":{"kind":"REWORK","_number":1,"created":"2023-03-02 06:55:26.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/37/876037/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/876037/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/876037/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/876037/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/876037/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/876037/1"}}},"commit":{"parents":[{"commit":"97ec843ec7a5fe9d1f287570110c207df2db60c9","subject":"Merge amphorav2 provider doc into amphora doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97ec843ec7a5fe9d1f287570110c207df2db60c9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 21:07:29.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-02 06:28:23.000000000","tz":60},"subject":"Update amphorav2 jobs","message":"Update amphorav2 jobs\n\n- remove duplicated jobs (amphora \u0026 amphorav2)\n- rename jobboard job, make it voting\n\nChange-Id: Ia1b1df9f3e7074dd71c7e4d1a94b307f7d878bee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3f908cbc397e8adb511e556e60511b7eb93f66d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3f908cbc397e8adb511e556e60511b7eb93f66d"}]},"branch":"refs/heads/master"},"456577ea47dbfc0969afb609f0ee7d3c90e01710":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-03-02 17:05:43.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/37/876037/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/876037/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/876037/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/876037/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/876037/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/876037/2"}}},"commit":{"parents":[{"commit":"c6a7a5add827a723f6984d8a9c12bc90563647a9","subject":"Merge amphorav2 provider doc into amphora doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c6a7a5add827a723f6984d8a9c12bc90563647a9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 21:07:29.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-02 15:59:15.000000000","tz":-300},"subject":"Update amphorav2 jobs","message":"Update amphorav2 jobs\n\n- remove duplicated jobs (amphora \u0026 amphorav2)\n- rename jobboard job, make it voting\n\nChange-Id: Ia1b1df9f3e7074dd71c7e4d1a94b307f7d878bee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/456577ea47dbfc0969afb609f0ee7d3c90e01710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/456577ea47dbfc0969afb609f0ee7d3c90e01710"}]},"branch":"refs/heads/master"},"97a341e31c4bc8f791aa1b8d60a8746df4858819":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-03-29 07:28:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/37/876037/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/876037/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/876037/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/876037/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/876037/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/876037/3"}}},"commit":{"parents":[{"commit":"3479bce008d4ef2cb184ae2291fbaa2f3ece9f39","subject":"Merge amphorav2 provider doc into amphora doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3479bce008d4ef2cb184ae2291fbaa2f3ece9f39"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 21:07:29.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-29 07:25:03.000000000","tz":-240},"subject":"Update amphorav2 jobs","message":"Update amphorav2 jobs\n\n- remove duplicated jobs (amphora \u0026 amphorav2)\n- rename jobboard job, make it voting\n\nChange-Id: Ia1b1df9f3e7074dd71c7e4d1a94b307f7d878bee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97a341e31c4bc8f791aa1b8d60a8746df4858819"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97a341e31c4bc8f791aa1b8d60a8746df4858819"}]},"branch":"refs/heads/master"},"965ab04aca08b52ed023f03d67e8ddc63a6af5de":{"kind":"NO_CHANGE","_number":4,"created":"2023-03-29 08:13:34.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/37/876037/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/876037/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/876037/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/876037/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/876037/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/876037/4"}}},"commit":{"parents":[{"commit":"3479bce008d4ef2cb184ae2291fbaa2f3ece9f39","subject":"Merge amphorav2 provider doc into amphora doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3479bce008d4ef2cb184ae2291fbaa2f3ece9f39"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 21:07:29.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-29 08:04:53.000000000","tz":-240},"subject":"Update amphorav2 jobs","message":"Update amphorav2 jobs\n\n- remove duplicated jobs (amphora \u0026 amphorav2)\n- rename jobboard job, make it voting\n\nChange-Id: Ia1b1df9f3e7074dd71c7e4d1a94b307f7d878bee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/965ab04aca08b52ed023f03d67e8ddc63a6af5de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/965ab04aca08b52ed023f03d67e8ddc63a6af5de"}]},"branch":"refs/heads/master"},"6f347f10c27e14041b261a3820ff44e39d6c15f0":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-04-05 15:04:01.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/37/876037/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/876037/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/876037/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/876037/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/876037/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/876037/5"}}},"commit":{"parents":[{"commit":"ecb6aac2151a134f79946bc9154b910129832818","subject":"Merge amphorav2 provider doc into amphora doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ecb6aac2151a134f79946bc9154b910129832818"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 21:07:29.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-05 15:04:01.000000000","tz":0},"subject":"Update amphorav2 jobs","message":"Update amphorav2 jobs\n\n- remove duplicated jobs (amphora \u0026 amphorav2)\n- rename jobboard job, make it voting\n\nChange-Id: Ia1b1df9f3e7074dd71c7e4d1a94b307f7d878bee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f347f10c27e14041b261a3820ff44e39d6c15f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f347f10c27e14041b261a3820ff44e39d6c15f0"}]},"branch":"refs/heads/master","description":"Rebase"},"6dbe7b8d1c2da4633323d61b9d9d290eafb20e24":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-04-07 08:29:50.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/37/876037/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/876037/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/876037/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/876037/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/876037/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/876037/6"}}},"commit":{"parents":[{"commit":"3b4ef33f14e11de6868b683a3a42d99f62824993","subject":"Merge amphorav2 provider doc into amphora doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b4ef33f14e11de6868b683a3a42d99f62824993"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-01 21:07:29.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-07 08:29:50.000000000","tz":0},"subject":"Update amphorav2 jobs","message":"Update amphorav2 jobs\n\n- remove duplicated jobs (amphora \u0026 amphorav2)\n- rename jobboard job, make it voting\n\nChange-Id: Ia1b1df9f3e7074dd71c7e4d1a94b307f7d878bee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6dbe7b8d1c2da4633323d61b9d9d290eafb20e24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6dbe7b8d1c2da4633323d61b9d9d290eafb20e24"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
