)]}'
{"id":"openstack%2Foctavia~876041","triplet_id":"openstack%2Foctavia~master~Iccae6ec561603653183911146a638272df631e92","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-08-01 10:23:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iccae6ec561603653183911146a638272df631e92","subject":"Don\u0027t update the management port when calculating delta","status":"MERGED","created":"2023-03-02 09:08:11.000000000","updated":"2023-08-01 10:24:25.000000000","submitted":"2023-08-01 10:23:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876041","meta_rev_id":"5ecf8aaf58cfb70b76409f6a9058b80dfafea5da","_number":876041,"virtual_id_number":876041,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-01 10:23:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2023-07-25 16:15:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},{"value":1,"date":"2023-07-28 01:42:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2023-07-25 17:11:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-01 08:20:09.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-01 08:20:09.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-07-25 15:56:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-01 08:20:09.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-02 11:35:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-25 16:15:45.000000000","updated_by":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"reviewer":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"state":"REVIEWER"},{"updated":"2023-07-25 17:11:13.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-07-28 01:42:49.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-08-01 08:20:09.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"0ad755f24dfe75157b0f676e151cfe0773efc957","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-02 09:08:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4fbd536f3eb89573c24d24e51ed357f5c7a17116","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-02 10:54:11.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"630a83d5901333dab5f445c5cb5af50c7f9a4597","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 11:35:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/534028e82c56415f9c9ad117794fe733\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ce5023d50134721b34bca2155a7dc2c : SUCCESS in 10m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c93368257a34d578c33838f675ec936 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65d809ffe6124c2aafda4beadbfb86c3 : SUCCESS in 6m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/327519aaa8724bd1a05a2f3ef09b10dd : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74d27dbc77e24a95ab277cb1a6e11e0e : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d1cc021a51848dbaebe6e53b01c18fe : SUCCESS in 5m 07s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/819a1751f53448ecaa65ae1028b95e65 : SUCCESS in 5m 24s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7535ea3408e34137a66af7cc0d65d507 : SUCCESS in 8m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0e43c09aa3954407869e2d8ab34831ba : SUCCESS in 4m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eb052078156f4d078b850b207383fe97 : SUCCESS in 6m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/708c6a26437b4cc1995211eb8a2dab8b : SUCCESS in 6m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/10aa8577c55a41ad8a44c105a55621e7 : SUCCESS in 7m 23s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/309594b8180748c8aa7a8f44bc94476e : SUCCESS in 1h 35m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7ec09e7440c24bb4b7f6a46760c8a643 : SUCCESS in 1h 52m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7d50ee30d103487b83629bae2cc223ca : POST_FAILURE in 38m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a7e11cfcfb3f43f991ec1ba6ee620dae : SUCCESS in 41m 11s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/72fe6ebf504a4d5e883326c57b39e95a : SUCCESS in 1h 06m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3a261ee255ca46ea85d34b7762f1e092 : SUCCESS in 34m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/765a7edcab5e456aba83b4fe70dd7350 : SUCCESS in 1h 58m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a203339ca2454684abf14512696a9f84 : SUCCESS in 2h 25m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/13f9bfb0aa5444acb413324473aef67b : SUCCESS in 1h 47m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/872550434e3f47c28c088d5f509207f1 : SUCCESS in 2h 09m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/99d48ceda12f46f8879d4b9ca4a82f3b : SUCCESS in 1h 35m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3820f12b70f4b54a7e2cd74a7796774 : SUCCESS in 1h 57m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"199196d45698d32703fbfb2e69fe4a9c3a2944a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-25 15:55:09.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c8cadf64e15da28ff79ccc22fb9f051a76c17934","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-25 15:56:27.000000000","message":"Patch Set 2: Backport-Candidate+2","accounts_in_message":[],"_revision_number":2},{"id":"0d06a5bfd0f625718507c68fd929e2316deafdaa","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2023-07-25 16:15:45.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c31da1b84f1b8c6dfd33d8adc0e0c3c69e2361b3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-25 17:11:13.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a89eb8ea678f03a7ea798ebfa3d76bcd6d267fa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 18:59:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/30d645050f7440a191c64b821af57a04\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a0d648f8ae14b1da4c5d7a54be39100 : FAILURE in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45b6e689ce114f64bf4ac97c9676803d : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/860feb8106a7428484b3ece5579085c6 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef88b2e9cf6c4dcc9370d8d4901b0833 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c266a13c57b148de9dafc3229063d4f1 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caaf82b4e1ff45c89c40fbb536a77a7e : SUCCESS in 11m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e075fa6a630746e9b65d20f92c41d952 : SUCCESS in 7m 02s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d589156cfe324e809a45d4dd0511e1c1 : SUCCESS in 6m 27s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/505c8185442f44a6923911fcb394ee1d : SUCCESS in 10m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bd495f5fb84d449996adbbb96c57c3d1 : SUCCESS in 5m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5fd7cac0cfe2460bb1bae2e2786c3b10 : SUCCESS in 9m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/af91792436614ea98f7e4c5ed8211663 : SUCCESS in 8m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b8553fdf17da4b7c845950499e5d1cfd : SUCCESS in 8m 48s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a6812b13e1e447198a57afcd47e832e7 : SUCCESS in 1h 36m 14s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/50ba547f70794db690f03f1264cf978f : SUCCESS in 2h 02m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/eb0acdb1efbe4ff0b81453e83a7c3c5b : SUCCESS in 1h 30m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/82fd4787bc97412980bb26b3e007537c : SUCCESS in 2h 05m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/759bb4d337964d2ba5d3f9cdd0cbd01c : SUCCESS in 1h 20m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7aa55fba60c84b7ea252971b94d38734 : SUCCESS in 40m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/274da49400f047da8e45f4151d1855f1 : SUCCESS in 37m 00s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/58019752d8be4688bd16df2013e179b6 : SUCCESS in 1h 18m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2cfd072a3e5942bf903c2fe8bf1a687d : SUCCESS in 35m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ceb12f64e6d74d8cbaaee7f8bfdd8d87 : TIMED_OUT in 2h 32m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0e07c47fb50f4b3ebcb6bf7ef9915d39 : TIMED_OUT in 3h 02m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5ffb27ad8ba04e018cc082a7965d7c03 : TIMED_OUT in 2h 32m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/43ded61c8d174280b825dcc81e6dee81 : SUCCESS in 2h 31m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"63b0e992fe37ae1526a5ea3e95fc9b831d162fac","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-26 09:41:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b0a2eda05a86e92d5063485eff49afad737385a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 12:45:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cb9879ff69746368debffad25eebb12\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68207e8a9af444fb9684c250741fa1e6 : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaf7b79a17604550b56a4b08e4363a98 : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61b22feeeefb4e20b38db207662f0619 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d598853392ea4e4c8d006d39d4207914 : SUCCESS in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bcb48f68df5460890e2774df5a518df : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ee079c062af48a4bc21c6103d755581 : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f4e41b5fea740f2af5d50684ffd33a0 : SUCCESS in 6m 22s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ea55535e4a93423db9973b10c7e361c9 : SUCCESS in 6m 20s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/15c589988a4146d7a6942bec4fc11514 : SUCCESS in 7m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cf2ca63e0ee04d45964b6d4a5a250290 : SUCCESS in 5m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b38babf9548e4c6b98440e6362be7b5c : SUCCESS in 7m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ddc27d4a64ab44d59a047c271d7d3365 : SUCCESS in 8m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8ae0697aa92a49d58a284bcc21d72cbb : SUCCESS in 7m 13s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4403fb252b924738a5cb9c843ec6d46f : SUCCESS in 1h 24m 34s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9d7ac232d6764f86b7c5794d064cce12 : SUCCESS in 2h 06m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6ecdf3616afa402099997598ee0d1c9e : SUCCESS in 1h 22m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a6081b13a17343279b458372b371d04f : SUCCESS in 2h 12m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e9ab0fc327af4b0d9473f7ed86fe9cad : SUCCESS in 1h 13m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3145f7b237954b3788375aca40b4c11f : SUCCESS in 41m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/74a30fde2cae4bda9b45dbd74e1c5148 : SUCCESS in 37m 29s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5efb0594069d48129cce1deeeb625455 : SUCCESS in 57m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/797e680c32be480b87601309cde8cbc2 : SUCCESS in 35m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6059e4c1ff1741a0a7f6cde536cebdad : TIMED_OUT in 2h 33m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0d19b6da8ab2433e998ce195be0d34a0 : TIMED_OUT in 3h 03m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b11724108f7c4b239527a4a4085244ec : TIMED_OUT in 2h 32m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/68601980c47b48c7859046e00081ed87 : SUCCESS in 2h 34m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e6025140761727e1824479cb1bbe34ba374fe6ce","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-07-28 01:42:49.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9ea3366b2d38421712050f0f39bdadfd8913f96a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-08-01 08:20:09.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"209d75fa1e244ed050e719a3a44f2b7ebdebe94f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 08:20:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0938d07c3098f5ed85fa6c7ddedad60a69cf0f7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 10:23:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/777057b805ab4932943c1ce820de4e2c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06621bf536144a19ba95405d6c5657bc : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14fba773b9af4b60ac9d7c03e4b7ac28 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/560c99ecb2334c0481df05d8d5cc29cf : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/311fc38694114fa0b1d6228de12b6668 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/467ed3c47e5e43349faeec9e0288c186 : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d8831c952084701a756af788b96e394 : SUCCESS in 4m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d00eff507b47488ea7585d70b94814a7 : SUCCESS in 4m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/01007f41f51a4345a60712096959b57c : SUCCESS in 8m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d2d920d646af4606a9bec93ec2838003 : SUCCESS in 1h 45m 42s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/68d960d990c94a89a79fd7fe44e149b1 : SUCCESS in 2h 02m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e69b9ff82b4b40b099eeba04782d710c : SUCCESS in 1h 17m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c94c6e783f9041e3b8cab21238747895 : SUCCESS in 43m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8c7062c2010c4a08846f1221e2e41cd6 : SUCCESS in 38m 35s","accounts_in_message":[],"_revision_number":2},{"id":"4216f1e4b0832f051d0d07999c9276565f2ebf26","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 10:23:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5ecf8aaf58cfb70b76409f6a9058b80dfafea5da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 10:24:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30f83cee5c004063ac44b7586fa60f29\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6b1aaca39aa4c51856d9147145191ac : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1dea83a33ef4e21bb01872a1afd34a4 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8ccc0d01764c234ce9d3dc0191b5a0952111e2ef","revisions":{"6ebc96dd25a73db13b712615ec286501c5e4645d":{"kind":"REWORK","_number":1,"created":"2023-03-02 09:08:11.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/41/876041/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/876041/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/876041/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/876041/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/876041/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/876041/1"}}},"commit":{"parents":[{"commit":"3475a4b3c2b0614d4a6f1e51d6b08d3abcc5f7ca","subject":"Merge \"Deprecate user_data_config_drive option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3475a4b3c2b0614d4a6f1e51d6b08d3abcc5f7ca"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-02 08:30:52.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-02 08:42:59.000000000","tz":-300},"subject":"Don\u0027t update the management port when calculating delta","message":"Don\u0027t update the management port when calculating delta\n\nCalculateDelta should be used to get the list of subnets/networks to\nplug (or unplug) into the amphora for the VIP and the members.\nIt should never reconfigure the management port.\n\nStory 2010623\nTask 47564\n\nChange-Id: Iccae6ec561603653183911146a638272df631e92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ebc96dd25a73db13b712615ec286501c5e4645d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ebc96dd25a73db13b712615ec286501c5e4645d"}]},"branch":"refs/heads/master"},"8ccc0d01764c234ce9d3dc0191b5a0952111e2ef":{"kind":"REWORK","_number":2,"created":"2023-07-25 15:55:09.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/41/876041/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/876041/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/876041/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/876041/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/876041/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/876041/2"}}},"commit":{"parents":[{"commit":"12d8e0de5de9ca754859542bcffd2311f5145d61","subject":"Merge \"Update doc for the OVN provider\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12d8e0de5de9ca754859542bcffd2311f5145d61"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-02 08:30:52.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-07-25 15:50:27.000000000","tz":120},"subject":"Don\u0027t update the management port when calculating delta","message":"Don\u0027t update the management port when calculating delta\n\nCalculateDelta should be used to get the list of subnets/networks to\nplug (or unplug) into the amphora for the VIP and the members.\nIt should never reconfigure the management port.\n\nStory 2010623\nTask 47564\n\nChange-Id: Iccae6ec561603653183911146a638272df631e92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ccc0d01764c234ce9d3dc0191b5a0952111e2ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ccc0d01764c234ce9d3dc0191b5a0952111e2ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
