)]}'
{"id":"openstack%2Foctavia~876480","triplet_id":"openstack%2Foctavia~master~I16430fa52db02e7445203994220673c1037d764c","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-07-07 11:23:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I16430fa52db02e7445203994220673c1037d764c","subject":"Fix upgrade check not working","status":"MERGED","created":"2023-03-05 22:18:44.000000000","updated":"2023-07-07 11:24:45.000000000","submitted":"2023-07-07 11:23:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876480","meta_rev_id":"586628c2f50bd77e3711ce5c975c4385596908d9","_number":876480,"virtual_id_number":876480,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-07 11:23:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-04-14 09:32:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-24 09:23:34.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":1,"date":"2023-04-24 09:36:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-24 09:23:34.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-04-14 09:32:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-06 00:50:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-14 09:32:05.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-04-24 09:23:34.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-04-24 09:36:48.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"89cd48224705af0a29fe6efe4f359c322872adca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-05 22:18:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f6fecef097613dd8f69931788bd39be275877d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 00:50:31.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1c1321650f145528578d9ff7b15ff8b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6015903572b480eb8b469572eda225d : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b45f4b49e3aa4c509e5e272ed02e297e : FAILURE in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9911090f38d46d29e8e3dc1a8fa3bf9 : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da8d7cc11a6c4b908d646188806db511 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dab2f1ebb6704bb1b9cb12b1e96a0912 : SUCCESS in 11m 30s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f2d8402ca6fc415485fbbe621e3f2d23 : SUCCESS in 5m 37s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/efbda08b52cf475185a800a54352aa4c : SUCCESS in 6m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ea348cb3e6424fcabe28caf68c45c925 : SUCCESS in 4m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bbdf2545aa07404b9853df32e98e6cef : SUCCESS in 6m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e17da9b5fdbe430aa69cf47dee7b8d68 : SUCCESS in 6m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4b2217dc530f49d9aba8b4bf2cd8c5b6 : SUCCESS in 8m 03s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6c2015207d334fe69147d2d9a519e308 : SUCCESS in 1h 05m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7409da61f3c94668ab91abb4afd10b27 : FAILURE in 2h 08m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/28960385087a43018ed62f694a4c8678 : SUCCESS in 34m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/82ee8f66bac54763a67bb982538dfc2d : SUCCESS in 41m 18s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a008bbab4fde45b4aecb8c7d5896c9ac : SUCCESS in 1h 00m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2de494e083ce4ece9d5fee73402cee6a : SUCCESS in 34m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2eb1cceb573745a3b12c1155169f07ef : SUCCESS in 1h 59m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e4a3ff2a0aff49de883e5cf927f2a135 : SUCCESS in 2h 24m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ed75655f846d4c43979e09fa6cd87d40 : SUCCESS in 2h 12m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1546f96248674f388537f90629e49ec0 : SUCCESS in 2h 14m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5aae4e5e70e340a08b40c40b4665538b : FAILURE in 2h 13m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5dc4fadd45ef401bb6b9166f2f6c7391 : SUCCESS in 2h 04m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ab5c4e24bc57c1fbbb56d09a99359de8d6d1f7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-06 15:30:25.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ab4d8b622c486f98cb1a80c95d2e92cd29dd7e5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 18:07:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d6e33607b884a8c8e0f468da43d55ef\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/582d4f539f0944ef8a7b18e8c00d4d2b : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68268e1e3ce94665afc7805285984fa0 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c5012c774464ef3be3ada885e4f4be1 : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3db5d8d5a72946e69f76c32ab8dd156d : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aa273881594480196d219e30206d94c : SUCCESS in 9m 59s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e248c0f60cc245adb3037f39b7bd053b : SUCCESS in 4m 31s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e9bce827ad03458ba2036abea5c50360 : SUCCESS in 7m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b9e104f59f954799ae965e050a7783b5 : SUCCESS in 4m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9ec0b86c432c482f87694e2d6b121946 : SUCCESS in 6m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c961d6fc7e2946d9860f6587dcba8be6 : SUCCESS in 7m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/606bcb7a0a5a43e8888af700af4ec093 : SUCCESS in 7m 38s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1921da4a181844dfa42d7b19cd8d8d37 : SUCCESS in 1h 29m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/16569b7675ed46f9b7accc0323538c58 : FAILURE in 20m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9349e1877b074963b336051c8ca50f27 : FAILURE in 21m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/46fa96bc2cee4467af734662706c4597 : SUCCESS in 37m 08s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/749bafd880734984b39172ee2833484c : SUCCESS in 1h 16m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/931b3883c2ef445bb879b3aab2367304 : FAILURE in 20m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c2568c7429e342d3b9cd3b34a25e9315 : FAILURE in 21m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d6ad67ef767d48c0b34c251a319d13a4 : SUCCESS in 2h 28m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9b3ad60406714a2e83079716d245ec5a : SUCCESS in 1h 57m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/312cb5626d5e456a9fd80f0a3fe4612b : FAILURE in 20m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c55efe5bcb9347358c5841803da8f467 : SUCCESS in 1h 36m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9745439cc4dd4805830a5ace823eaaac : TIMED_OUT in 2h 35m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fb035b028666d6f59dcf2810653562d5579ca232","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-08 21:44:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fe32f7cd71cd0c4b8c51306543caedcd72076737","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 00:38:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/350bcc066b224fd69e350df13d9945ef\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/083c2af090524cca8a56badb3fed9d08 : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1bb38b7fd164dd99e1fafeed6ead2c2 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd37c1543d354c118de7a24aa649bfc3 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa2aa0404db1430694c954a9c79c1db8 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26c1d5a4277445aebf7b3d405efa8db2 : SUCCESS in 9m 48s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/dda606294e2148b0a5ca4d5985b61cb2 : SUCCESS in 4m 43s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b0f13a4b0445419b80e92fe6cb30a8e7 : SUCCESS in 7m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9958736225764a8bbda8fd86b0158253 : SUCCESS in 4m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0f8521f06b0f439ea268fbf4d8dc4ce2 : SUCCESS in 7m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/816b7a3775ff4322934b5b617c0f4912 : SUCCESS in 7m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2d30aeaf1b9a480fa343b5524f84b470 : SUCCESS in 6m 39s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a77989f5359942eeb256c227f34de860 : SUCCESS in 1h 22m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/757f42a72b324af3b1cedc05f4866e71 : FAILURE in 18m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fde27a221c254c1cbe6b848274bb9b46 : FAILURE in 21m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f756701ef23f4bdaa140ea1c96b58863 : SUCCESS in 37m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/678060d99ca94ac7bb8050dc7cbfe479 : SUCCESS in 1h 12m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5640dc3e9ba140308a2a19920e92351d : FAILURE in 18m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/096b2d3cf64f40109d7b9436a127aadb : FAILURE in 19m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9d3627ea91bf4419b48a1a56c31f4e26 : SUCCESS in 2h 46m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4cfd03d8ca7f4eeb8c88de0fcee22f5c : FAILURE in 20m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9f19dd029c2a45ea8fa27ecf94ab07ab : TIMED_OUT in 2h 33m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7e0c5fd1d1bc4325a3f76d94c6528d2c : FAILURE in 19m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef1daa58041f43e4b0f3fe5f476d20b8 : TIMED_OUT in 2h 35m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0970c2eaa92037fce4871dbf17415184a24a32a9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-14 09:32:05.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b72376d96318fdc4882bbc2f050c9e58226358b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 18:34:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/becc5da491744b63bca70cb28bc0e484\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02a800e64a624c1f9ad2c2a3e012df8e : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00c52e7bbe0c4e67b01001f4f825dcae : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36079c4e94064159b0182c4eb74f3fd5 : SUCCESS in 8m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4b5acd7322743c69ef288d1f2e328c2 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7972f2821d464b1a93d4bd2589f08ce4 : SUCCESS in 12m 06s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/80e0268f9e78438d9d7c37b11b866e8a : SUCCESS in 8m 06s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0612770f518044a2a5ddeefc33a6377b : SUCCESS in 8m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ebe022e5d6104062a27123b0d8ce6f3b : SUCCESS in 5m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f19f18773c754e8ba303673eb2d9a153 : SUCCESS in 7m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c54b8461433f4269a0ce2281f1606403 : SUCCESS in 7m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/358185c44a554409ae22ac1a3ff892a5 : SUCCESS in 7m 35s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5b288d0a120049ff93c57af69bf52838 : SUCCESS in 2h 02m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c5b2d80aea3848beb60d6eb255e690bf : SUCCESS in 2h 07m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/901bd73f24004223b8ada762ed044eef : SUCCESS in 37m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/972c66293fcf444c9b7e40b3cce2d2f3 : SUCCESS in 39m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/88ca9cae59274a2e91f63e2cbaa620c7 : SUCCESS in 58m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e1bdf3d618384f06b38cf0a70332e790 : SUCCESS in 30m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ad22894194e745779be99254330387cc : SUCCESS in 2h 09m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f5f870a36cd0461aa5bf02a0cd6d65d5 : SUCCESS in 2h 18m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5afd543910654de0939578e72c8ed73c : SUCCESS in 1h 37m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bc0b9e3fd47a474f90502e2775622289 : SUCCESS in 2h 08m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/95610e89edec4e21a1cf37465d4259f7 : SUCCESS in 2h 22m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/6467670e134244e58d7952c86defda90 : SUCCESS in 2h 00m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aafb0089b56edd147da571dcf5c7153fcf77ed1c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-24 09:23:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c328f8628e524d60d7b35aaa928e55d27f06304a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 09:24:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"efc5fc901820d03610ba003096e48d84745a3a93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 09:35:16.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4516d11f0266401bb681cba651e4fa2a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3df513bb473345f6bb51de4e0c028d5e : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92279103fe3043c69810d817bee93313 : CANCELED\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94b2ddd63db94af9beb53556bda6c6c4 : CANCELED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b89956c6fad4fcaae198d0562d8fb0b : CANCELED\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d1d1411d9de24b75a04a4e91f7dbe726 : CANCELED\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/72d0e7488b044c99bb98e8bd6b4f0568 : CANCELED\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/58e1d4c9a6054c509f0aee01c4051ed9 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/262036042f674cf6be71d9e313122284 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e905949999ed4053939bb3035c9a1cf5 : FAILURE in 5m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dc936a16f30c46c2a47c70703d5ba160 : CANCELED","accounts_in_message":[],"_revision_number":2},{"id":"9d3d06d4c13e47de72e0a3545509ac99eb26d6e0","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-04-24 09:36:48.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"01a5b680a5f04511ebb4126cd992e457cef93555","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-05-05 06:09:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2c3c9b41150da094c67d2857411d049ebe68b0ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 08:58:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eed10ef8f3544df1b206f234833bf7c2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31dd91cbb5fc42a0a79c710045ae4977 : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afc70c9b2a544807b90ed8bb308f2a68 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a859ac33e37f4a568ea0afe34eb14ba6 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/334bd52dd15942eb9d7a870d60d4c295 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05efd462e8d04e1bac0f8e9a3cc4959f : SUCCESS in 10m 45s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3ab7a028056744dbbc7b17317bf1c471 : SUCCESS in 5m 01s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1249b1bec4504efc967258a0d05b9175 : SUCCESS in 7m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/735feb5496e84621840950d15ba146da : SUCCESS in 4m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/549f879837744c8f802b5c568405c324 : SUCCESS in 7m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/36a0baf8c0c849ff96ae80be92bf28a0 : SUCCESS in 7m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a6c622a10d44478f858c926f3c2c5df2 : SUCCESS in 8m 12s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0103a9e94077470e8aeef8a510adf645 : SUCCESS in 1h 52m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/41431dec4d2b440d9196d7245f38f8b2 : SUCCESS in 2h 19m 02s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/f585428b1ccb40ffaf5bab2002ca1aab : TIMED_OUT in 2h 34m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b14ea211ecfc45018f90f18e043251fc : NODE_FAILURE Node request 300-0021143243 failed in 0s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c1f4a8a78d10460f9d698d0f9ad6e6af : SUCCESS in 46m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/14085ee6a7624497969c119d538b7b3b : SUCCESS in 1h 12m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/58ab90f39ad74ca88983a603c6a390b6 : NODE_FAILURE Node request 300-0021143246 failed in 0s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6034d715137a4b5b826e2d74c4958a13 : NODE_FAILURE Node request 300-0021143247 failed in 0s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0127e1b69a4943d1bd3fd0ddd1b04f21 : NODE_FAILURE Node request 300-0021143248 failed in 0s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a26b217cabfc4aa98ba229a1f5c869df : TIMED_OUT in 2h 34m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/082419f008524c8e9f6cef14b87de226 : TIMED_OUT in 2h 35m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1e8b94b667acbda23fef92e1b610631559d2d662","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-05-07 17:18:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6ea9d96bdf88dbee4a9a4e9b61b32dad90a42051","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-07 20:31:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a30347dd55904bd399dcd9f111d94e5a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d46e681bb134807af9addc616c24621 : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b36bc74c7fa548beba8ac6356f1f33d4 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/862da30dab114e4d938d697acc4932fc : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4553565205074ce4bde3fd5774c99299 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2aa4ff64ef443ae80e8f0ce0d9d131c : SUCCESS in 11m 10s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1428ab3f33a6406ebfcd55af12f71979 : SUCCESS in 4m 46s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a62609487e10474ba7d95c6e30269b54 : SUCCESS in 7m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/101e2bbd316f47cb858276284997312c : SUCCESS in 3m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/be9b5301e23a473e827dbea97e4c9273 : SUCCESS in 6m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4413c5d3d0124a2ab33a7a0269dec193 : SUCCESS in 6m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/70860b18be84481caf4bc1c70fe409f7 : SUCCESS in 7m 45s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ed90de3cbec34c37b433885753b7db8b : SUCCESS in 57m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/db2a7e4e67ea4eeb849ec78cb5d06293 : SUCCESS in 2h 09m 05s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/9b28e11ea70748e1958aa78d34e7a35b : TIMED_OUT in 2h 33m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cb70b287471e4e55aa2d85a56f9efe67 : SUCCESS in 45m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3826446da6f9418284e60d600ff74c36 : SUCCESS in 52m 25s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5659a63d59cb4afb8c4c5b44e0bdb04d : SUCCESS in 57m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a26b761868e437a9cfeb1e37ea9ee6f : SUCCESS in 35m 08s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d981a93f2a854d6db0299c3beedd8baf : TIMED_OUT in 2h 33m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dc0c548dc07342aab49afb8eb545790b : SUCCESS in 2h 55m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/594d0e3587b44ca1bae991f4ad64b5f0 : TIMED_OUT in 2h 33m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2e1d8ea5e284d30b7a49ea6a137800d : SUCCESS in 2h 20m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2f193942c69db8cd019a7f3cf55267230fd42f15","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-08 07:06:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"41051a8d8772cb473e8b3532f5530ce7f834561d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 10:13:26.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6fa7823e98ff42b7bc4b36a75e08b225\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b519ae82f3e47b7a5798f4f114f7ca5 : SUCCESS in 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cee53a6a067345d2af9c1305b8ea252f : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b936a3499a5c46b59390be932d03ecb0 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/857b2e4e92874f97b6065a1d1d324099 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb5da4c4bb904911a4aa38b4bc62da50 : SUCCESS in 13m 25s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/829e49979a9d49e1b4809f1a6b7e4bd3 : SUCCESS in 5m 49s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e49667b2d28a4c209b7eb831a62f6b35 : SUCCESS in 8m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9ca48bc78df342a6937419ecb7cab0c1 : SUCCESS in 4m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/64aad239d9aa4a50aa9e80147cbb8ea9 : SUCCESS in 7m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b5e34f4a75a34a75bc2bc68d1bd96f8f : SUCCESS in 7m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/820beeeec9c5472bb1de4268d521f6d9 : SUCCESS in 7m 22s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a4da4327af6744c9a4ea8cf6712e87bb : SUCCESS in 1h 32m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5de2c6d5d0c4c5e94eb65044320b99c : SUCCESS in 2h 26m 56s\n- octavia-v2-dsvm-scenario-jobboard https://zuul.opendev.org/t/openstack/build/f42dabe4bf144fc0ab58d27b6193b83c : TIMED_OUT in 2h 33m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/76e0e495e2c7449fa015b920f2bd2438 : SUCCESS in 42m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4bfbedb13cbf472fb8ef72f38c7cc161 : SUCCESS in 50m 08s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b44206f88ec04b1b81c9351f62b5b8ed : SUCCESS in 1h 06m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0fea12b891b745728471377eedc30791 : SUCCESS in 42m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b4afa08704404c95b309af55087fc5c6 : SUCCESS in 2h 21m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/32e12141d1fb41cbab2afa938d23fcea : TIMED_OUT in 3h 04m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6149da0672c548e48c1d1af0e3ac3beb : SUCCESS in 2h 31m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b50a4bce9b234cf8bbb3b9d692b4f74f : SUCCESS in 2h 34m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"56ca5428efbd802aa5456635bfc4acb390652ca3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-07 06:07:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b3e66ed5058dec20a3e53b14040f1f3450aa6a08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 09:11:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35a125188ae641f09817282837ab81df\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6cf757d40e7148ac9f3b31746e4f7096 : SUCCESS in 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db0c9b7bf6a54f42bc8ab06636c41ebe : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/487303e8cdd24d368a35c5d520808a2b : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f338c1eaec34495795067d2109592b31 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/050041c90b0a471d841c27c4df22a8db : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/300ea791c9e541e58ae8f2078c083fbd : SUCCESS in 10m 45s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/477f6317496c47e7b4eaa0368ae7c63f : SUCCESS in 4m 56s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ec428c2f4d584de484b0bdfabe9d3869 : SUCCESS in 11m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e435c93f77524db0a60e912f37d51a25 : SUCCESS in 5m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ffcdb5236eed45a79ec94e5df8bd8459 : SUCCESS in 6m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/024d5619bc634d05919a7e747226c9b5 : SUCCESS in 8m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/09b9354e5af44fa49cb63db879a90e0d : SUCCESS in 8m 54s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f177e509799a47a49974f464781f3bb1 : SUCCESS in 1h 28m 27s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a9d686f882b4476988bcb88d57a4f875 : SUCCESS in 2h 01m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3c114a038670403abab01c9b64963a4d : SUCCESS in 1h 24m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/dc0f05f42cf44ada9b3045df2117223d : SUCCESS in 2h 13m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/41f4cec4f06f49e3b0613db646cdcb09 : SUCCESS in 1h 39m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/19897dd182c7410e8c112973d5440976 : SUCCESS in 39m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/63400ae02a4d44138a019a7cd4aa2306 : SUCCESS in 40m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1c9dadd8425441f69ec142174e16894c : SUCCESS in 54m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5c3e650f58e84a81b999c573159eacc8 : SUCCESS in 33m 08s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fbc831c265644736b1b562f9ba8bc135 : TIMED_OUT in 2h 33m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b40bb5ffb2f74bd9b96e583d6ccac05e : TIMED_OUT in 3h 03m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3f80d0cff1e44d93bd999e7712c30916 : TIMED_OUT in 2h 32m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f72091cb0df4612bbbf3036abd9921f : SUCCESS in 2h 08m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c5dabc61035e4baeebd7a5e83e4f43605d3d5f60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 09:12:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"69f8c18ccfff6c623f28ebd04d4174df627c195b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 11:23:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fdd4416f9314f13a2df6f2c22e54457\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f533438c2064046855fdaa5fc888993 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80c26e295bd34714aef51c634537676c : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf6ecdc45bf44c1d911049dd2d255627 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/298d7d68d7bc48cbaf921f308b24d780 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/122b3e7bcc2b4b0088908a8a2455a3ce : SUCCESS in 13m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f9608bbfbb384fa986bcf5d22599b3f7 : SUCCESS in 4m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/af8dfdee6cb94fc1b9449c309d258902 : SUCCESS in 8m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a3e16c77c8fb494ca20c89df0a04a2d3 : SUCCESS in 2h 02m 19s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/59a4bca5a4e9452084fd5428ec58c77c : SUCCESS in 2h 02m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1095a10b76204b4b934baf3aab339943 : SUCCESS in 1h 17m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/04fdcf794c7e4234a817efb45a58e0fe : SUCCESS in 40m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4ac892de9dd7498b9bfd5d93d76d0f60 : SUCCESS in 51m 20s","accounts_in_message":[],"_revision_number":2},{"id":"39b6958642a591b251353465fef8aea904e98c2c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 11:23:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"586628c2f50bd77e3711ce5c975c4385596908d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 11:24:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/559e96cb25334af68acec31de27e66b9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b78940d85f9b47abb749720d034e6fc5 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"47de8fa3444cc7b1b68d486ad8c06d294a9ef9de","revisions":{"c27e90136203579e6e8c956e06780ce40d1a7ad1":{"kind":"REWORK","_number":1,"created":"2023-03-05 22:18:44.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/80/876480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/876480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/876480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/876480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/876480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/876480/1"}}},"commit":{"parents":[{"commit":"e0f67604428ebb24af2136f4588091bc94780887","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0f67604428ebb24af2136f4588091bc94780887"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-03-05 22:16:16.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-03-05 22:16:16.000000000","tz":0},"subject":"Fix upgrade check not working","message":"Fix upgrade check not working\n\nThe octavia-status upgrade check command fails because\nthe initialization of Policy() requires oslo_config to\nbe initialized already which it isn\u0027t.\n\nThis can be reproduced in master by running:\n\n  tox -e venv -- octavia-status upgrade check\n\nI can observe this issue multiple relases back which\nprobably means this has been broken for a long time.\n\nWe shouldn\u0027t need to init Policy() because it just\nloads policy rules which is not checked in upgrade\ncheck anyway.\n\nChange-Id: I16430fa52db02e7445203994220673c1037d764c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c27e90136203579e6e8c956e06780ce40d1a7ad1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c27e90136203579e6e8c956e06780ce40d1a7ad1"}]},"branch":"refs/heads/master"},"47de8fa3444cc7b1b68d486ad8c06d294a9ef9de":{"kind":"REWORK","_number":2,"created":"2023-03-06 15:30:25.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/80/876480/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/876480/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/876480/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/876480/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/876480/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/876480/2"}}},"commit":{"parents":[{"commit":"e0f67604428ebb24af2136f4588091bc94780887","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e0f67604428ebb24af2136f4588091bc94780887"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-03-05 22:16:16.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2023-03-06 15:30:25.000000000","tz":0},"subject":"Fix upgrade check not working","message":"Fix upgrade check not working\n\nThe octavia-status upgrade check command fails because\nthe initialization of Policy() requires oslo_config to\nbe initialized already which it isn\u0027t.\n\nThis can be reproduced in master by running:\n\n  tox -e venv -- octavia-status upgrade check\n\nI can observe this issue multiple relases back which\nprobably means this has been broken for a long time.\n\nWe shouldn\u0027t need to init Policy() because it just\nloads policy rules which is not checked in upgrade\ncheck anyway.\n\nChange-Id: I16430fa52db02e7445203994220673c1037d764c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/47de8fa3444cc7b1b68d486ad8c06d294a9ef9de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/47de8fa3444cc7b1b68d486ad8c06d294a9ef9de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
