)]}'
{"id":"openstack%2Foctavia~876885","triplet_id":"openstack%2Foctavia~master~I987211111ebff8d1be50ffd33ba91dc4adf4a709","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-03-09 10:26:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I987211111ebff8d1be50ffd33ba91dc4adf4a709","subject":"Workaround an openstacksdk bug","status":"MERGED","created":"2023-03-08 17:55:31.000000000","updated":"2023-03-09 10:27:40.000000000","submitted":"2023-03-09 10:26:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876885","meta_rev_id":"c1ad519b1afc23e41ee84925f82a4f050cd3e067","_number":876885,"virtual_id_number":876885,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-09 10:26:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-03-09 07:54:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-08 18:33:30.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2023-03-09 07:54:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-08 18:33:30.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-03-08 20:55:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-09 07:54:05.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"af39ae9ea9215a2465f4cdaa9755d0140eca520a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-08 17:55:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4eed0db71222e8dea1a05815f00ecbfbfe66bf9f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-08 18:33:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bdfecd069fd1473e730b7ba583f26e2fba2730bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 20:55:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b74afc0d1b2f47c1bd44626e8792ef67\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4c67a72c3a14251a525be65038c07de : SUCCESS in 9m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8493ef35e7954c9d8cc67782a9a5909a : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5afd26c0de6b4d60b335886c8891c898 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9499cce817644ca9b2bd401440c9b54d : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09bc8c1b84b7440ca86aeb6058f0f9e2 : SUCCESS in 11m 22s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/719424d7e92345a2ad5efb8e1695fcae : SUCCESS in 5m 01s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/47c551cb4f7343b885b3e0e4aa0ad3d4 : SUCCESS in 6m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/af8ea7d9cdd34867a8bbc25e03062fac : SUCCESS in 4m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/93087060131143a69658ce9fc2559706 : SUCCESS in 6m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e1f7c68a08914f25a30560864901f365 : SUCCESS in 7m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b37690a1ce37448198c56cb4b45a3900 : SUCCESS in 7m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/28e8dfa3aed64fdd92e542f2f4bda1af : SUCCESS in 1h 32m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1ad45df8eb0543719184714f6199dd1a : FAILURE in 23m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8d3715d84ec74711b5333184d88de5fd : FAILURE in 22m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c3692f93bf79407fa98ec7933a2cb88d : SUCCESS in 39m 17s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a0dd3b455a1048ad885d03f0ec9c6bdc : SUCCESS in 1h 20m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1347be4474ea4d93a1f7e1d5b86910e9 : FAILURE in 23m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d3994a9854d6412191798da03a4677ff : FAILURE in 22m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bb3855a307bf448ba8fbc73c9a69f28e : SUCCESS in 2h 56m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/466b2521f127456a8a6f8f2f40fcffb3 : SUCCESS in 2h 30m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/04adbcb6ee224cf682d1a475dd462c91 : SUCCESS in 2h 16m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bb36034e54464eaba6065308e724e344 : SUCCESS in 2h 34m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/db236324d44c456e8801102acc58f4b8 : FAILURE in 2h 11m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b0cbf84cf10bb2d3ff724a4d357eac71b2696968","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-08 21:15:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2efd8949c398d6d5b5a132e2c1d17bc6d2ed053a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 23:57:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1d4d08562324bcf9088a07b6a1c785e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7a52e1da63440e2bf8616d6d85a23cc : SUCCESS in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e5f9f7a5b6941ae8a41e382b2b04f09 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42e1a74e3f9b4e339458b545a9570d70 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb08148086194c88a5aed05aeed89493 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7101b20d1ada4c51ac4fad4331145941 : SUCCESS in 11m 11s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1a488c23bdc247e4bdc559eecb5cec24 : SUCCESS in 4m 46s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c83ef61bc82f4f8e8839ed5f226d1cce : SUCCESS in 7m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/67f06a3fae2f44b6b03b12aeb90a0ec5 : SUCCESS in 4m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/40bd3b2f926a4012b50fabe97386e25b : SUCCESS in 7m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d7c9f98c639041fc84420d874c6fa550 : SUCCESS in 6m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1d45dd8d66d443a19ea7cbf23ae93d48 : SUCCESS in 7m 17s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ec03d455b0be4aeba3ba710798081738 : SUCCESS in 1h 19m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1f2efb0709a84b40990eff5d41abaa3c : SUCCESS in 1h 45m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/01e8779255784bc9aa4e5115cb944665 : SUCCESS in 41m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0e8206fb4bbc4007b6e5be1604824024 : SUCCESS in 50m 57s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c891f4fb31d44f3ca27860217d0cb833 : SUCCESS in 1h 00m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c58cd4b4509c4102b7d9a9bea0b04052 : FAILURE in 22m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e2eef9447b9843518af435b213db3418 : SUCCESS in 1h 56m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c37279e921fb46d4afa38939a266f39a : SUCCESS in 2h 10m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0b6228e4f78746aa9091afff2e0f2f15 : SUCCESS in 2h 07m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f2ba298283cf480fbd94cd9b840fdaaf : FAILURE in 22m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fbe6f9a1249546ca85dbce413c799f3a : SUCCESS in 2h 14m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d7a6f1ed3654841a78fada8b2614d0e : TIMED_OUT in 2h 34m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d09c967dbabe5fa83f7150df70fb81e316093d35","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-09 07:54:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"417ae9cd2b8d9c52abe9c69102c1daa765ccb955","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 07:54:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c6ad1f70b8ae9597b47f33127e63f3f4909bf428","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 10:26:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/138975573cec4f15850cb3358799d37b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d582fab6bb94c639f38f486d18faf3a : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/778a868703e64d6e8ded88ed9591e60a : SUCCESS in 5m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15baabc07e0d4afab4695ec5d3d65a90 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c61ee0c342b411ea1b1032c3280419b : SUCCESS in 12m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/47efe075be694cf0ade57fd82c39af70 : SUCCESS in 4m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cf94baeb974b408688ccf3d011cad13b : SUCCESS in 10m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e6ed8d14617a428f9f9c93702b874e50 : SUCCESS in 1h 29m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/556c39ff700149cab91428a98c7085c6 : SUCCESS in 2h 20m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/39ebee7f028f4023875c041d7990cb6e : SUCCESS in 44m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/98263923c3294105a4d2b743dd312e34 : SUCCESS in 49m 44s","accounts_in_message":[],"_revision_number":1},{"id":"fd8def9b27a61ebbe61b5e83d1877699d30b584c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 10:26:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c1ad519b1afc23e41ee84925f82a4f050cd3e067","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 10:27:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48cbba7d0e784adf9e074ad4b806fa6a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8de2592d6e7e40298c1abc7d2ad93ece : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ed6f37519f27335b136340f14f658d49adfe540","revisions":{"5ed6f37519f27335b136340f14f658d49adfe540":{"kind":"REWORK","_number":1,"created":"2023-03-08 17:55:31.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/85/876885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/85/876885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/876885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/876885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/876885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/85/876885/1"}}},"commit":{"parents":[{"commit":"fe63fd4ca5b4c62f9e5720d1194cc090f1bea0b0","subject":"Merge \"Fix example policy file system-reader role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fe63fd4ca5b4c62f9e5720d1194cc090f1bea0b0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-08 17:53:03.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-08 17:55:29.000000000","tz":0},"subject":"Workaround an openstacksdk bug","message":"Workaround an openstacksdk bug\n\nThere was a bug[1] in the openstacksdk implementation that could cause load balancer delete calls to fail. This impacts the octavia dashboard ability to delete load balancers.\nThis patch adds a workaround for that bug until the impacted versions of openstacksdk have all gone EOL.\n\n[1] https://review.opendev.org/c/openstack/openstacksdk/+/876669\n\nChange-Id: I987211111ebff8d1be50ffd33ba91dc4adf4a709\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed6f37519f27335b136340f14f658d49adfe540"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed6f37519f27335b136340f14f658d49adfe540"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX and -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}},{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX and -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX and -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}}]}
