)]}'
{"id":"openstack%2Foctavia~877116","triplet_id":"openstack%2Foctavia~master~Iad76be67631bb0393158accad2937273c9ede0c4","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-03-16 15:49:35.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iad76be67631bb0393158accad2937273c9ede0c4","subject":"DNM: Testing DIB diskspace issue","status":"ABANDONED","created":"2023-03-10 18:41:49.000000000","updated":"2023-03-16 15:49:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"06b61c4c96fcde57c2e48e22e00fea8ce746797c","_number":877116,"virtual_id_number":877116,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-03-10 21:35:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-10 21:35:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"999ed3499fe031ff1c7ac850b7ae2259f5332e8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-10 18:41:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80373862cc5f3c2dcd9efc12891f608e1ce6a0ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 21:35:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7fc8405a6d84380a12a541f2b089509\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81de82b613ed4aa992f6268b1ef3d275 : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b1b639d5e0c4f608abb37d22737eac8 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d680c6262b9c4095a523851cfb07c791 : SUCCESS in 6m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f019c3bc20f4b26a99d2c365afffc4f : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dc5948ef14d491e90dd30926267dc10 : SUCCESS in 12m 26s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ef1e2b5e10fd4c9ba7b5d5577ccc0058 : SUCCESS in 6m 31s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d7827512b95a44029f158170d1cfb90c : SUCCESS in 8m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/504c36047cf44fc29734bb528dabe245 : SUCCESS in 4m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f16fa4bed50a4724bd45f82ab46b8cd6 : SUCCESS in 9m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/66f196a9e9d142d49e6f27cb51b7a4a8 : SUCCESS in 8m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f75c3f279fbd41f28d4e28082945c89f : SUCCESS in 8m 39s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/370c14e336194430a6d3703d1f467f3b : SUCCESS in 1h 37m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa7c73679bec494dbde1cebab4b249b7 : TIMED_OUT in 2h 33m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c55cb8645e774ded92097c361ab7aea7 : SUCCESS in 43m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/93ae13b3c0424df6973cbd0fd1e32139 : SUCCESS in 48m 45s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/dd37cad3c503485986bbe2c7f4273669 : SUCCESS in 1h 08m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f35dd2ef77f466f99d5a19dda6fc64f : SUCCESS in 33m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/52a411cfdf4641a0ad404e106b0990fb : SUCCESS in 2h 10m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/47fad80b76b04e18860b3d9e047cfb8c : SUCCESS in 2h 09m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fa72614501df40e4a74dd5a6d3d2d01b : SUCCESS in 2h 19m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/32a59638705c48969446417473e2c30d : FAILURE in 21m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2264b7f2b4374557b12141731a34cb46 : FAILURE in 22m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b24c6637f22406eb36a78b29a2394f4 : TIMED_OUT in 2h 34m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06b61c4c96fcde57c2e48e22e00fea8ce746797c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-16 15:49:35.000000000","message":"Abandoned\n\nWe found the issue","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cfa15801d6f9e719e00f2307dacb942ed59284fd","revisions":{"cfa15801d6f9e719e00f2307dacb942ed59284fd":{"kind":"REWORK","_number":1,"created":"2023-03-10 18:41:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/16/877116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/877116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/877116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/877116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/877116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/877116/1"}}},"commit":{"parents":[{"commit":"5ed6f37519f27335b136340f14f658d49adfe540","subject":"Workaround an openstacksdk bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed6f37519f27335b136340f14f658d49adfe540"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-10 18:41:19.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-10 18:41:47.000000000","tz":0},"subject":"DNM: Testing DIB diskspace issue","message":"DNM: Testing DIB diskspace issue\n\nChange-Id: Iad76be67631bb0393158accad2937273c9ede0c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cfa15801d6f9e719e00f2307dacb942ed59284fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cfa15801d6f9e719e00f2307dacb942ed59284fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
