)]}'
{"id":"openstack%2Foctavia~877141","triplet_id":"openstack%2Foctavia~master~I556c89dbc3aa73415d084187eea21d1790decbd5","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-03-15 09:32:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I556c89dbc3aa73415d084187eea21d1790decbd5","subject":"Purge some dev tools from the amphora image","status":"MERGED","created":"2023-03-10 22:28:49.000000000","updated":"2023-03-21 15:42:44.000000000","submitted":"2023-03-15 09:32:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877141","meta_rev_id":"256c7e78fdee5ce1012f6de864bab296b024266f","_number":877141,"virtual_id_number":877141,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-15 09:32:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-03-15 07:03:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-13 08:39:21.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2023-03-15 07:03:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2023-03-15 07:03:58.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-11 01:15:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-13 08:39:21.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-03-15 07:03:58.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"9ba52a07cdd86d88bf17aa9a437a8d726c249619","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-10 22:28:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf169d81665830e5f83dc4d8eb5f09055d985d73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 01:15:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b78b7cf65d1d4441ac7616902950c8f8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b1c7c93d109459c9ccdd016b808b56e : SUCCESS in 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2889ac1f37d048f38cc189e8cfc35bf6 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b789da9cf0b642b686e8b22838dd5f76 : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c3658e575494ee39028e387b9abf933 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9857283b79ba4208bc04f590404d7d37 : SUCCESS in 10m 59s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ecc638978010496eb78a3958419f8650 : SUCCESS in 5m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3a0df2fd6a8a4220932e30586ade4108 : SUCCESS in 8m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/55c0f039d0c64b9b8f227bda51be7231 : SUCCESS in 5m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0684147b5c564ff18ffc15f9428cd454 : SUCCESS in 7m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/83c919e9e8fc4ab0bf284f238c33f5de : SUCCESS in 7m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4225b7ba8a424689b4b0ea9f4947deb5 : SUCCESS in 7m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/afcb094480e546f9b51adee4365509d3 : SUCCESS in 1h 24m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ed4ec634a4c429eb99493af67ec92a9 : SUCCESS in 1h 55m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b06c3c47b3154f9b9b53177100c44859 : SUCCESS in 41m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b5c5e9f0c56840f38c178465add06eb2 : SUCCESS in 45m 22s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3fbd79e29cee4eda958f4743e29a3161 : SUCCESS in 59m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c912cd040d444d6ab2177be8ad6ea81 : SUCCESS in 35m 21s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a22627b0fbe44f4fa847d491424e7748 : SUCCESS in 2h 09m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/598dba020e5c4a1c94766dd82438f1dc : SUCCESS in 2h 42m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a1e6c3bf3e7640d295b26302a4e59dbe : SUCCESS in 1h 53m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5093991cfab646b48f423e06ddc498c8 : SUCCESS in 1h 51m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1ecbfdae600c4bd8949fded74bd7caec : SUCCESS in 2h 18m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d26f18e69854fba8a0170d4a1b7922d : SUCCESS in 2h 08m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3160ad9f46f4d65ff680225194df4e033fd782f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-13 08:39:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e1bb75612f028e0bd3389b6dfbe4f5cee0967c37","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-15 07:03:58.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ac05bad3e8c00825f9dc3f28d00688eed03e20a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 07:04:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"96d552c52a5f22d2e948e9e034ab28069461aea8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 09:32:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/562a20a1b01d44b5865adb7c8faabbbe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3b7799d008446449f220fb53d49e8f0 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70ff9a5ef7cd4514bd557af76b5f21af : SUCCESS in 8m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51eec2a283ef48c9953752b4bcbade5c : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9de37140c4bc432da10eb653deb4a4a2 : SUCCESS in 12m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/741dff5e2d9c40f087ed8800dff7db96 : SUCCESS in 4m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/657c2061748842afa9b5a1fec2b92bdb : SUCCESS in 7m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/93a790c745af4db18fe501d4da6ca602 : SUCCESS in 1h 24m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/69cc6fb302d74064be499eead9af5f1e : SUCCESS in 2h 16m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6f512faeaafc43a9b931e4b9b50032d6 : SUCCESS in 35m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2b892e0efa4e48b2889d06946dbb9ece : SUCCESS in 40m 54s","accounts_in_message":[],"_revision_number":1},{"id":"f9d7ce8d60f78d061080e75e93cc9b16d392a92f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 09:32:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"22b8ca424c5b7c1a2dc7c80ca3ea1a9afde542be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 09:33:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d58fc79c43ef451887c4fc7a8521a535\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23c704bade14488cafa0141f93b255a8 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1},{"id":"256c7e78fdee5ce1012f6de864bab296b024266f","tag":"autogenerated:gerrit:revert","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-21 15:42:44.000000000","message":"Created a revert of this change as I39cfb836fc65d8466ee3b848da06e3a392c1c30d","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"764cbf014f494f28de5dfdae3876afec2f5f705d","revisions":{"764cbf014f494f28de5dfdae3876afec2f5f705d":{"kind":"REWORK","_number":1,"created":"2023-03-10 22:28:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/41/877141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/877141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/877141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/877141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/877141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/877141/1"}}},"commit":{"parents":[{"commit":"5ed6f37519f27335b136340f14f658d49adfe540","subject":"Workaround an openstacksdk bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed6f37519f27335b136340f14f658d49adfe540"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-10 22:27:11.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-10 22:28:48.000000000","tz":0},"subject":"Purge some dev tools from the amphora image","message":"Purge some dev tools from the amphora image\n\nSome of the image build time tools are not being uninstalled from the amphora\nimage during cleanup. This patch adds a package uninstall callout for those\npackages.\n\nChange-Id: I556c89dbc3aa73415d084187eea21d1790decbd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/764cbf014f494f28de5dfdae3876afec2f5f705d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/764cbf014f494f28de5dfdae3876afec2f5f705d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
