)]}'
{"id":"openstack%2Foctavia~877281","triplet_id":"openstack%2Foctavia~master~Ia7f6e20a40fe7113276981a160789624547420e1","project":"openstack/octavia","branch":"master","topic":"acmev2_rfe","attention_set":{"27442":{"account":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"last_update":"2023-07-19 08:33:37.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"35848":{"account":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"last_update":"2023-11-17 12:19:10.000000000","reason":"\u003cGERRIT_ACCOUNT_20498\u003e replied on the change","reason_account":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"}},"11290":{"account":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"last_update":"2023-03-29 11:33:22.000000000","reason":"\u003cGERRIT_ACCOUNT_35848\u003e replied on the change","reason_account":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-04-25 11:41:23.000000000","reason":"\u003cGERRIT_ACCOUNT_35848\u003e replied on the change","reason_account":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-04-14 10:41:58.000000000","reason":"\u003cGERRIT_ACCOUNT_35848\u003e replied on the change","reason_account":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"}}},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-10-24 12:03:24.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Ia7f6e20a40fe7113276981a160789624547420e1","subject":"Add new spec ACMEv2 Cert Management","status":"NEW","created":"2023-03-13 13:35:46.000000000","updated":"2023-11-20 13:20:56.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":109,"unresolved_comment_count":6,"has_review_started":true,"meta_rev_id":"534e21f9d813ca025c6be697713afb3bdaf871d9","_number":877281,"virtual_id_number":877281,"owner":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-07-19 12:57:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":1,"date":"2023-11-17 12:19:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":-1,"date":"2023-11-20 13:20:56.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-13 13:48:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-16 15:47:21.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2023-03-28 19:38:51.000000000","updated_by":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"reviewer":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"state":"CC"},{"updated":"2023-04-17 10:51:45.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"CC"},{"updated":"2023-04-21 23:04:35.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-06-07 15:43:20.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"CC"},{"updated":"2023-06-21 16:14:20.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2023-11-17 12:19:10.000000000","updated_by":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"reviewer":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"state":"REVIEWER"},{"updated":"2023-11-20 13:20:56.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"4ffef6341b65092ae6fe9d7282b84e4f6ff9d66d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-03-13 13:35:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a41a4113ed295838734ca1cbe5d00058e1ed22c6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-03-13 13:36:05.000000000","message":"Topic lets_encrypt_rfe removed","accounts_in_message":[],"_revision_number":1},{"id":"5e9f1f49cfd42deb3eeaf26c3323a37c0161cc72","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-03-13 13:36:31.000000000","message":"Topic set to rfe","accounts_in_message":[],"_revision_number":1},{"id":"a4da52fbc5984381d219e06725e070cdf20eeb9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 13:48:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/60c2e6ef35a943739449dc22635b2c2e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bfc2bef13c9497f9c14820d26bcce75 : FAILURE in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e169a93bee4447ba44d15d97ffebf32 : SUCCESS in 10m 59s","accounts_in_message":[],"_revision_number":1},{"id":"bc0985970775eef17bb654abcad6e302b18d2d05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-03-13 13:50:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fc0334333c019ddda32f9d52852890584f6293e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-03-13 13:52:03.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e6cfd720babd482c3efbea9c5c28fd0c3f3a9594","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-03-13 13:52:19.000000000","message":"Topic lets_encrypt_rfe removed","accounts_in_message":[],"_revision_number":3},{"id":"014a6deac63d6494c1d20dc83aaf4aaefbe91711","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-03-13 13:52:24.000000000","message":"Topic set to lets_encrypt","accounts_in_message":[],"_revision_number":3},{"id":"59c3f3ccb178b6ba153a82521393e9afda594c66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 14:10:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe49b2f7146d44d788440a441283a83b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac6860496130442393004048760703e2 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f726f00b41414881a3fef9825259d9c6 : SUCCESS in 12m 19s","accounts_in_message":[],"_revision_number":3},{"id":"ed875cdaf6dfff463f6fca2ded52cf4ffda4bb7a","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2023-03-28 19:38:51.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"eb0319e203f5e33a8794e1323ef3e6f0c387b3a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-03-29 11:29:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bcb17dda88960246b187c2c83188ab5eed772d00","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-03-29 11:33:22.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f36e26c020649fecf61b7e07959eea7fb57e124b","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-03-29 11:33:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cd0a7067368bbe24c4bf0f8c64024b3d455111a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 11:46:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0f7fc4294d045e9a5bfc74a14beeab5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb8e01d5ad0e457f83e1c8eb40aa6a52 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e09d4853d8094332bdefdbc211461975 : SUCCESS in 11m 35s","accounts_in_message":[],"_revision_number":4},{"id":"67200a8b8fc111874e8c8b90231c100ef98d8083","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-04-06 14:06:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f715be66319a246052095a9b0585e3b8cc5c7b04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 14:22:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e34bca1b3309470daf06b3d8aac011c6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a654cfeb76e42e0b446f15f6fa224e5 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4d4a52defbf4dca896d73bb5416faa1 : SUCCESS in 10m 22s","accounts_in_message":[],"_revision_number":5},{"id":"9269eb37720eed6a9118c3c064a7f66480fd9d22","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-14 09:09:33.000000000","message":"Patch Set 5:\n\n(7 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b53235a946d752585f6ddf5fc9c76afb7954ba84","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-04-14 09:37:20.000000000","message":"Patch Set 5:\n\n(7 comments)","accounts_in_message":[],"_revision_number":5},{"id":"95175f1a3b8bbc8aa9b23de97a88a47c38503fee","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-14 09:55:49.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4031b43ded3b3652701bb369fdcf3cb4ac957f59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-04-14 10:40:59.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"d9eccba0cb75ef3cf274c2f63f1dbf386841322e","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-04-14 10:41:58.000000000","message":"Patch Set 5:\n\n(7 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c78ec45b3e7ad6bd5429fa70a6b13acb280e735a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-04-14 10:45:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"007526f0d9a8f2f443b4d53eb4ef910d2d2a7b42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 11:04:43.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f8233a2d42849499fcf02fd258080ab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9426d86f8c534858a21137525097734a : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50c12ca650e0463aa22b077f3e692849 : SUCCESS in 11m 55s","accounts_in_message":[],"_revision_number":7},{"id":"38c476ad35285d99f8d4af9fd2ece86c7e3f6786","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-04-14 11:07:13.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"02c20edddc137aa813b9a6b1cac94df983e1dd56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 11:24:51.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/935d15bcccc64772b2f5636a1d189f58\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/933c6594ffcc4ab2805cd679078e4665 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01dbc18e5e304dce916a05566fe69a00 : SUCCESS in 10m 56s","accounts_in_message":[],"_revision_number":8},{"id":"f7b1c2a7c164bf0fb1cea62aea83c37fbbf69ca2","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-17 10:51:45.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"79d074c92c85d210468344021184f6c9314c242d","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-04-18 07:50:52.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"f5b241616b73ccf11039c8e78f1876afe4b6a485","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-04-18 08:38:57.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"23c91d4ef470a161ade95b8e99086c142c49e964","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-21 23:04:35.000000000","message":"Patch Set 8: Code-Review-1\n\n(30 comments)","accounts_in_message":[],"_revision_number":8},{"id":"41a498330cb2e972c80db04b23b2792158cd9100","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-04-25 11:41:23.000000000","message":"Patch Set 8:\n\n(22 comments)","accounts_in_message":[],"_revision_number":8},{"id":"670b812510d6aec7870f6b212f07c1d6e952d64c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-04-25 11:43:26.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"d142ea16352ae877f1095dc4f45f2981576b1e34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 11:54:41.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a7ad2d9d73041639339b64866089034\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a7b290824f14d1f92605a3358ec8941 : FAILURE in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9c56c1d8c1d400aaa98d28ae866db46 : FAILURE in 6m 00s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.1/letsencrypt.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.1/letsencrypt.rst","accounts_in_message":[],"_revision_number":9},{"id":"c6594f6a13d56bf58d56565e09d646448f4e1d20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-04-25 12:25:05.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"62da3ae7ecdf448b759df61f55e7d2fd8ab59d1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 12:39:31.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1b5d55b6e644ee0ab25f4fc17ea3f31\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/858ac70555c048c6a9eb0f77a12be2cb : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a9b6a894cf64ebc9461185411338b9a : SUCCESS in 13m 00s","accounts_in_message":[],"_revision_number":10},{"id":"a33811892fe5db7fb6f0a3e3c0250d0afceba9b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-04-26 09:34:08.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"ae3d18657590569c8be1b8a9e9b5b151fc0bc8de","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-04-26 09:34:23.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"722caa4ea49a4219b96ec40be369bfabbdec2749","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 09:54:36.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/292a40dd19c4438994117a2a9810657e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14e051b16e2c4f3ea884273495f88477 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84868223ea5f459d85991eb31315ee78 : SUCCESS in 14m 30s","accounts_in_message":[],"_revision_number":11},{"id":"9a0e9b5df3a4f9330d1ff6f4dcb587ed9f7ce5b7","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-04-26 11:03:58.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"0f35ce4cb5f24f12fc7add61ec35c3efff3fc815","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-06-07 15:43:20.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"6a8bb39ab37ecf2acbde76b5a40d8bc0055388d6","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-06-07 16:00:50.000000000","message":"Patch Set 11:\n\n(7 comments)","accounts_in_message":[],"_revision_number":11},{"id":"1381d17d5b7e35ae1138cfb17a5fc73a30b4cc08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-06-07 16:11:21.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"abd15d1bb7f2e4c4e3e27672ae92d1977f7cf5ac","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-06-07 16:19:47.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"b26be4965cf4b2c0cf0ae7431cf9d452ac74353c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 16:28:17.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08a17689ef394f7f9d43f9efc0d62efa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9c156c3e5b747b2bb7b4220f1500c99 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0c41112ccaa48b4a6f9d6d94bbb041a : SUCCESS in 11m 53s","accounts_in_message":[],"_revision_number":12},{"id":"7f543d6dad2e4de0742b9510db367a2e06b19378","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-06-21 16:14:20.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"339553f69511a63c8f150b8aedf0762354e5f279","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-07-19 08:33:37.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"a6f94d22aaa3de3b2894c966fd6493b46e3fb6ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-07-19 08:43:16.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"b5ed4349297497e32e745b6fb849d34ebd9faad3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 08:50:20.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc93cf537edb42b0a81b03090a8d4277\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3953234f00ba4d57b409ba02fe6b7f70 : FAILURE in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4be4ef30340439daee3c3d3611ab33c : FAILURE in 6m 32s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/contributor/specs/version1.1/acmev2.rst in the commit\u0027\n  Comments left for invalid file doc/source/contributor/specs/version1.1/acmev2.rst","accounts_in_message":[],"_revision_number":14},{"id":"b99c4041a19945ed0394dd0d0fe06c6616239332","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-07-19 09:40:29.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"a8a711656ea0f1bcc501a5e26b82a3c1116b2b21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 09:56:28.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d23747a12b7f4fee9b6ebab81f3235c6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c9d317b0ccf42898854c84d61ccd415 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7b69a76c84f4870b0d7e518abc990a6 : SUCCESS in 11m 01s","accounts_in_message":[],"_revision_number":15},{"id":"d56a9ba9f554b1360bf111b9312aa91c5fa13188","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"date":"2023-07-19 12:37:59.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"a65a51554cf6e23d9f23b1d2073ae300e14bef06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 12:57:55.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64bc7a4e92fd4168ac1d993b0b61b868\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dc866ddc5bc49a6b742bbe1a1c2f31c : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1890023e62342a5a5edcdfbb1d25fc5 : SUCCESS in 12m 25s","accounts_in_message":[],"_revision_number":16},{"id":"526360f7a47ad57a210f451fc054d037d1876ba9","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-24 12:03:24.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"7e9bfb11acf337fcbada2ab83eae375a5ed0dfe7","author":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"date":"2023-11-17 12:19:10.000000000","message":"Patch Set 16: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"534e21f9d813ca025c6be697713afb3bdaf871d9","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-11-20 13:20:56.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"80c2f862252641355fd5e91548817a37fc2fc444","revisions":{"cf7fece8761fe54fb9cae10b17d757a6d3eaa5c1":{"kind":"REWORK","_number":1,"created":"2023-03-13 13:35:46.000000000","uploader":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"ref":"refs/changes/81/877281/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/877281/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/877281/1"}}},"commit":{"parents":[{"commit":"5ed6f37519f27335b136340f14f658d49adfe540","subject":"Workaround an openstacksdk bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed6f37519f27335b136340f14f658d49adfe540"}]}],"author":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-03-13 13:35:37.000000000","tz":60},"committer":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-03-13 13:35:37.000000000","tz":60},"subject":"Add new spec Let\u0027s Encrypt support","message":"Add new spec Let\u0027s Encrypt support\n\nChange-Id: Ia7f6e20a40fe7113276981a160789624547420e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf7fece8761fe54fb9cae10b17d757a6d3eaa5c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf7fece8761fe54fb9cae10b17d757a6d3eaa5c1"}]},"branch":"refs/heads/master"},"905a58b07a71dd39bf11657656bbec1c249d2392":{"kind":"REWORK","_number":2,"created":"2023-03-13 13:50:16.000000000","uploader":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"ref":"refs/changes/81/877281/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/877281/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/877281/2"}}},"commit":{"parents":[{"commit":"5ed6f37519f27335b136340f14f658d49adfe540","subject":"Workaround an openstacksdk bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed6f37519f27335b136340f14f658d49adfe540"}]}],"author":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-03-13 13:35:37.000000000","tz":60},"committer":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-03-13 13:50:08.000000000","tz":60},"subject":"Add new spec Let\u0027s Encrypt support","message":"Add new spec Let\u0027s Encrypt support\n\nChange-Id: Ia7f6e20a40fe7113276981a160789624547420e1\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/905a58b07a71dd39bf11657656bbec1c249d2392"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/905a58b07a71dd39bf11657656bbec1c249d2392"}]},"branch":"refs/heads/master"},"3050d3cdb7082ffbb9ec6cbd1e9d19b706c4c2ec":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-03-13 13:52:03.000000000","uploader":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"ref":"refs/changes/81/877281/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/877281/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/877281/3"}}},"commit":{"parents":[{"commit":"5ed6f37519f27335b136340f14f658d49adfe540","subject":"Workaround an openstacksdk bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed6f37519f27335b136340f14f658d49adfe540"}]}],"author":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-03-13 13:35:37.000000000","tz":60},"committer":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-03-13 13:51:38.000000000","tz":60},"subject":"Add new spec Let\u0027s Encrypt support","message":"Add new spec Let\u0027s Encrypt support\n\nChange-Id: Ia7f6e20a40fe7113276981a160789624547420e1\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\nStory: 2010644\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3050d3cdb7082ffbb9ec6cbd1e9d19b706c4c2ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3050d3cdb7082ffbb9ec6cbd1e9d19b706c4c2ec"}]},"branch":"refs/heads/master"},"ed75ff2a39a75ab176f7e273e4a8def850f3bba0":{"kind":"REWORK","_number":4,"created":"2023-03-29 11:29:31.000000000","uploader":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"ref":"refs/changes/81/877281/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/877281/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/877281/4"}}},"commit":{"parents":[{"commit":"5ed6f37519f27335b136340f14f658d49adfe540","subject":"Workaround an openstacksdk bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed6f37519f27335b136340f14f658d49adfe540"}]}],"author":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-03-13 13:35:37.000000000","tz":60},"committer":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-03-29 11:29:23.000000000","tz":120},"subject":"Add new spec Let\u0027s Encrypt support","message":"Add new spec Let\u0027s Encrypt support\n\nChange-Id: Ia7f6e20a40fe7113276981a160789624547420e1\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\nStory: 2010644\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed75ff2a39a75ab176f7e273e4a8def850f3bba0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed75ff2a39a75ab176f7e273e4a8def850f3bba0"}]},"branch":"refs/heads/master"},"ed14f10c42b892744f42f1001badfdc105ff0217":{"kind":"REWORK","_number":5,"created":"2023-04-06 14:06:31.000000000","uploader":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"ref":"refs/changes/81/877281/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/877281/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/877281/5"}}},"commit":{"parents":[{"commit":"5ed6f37519f27335b136340f14f658d49adfe540","subject":"Workaround an openstacksdk bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed6f37519f27335b136340f14f658d49adfe540"}]}],"author":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-03-13 13:35:37.000000000","tz":60},"committer":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-04-06 14:06:24.000000000","tz":120},"subject":"Add new spec Let\u0027s Encrypt support","message":"Add new spec Let\u0027s Encrypt support\n\nChange-Id: Ia7f6e20a40fe7113276981a160789624547420e1\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\nStory: 2010644\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed14f10c42b892744f42f1001badfdc105ff0217"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed14f10c42b892744f42f1001badfdc105ff0217"}]},"branch":"refs/heads/master"},"ee158ecfb665971e2e6ce252b610d56ba2bfc9e1":{"kind":"REWORK","_number":6,"created":"2023-04-14 10:40:59.000000000","uploader":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"ref":"refs/changes/81/877281/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/877281/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/877281/6"}}},"commit":{"parents":[{"commit":"5ed6f37519f27335b136340f14f658d49adfe540","subject":"Workaround an openstacksdk bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed6f37519f27335b136340f14f658d49adfe540"}]}],"author":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-03-13 13:35:37.000000000","tz":60},"committer":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-04-14 10:40:53.000000000","tz":120},"subject":"Add new spec Let\u0027s Encrypt support","message":"Add new spec Let\u0027s Encrypt support\n\nChange-Id: Ia7f6e20a40fe7113276981a160789624547420e1\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\nStory: 2010644\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ee158ecfb665971e2e6ce252b610d56ba2bfc9e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ee158ecfb665971e2e6ce252b610d56ba2bfc9e1"}]},"branch":"refs/heads/master"},"1bcddf15b8e600f36a251af22bcc822c67466a22":{"kind":"REWORK","_number":7,"created":"2023-04-14 10:45:28.000000000","uploader":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"ref":"refs/changes/81/877281/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/877281/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/877281/7"}}},"commit":{"parents":[{"commit":"5ed6f37519f27335b136340f14f658d49adfe540","subject":"Workaround an openstacksdk bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed6f37519f27335b136340f14f658d49adfe540"}]}],"author":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-03-13 13:35:37.000000000","tz":60},"committer":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-04-14 10:45:20.000000000","tz":120},"subject":"Add new spec Let\u0027s Encrypt support","message":"Add new spec Let\u0027s Encrypt support\n\nChange-Id: Ia7f6e20a40fe7113276981a160789624547420e1\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\nStory: 2010644\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1bcddf15b8e600f36a251af22bcc822c67466a22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1bcddf15b8e600f36a251af22bcc822c67466a22"}]},"branch":"refs/heads/master"},"ff775e331716451cbb7f29c87be647a5e33cddd9":{"kind":"REWORK","_number":8,"created":"2023-04-14 11:07:13.000000000","uploader":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"ref":"refs/changes/81/877281/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/877281/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/877281/8"}}},"commit":{"parents":[{"commit":"5ed6f37519f27335b136340f14f658d49adfe540","subject":"Workaround an openstacksdk bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed6f37519f27335b136340f14f658d49adfe540"}]}],"author":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-03-13 13:35:37.000000000","tz":60},"committer":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-04-14 11:07:08.000000000","tz":120},"subject":"Add new spec Let\u0027s Encrypt support","message":"Add new spec Let\u0027s Encrypt support\n\nChange-Id: Ia7f6e20a40fe7113276981a160789624547420e1\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\nStory: 2010644\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff775e331716451cbb7f29c87be647a5e33cddd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff775e331716451cbb7f29c87be647a5e33cddd9"}]},"branch":"refs/heads/master"},"121439b0d92f36e796225f830b92df5671516ce8":{"kind":"REWORK","_number":9,"created":"2023-04-25 11:43:26.000000000","uploader":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"ref":"refs/changes/81/877281/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/877281/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/877281/9"}}},"commit":{"parents":[{"commit":"5ed6f37519f27335b136340f14f658d49adfe540","subject":"Workaround an openstacksdk bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed6f37519f27335b136340f14f658d49adfe540"}]}],"author":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-03-13 13:35:37.000000000","tz":60},"committer":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-04-25 11:43:07.000000000","tz":120},"subject":"Add new spec Let\u0027s Encrypt support","message":"Add new spec Let\u0027s Encrypt support\n\nChange-Id: Ia7f6e20a40fe7113276981a160789624547420e1\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\nStory: 2010644\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/121439b0d92f36e796225f830b92df5671516ce8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/121439b0d92f36e796225f830b92df5671516ce8"}]},"branch":"refs/heads/master"},"d449cf4ba5fead874928ce3228314acf58a8dcb4":{"kind":"REWORK","_number":10,"created":"2023-04-25 12:25:05.000000000","uploader":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"ref":"refs/changes/81/877281/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/877281/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/877281/10"}}},"commit":{"parents":[{"commit":"5ed6f37519f27335b136340f14f658d49adfe540","subject":"Workaround an openstacksdk bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed6f37519f27335b136340f14f658d49adfe540"}]}],"author":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-03-13 13:35:37.000000000","tz":60},"committer":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-04-25 12:24:45.000000000","tz":120},"subject":"Add new spec Let\u0027s Encrypt support","message":"Add new spec Let\u0027s Encrypt support\n\nChange-Id: Ia7f6e20a40fe7113276981a160789624547420e1\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\nStory: 2010644\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d449cf4ba5fead874928ce3228314acf58a8dcb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d449cf4ba5fead874928ce3228314acf58a8dcb4"}]},"branch":"refs/heads/master"},"d9eff928b36b06ab83222d7fe32725c336ff02f8":{"kind":"REWORK","_number":11,"created":"2023-04-26 09:34:08.000000000","uploader":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"ref":"refs/changes/81/877281/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/877281/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/877281/11"}}},"commit":{"parents":[{"commit":"5ed6f37519f27335b136340f14f658d49adfe540","subject":"Workaround an openstacksdk bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed6f37519f27335b136340f14f658d49adfe540"}]}],"author":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-03-13 13:35:37.000000000","tz":60},"committer":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-04-26 09:34:02.000000000","tz":120},"subject":"Add new spec Let\u0027s Encrypt support","message":"Add new spec Let\u0027s Encrypt support\n\nChange-Id: Ia7f6e20a40fe7113276981a160789624547420e1\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\nStory: 2010644\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d9eff928b36b06ab83222d7fe32725c336ff02f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d9eff928b36b06ab83222d7fe32725c336ff02f8"}]},"branch":"refs/heads/master"},"cf8a6f4ed7e3a9ced97c76fad2b24b71a699740a":{"kind":"REWORK","_number":12,"created":"2023-06-07 16:11:21.000000000","uploader":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"ref":"refs/changes/81/877281/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/877281/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/877281/12"}}},"commit":{"parents":[{"commit":"5ed6f37519f27335b136340f14f658d49adfe540","subject":"Workaround an openstacksdk bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed6f37519f27335b136340f14f658d49adfe540"}]}],"author":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-03-13 13:35:37.000000000","tz":60},"committer":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-06-07 16:11:11.000000000","tz":120},"subject":"Add new spec Let\u0027s Encrypt support","message":"Add new spec Let\u0027s Encrypt support\n\nChange-Id: Ia7f6e20a40fe7113276981a160789624547420e1\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\nStory: 2010644\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf8a6f4ed7e3a9ced97c76fad2b24b71a699740a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cf8a6f4ed7e3a9ced97c76fad2b24b71a699740a"}]},"branch":"refs/heads/master"},"d75d19e3fcf0456be9238df3a5fbc22dea96849d":{"kind":"REWORK","_number":13,"created":"2023-07-19 08:33:37.000000000","uploader":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"ref":"refs/changes/81/877281/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/877281/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/877281/13"}}},"commit":{"parents":[{"commit":"5ed6f37519f27335b136340f14f658d49adfe540","subject":"Workaround an openstacksdk bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed6f37519f27335b136340f14f658d49adfe540"}]}],"author":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-03-13 13:35:37.000000000","tz":60},"committer":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-07-19 08:32:57.000000000","tz":120},"subject":"Add new spec ACMEv2 Cert Management","message":"Add new spec ACMEv2 Cert Management\n\nChange-Id: Ia7f6e20a40fe7113276981a160789624547420e1\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\nStory: 2010644\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d75d19e3fcf0456be9238df3a5fbc22dea96849d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d75d19e3fcf0456be9238df3a5fbc22dea96849d"}]},"branch":"refs/heads/master"},"72744dde944db75acb98eb40ee8a79382a379a29":{"kind":"REWORK","_number":14,"created":"2023-07-19 08:43:16.000000000","uploader":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"ref":"refs/changes/81/877281/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/877281/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/877281/14"}}},"commit":{"parents":[{"commit":"5ed6f37519f27335b136340f14f658d49adfe540","subject":"Workaround an openstacksdk bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed6f37519f27335b136340f14f658d49adfe540"}]}],"author":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-03-13 13:35:37.000000000","tz":60},"committer":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-07-19 08:40:42.000000000","tz":120},"subject":"Add new spec ACMEv2 Cert Management","message":"Add new spec ACMEv2 Cert Management\n\nChange-Id: Ia7f6e20a40fe7113276981a160789624547420e1\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\nStory: 2010644\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/72744dde944db75acb98eb40ee8a79382a379a29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/72744dde944db75acb98eb40ee8a79382a379a29"}]},"branch":"refs/heads/master"},"7a3cd57a4d7e2e580443102c4582a704cd6fa55e":{"kind":"REWORK","_number":15,"created":"2023-07-19 09:40:29.000000000","uploader":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"ref":"refs/changes/81/877281/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/877281/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/877281/15"}}},"commit":{"parents":[{"commit":"12d8e0de5de9ca754859542bcffd2311f5145d61","subject":"Merge \"Update doc for the OVN provider\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12d8e0de5de9ca754859542bcffd2311f5145d61"}]}],"author":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-03-13 13:35:37.000000000","tz":60},"committer":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-07-19 09:40:19.000000000","tz":120},"subject":"Add new spec ACMEv2 Cert Management","message":"Add new spec ACMEv2 Cert Management\n\nChange-Id: Ia7f6e20a40fe7113276981a160789624547420e1\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\nStory: 2010644\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7a3cd57a4d7e2e580443102c4582a704cd6fa55e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7a3cd57a4d7e2e580443102c4582a704cd6fa55e"}]},"branch":"refs/heads/master"},"80c2f862252641355fd5e91548817a37fc2fc444":{"kind":"REWORK","_number":16,"created":"2023-07-19 12:37:59.000000000","uploader":{"_account_id":35848,"name":"Julian DA CUNHA","display_name":"Julian","email":"julian.da-cunha@corp.ovh.com","username":"jdacunha"},"ref":"refs/changes/81/877281/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/877281/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/877281/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/877281/16"}}},"commit":{"parents":[{"commit":"12d8e0de5de9ca754859542bcffd2311f5145d61","subject":"Merge \"Update doc for the OVN provider\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12d8e0de5de9ca754859542bcffd2311f5145d61"}]}],"author":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-03-13 13:35:37.000000000","tz":60},"committer":{"name":"Julian Da Cunha","email":"julian.da-cunha@corp.ovh.com","date":"2023-07-19 12:37:52.000000000","tz":120},"subject":"Add new spec ACMEv2 Cert Management","message":"Add new spec ACMEv2 Cert Management\n\nChange-Id: Ia7f6e20a40fe7113276981a160789624547420e1\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\nStory: 2010644\nSigned-off-by: Julian Da Cunha \u003cjulian.da-cunha@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/80c2f862252641355fd5e91548817a37fc2fc444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/80c2f862252641355fd5e91548817a37fc2fc444"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
