)]}'
{"id":"openstack%2Foctavia~877414","triplet_id":"openstack%2Foctavia~master~Ibd4da09079e83789d6cfe3658fcf1f266f5cf8b4","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"11290":{"account":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"last_update":"2023-03-29 21:46:29.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-03-29 17:22:17.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Ibd4da09079e83789d6cfe3658fcf1f266f5cf8b4","subject":"Fix ORM caching for with_for_update calls","status":"MERGED","created":"2023-03-14 17:14:16.000000000","updated":"2023-03-29 21:47:37.000000000","submitted":"2023-03-29 21:46:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877414","meta_rev_id":"33ce1e554a8223cf9ddeb56283b993ce89e18874","_number":877414,"virtual_id_number":877414,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-29 21:46:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"all":[{"value":1,"date":"2023-03-28 11:07:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-29 13:51:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2023-03-29 16:29:37.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2023-03-29 16:29:37.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"recommended":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"all":[{"value":1,"date":"2023-03-28 11:07:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2023-03-29 16:29:37.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-14 20:17:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-14 21:11:01.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-03-15 08:57:55.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"state":"REVIEWER"},{"updated":"2023-03-26 20:44:31.000000000","updated_by":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"reviewer":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"state":"REVIEWER"},{"updated":"2023-03-29 16:29:37.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"6f0ef31044d16f15fdb1f689df6c0e877b1f8730","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-14 17:14:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60efeabb2d9eac690cf47868f4951b4d6010ff01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 20:17:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/795ad028fff84a7691347e4f1d55bdb8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/933bf5acb84a4179b5d32e2e5ac28304 : SUCCESS in 12m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59750058850148e790e4c4878562e27b : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2e5350cd8234015af1bdffdca695c04 : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5093438d3df46da93b55315b920d4e0 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f13b8694c814a659da2ffd7d4daf825 : SUCCESS in 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ecfcfad83a74ce6ad8b540088f867d0 : SUCCESS in 5m 36s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/264ffa770f7a4c5ead342e886d312725 : SUCCESS in 5m 14s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/653c6297481348f6844533d97d815856 : SUCCESS in 8m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3f2a4bd1b792463293a40ed136ac895e : SUCCESS in 4m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1fb9f04a05f64829944acfbb0a08ad7a : SUCCESS in 7m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7f24c815f8a94046ba74f8bebb20183f : SUCCESS in 8m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4d473b0fc5764dd3b02423f0da6cc952 : SUCCESS in 7m 01s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/39e5808f46d74b7097604a3c4e06d9f4 : SUCCESS in 1h 35m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4e8fb812469947e9a114406e3405adf1 : SUCCESS in 2h 24m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/09a2d8b948bb4203ace6a5b275ce1bce : SUCCESS in 45m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/da2f8b1410454eecbb7d2b42579ca977 : SUCCESS in 50m 56s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a5d615a13406489087462226b36dfa2e : SUCCESS in 1h 11m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afabe040cd504da6bfa5e701f0b4fc07 : SUCCESS in 44m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8eb3cd313e6f44c68673628c9ec3fdba : FAILURE in 27m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/964317ecd05f43688ca54620909bb204 : FAILURE in 2h 56m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cbe02456637b47d0b70cdeef8a1f96a3 : TIMED_OUT in 2h 34m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ffb48e44d9034733a4efe52c0f899b67 : TIMED_OUT in 2h 33m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/505f5ae04bf44b0a9a17fb2b5a6ed145 : SUCCESS in 2h 25m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d68b7c2d635644d092deb231031d2872 : SUCCESS in 2h 29m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44a3d229c031850be7b112a0e23a1fa12d28528f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-14 21:11:01.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e7b0ea12a52b0bd42060cd764071ef4a062a73f9","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-03-15 08:57:55.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f235e1d9a5ca72e66dfe9456099827258bdddaa4","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-03-15 08:58:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10577629c0d604a2a6cb77d7f4b770b0e13aa551","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-20 10:39:55.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"85169833d6c81942eaed3a5fa71cc187f74dc179","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2023-03-26 20:44:31.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3b493a147601019fc375fbc36aeb7013710139fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-27 06:54:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"48d8aa7500bf3e00bf8aa2c86bf762d6edf46556","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-27 06:54:10.000000000","message":"Patch Set 1: Backport-Candidate+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9660a2fd4fd60af78b8dc031d5361951b97566a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 09:09:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5544abe66bf248dbafdfc9b39a56f59b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3070b68e0414cbbb300277ea7bd046d : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/215a4d082aa54555b858c098239c816c : FAILURE in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fefa51f63ca247c2b40a57b42444aa42 : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db21a79ecd1d4a639c82aac615f306ad : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36c0b837f9354bb2893b0a0efb4019bc : SUCCESS in 10m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1169dbdb09647929e36d1518c0d57c7 : SUCCESS in 6m 08s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/21bebd1900ce491f8804ed2623f0e897 : SUCCESS in 5m 06s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/59931fbfe1ee4ec491026f5efddb8ca5 : SUCCESS in 7m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/aaf9fa100b5e47cc942590952f9b0859 : SUCCESS in 4m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9871b878a4e44c29809ece61b452da2d : SUCCESS in 7m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/db8247c16f684099896faf0a38bc0d14 : SUCCESS in 6m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3e0059b29db74e6ab7bb698e0b3d760c : SUCCESS in 7m 31s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cb1b4dbd96394d22a8bae0dac5eac1c9 : SUCCESS in 1h 04m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c9b8ca4b299b48d5a8831ebe4f9d31e3 : SUCCESS in 1h 40m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/da5e3e3a080a491e83ea43a8b3e84a33 : SUCCESS in 33m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1b8f223368fd46fd8cb21f016379c78c : FAILURE in 30m 33s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0f005ad2fcda4cc7ad7d6877daf7d61b : SUCCESS in 52m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/427e28a8d18e40f3b490d6768ea90dd2 : SUCCESS in 32m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/828cb79c091d4dc1964583fd3f8891bb : SUCCESS in 2h 00m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/96f800ce15a04fa1abb7af6370944c74 : SUCCESS in 2h 13m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2a096518c4b24c84a0ac5557574c2d2f : SUCCESS in 2h 01m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4512339b168f4e489bc656bba9b4b50e : SUCCESS in 1h 48m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5950581a9b0a410e9767825d8141a22f : SUCCESS in 2h 03m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b75c446e120b41fa8e6dbc1dcaefd2cc : SUCCESS in 2h 01m 24s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/plug.py\n  Comments left for invalid file octavia/amphorae/backends/utils/interface.py","accounts_in_message":[],"_revision_number":2},{"id":"10546b459ff98baba8906fbf5d2d506a40420fb9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-28 07:45:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f4dd100384f2b98c7b33601a07e7d495d0360a7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 10:59:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/34008114ddf34f8b937a3c803f45a60a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f98877c2f7e94f94abe20f3b831b46e6 : SUCCESS in 11m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0cf648db80f4bef80eb6d0114f06aab : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5dcb5b91d8bf4ef5b7ae253984e16dc5 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3db0009f92404bc59ff200349531c726 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f1318ba23d746309b411cc41f884fba : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4a6ae73711b4dbdbe3131302ee6fc75 : SUCCESS in 5m 46s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7d8464bf63854d22b180f9627a18a7f5 : SUCCESS in 5m 28s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1f6831fef2ce44118f6309c7324c11b1 : SUCCESS in 8m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8a27930cfd9b468f9efb9cddeeec4e4b : SUCCESS in 4m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6bdcaa3a94f84da7a09520077f81db36 : SUCCESS in 6m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9e3970dbea2c45cab56b12552ea912b5 : SUCCESS in 8m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f01b9148462c48d9931b0c8679a7323e : SUCCESS in 6m 48s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/001ae3bc57e24fb983668b57167fa362 : SUCCESS in 1h 26m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ff97ec20539546488c40b3bb4bc40c2d : TIMED_OUT in 2h 33m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d57fac0643194e1ebdefb6942571f13d : SUCCESS in 43m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e57d5ca58cfe4b66b3663cea919eb74f : SUCCESS in 40m 02s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d0da1617ef524b6781edc1fca11b8d0b : SUCCESS in 59m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0ff420391c20436cbc2623646e7675f6 : SUCCESS in 38m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/60e6afc531304cedb85a6cba736e7f6f : TIMED_OUT in 2h 33m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8171eb060fb44ed09c5e63630a2496cb : TIMED_OUT in 3h 04m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/04a40dc336654147b557a9f3ea7da1e5 : SUCCESS in 2h 01m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9bf48cae49154a05892b0b78fac95faf : TIMED_OUT in 2h 33m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ae00eff9da6d4a69aa6a0b01fbd63dd7 : SUCCESS in 2h 20m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/eeb502ec52a24b2f819fb45310d290d1 : SUCCESS in 2h 26m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f3f30103e3911dfa93c57674e2dfc409de11b13d","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-03-28 11:07:16.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5e6b57eb3a399a5c63ede94c4ece5567b1764330","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-28 11:40:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ab500404ff6b7c2222d9b8a52b284c0d171910ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 14:30:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eb81da7f3bc405e91c8c78bcf80a5e9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cba42c5bf82e417fb03763d5fb10afe8 : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58cb3fbdac234d43a7c3307a9cc2b487 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b6089d0b63448909ac14b0a3096df0f : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae81b05b1bcc4803bdf8153bbc0a2d46 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30ba3360be2244cba19d7ae8c8cfb8d2 : SUCCESS in 12m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2502e4fe8d546aa9ac058dd345ae0e7 : SUCCESS in 4m 57s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/172fa03f45964f8bb80583c71f6b714a : SUCCESS in 5m 07s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0aa88a63825847aa888be27b5f082a15 : SUCCESS in 8m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/713897219bda427e9697f1e6af5076ab : SUCCESS in 4m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9a8bec813d8648bfacabb982dc271254 : SUCCESS in 6m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c19201e4d13048498f31c64cf0d6b849 : SUCCESS in 7m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5789aac6c1954c72b37552489d89b5a3 : SUCCESS in 7m 27s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/036bc54fcfae4367a0bf0e2c720c2f96 : SUCCESS in 2h 01m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/db04e4d244734b3b8b40ffbeaccf0291 : SUCCESS in 1h 57m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/23e0db19a7ca44bfbca47c3cbb19c1aa : SUCCESS in 40m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3593869e101642ab86ac4cc3a337b85a : SUCCESS in 41m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/19a9c08a9aec4bd99d5a9b36e4940607 : SUCCESS in 57m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a8b90ccd429c4ca585377f44d16a511d : SUCCESS in 36m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/94da727ab0b2466eb04bf903a1682045 : SUCCESS in 2h 08m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b1f17c443d864bc8b33e81468938846a : SUCCESS in 2h 46m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8dd42046f5544eea97ad49fb71a8a49e : SUCCESS in 1h 48m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/375a7bca15a04fedb7655b81e84347f4 : SUCCESS in 1h 55m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/bf073bf10ec44d28961f7142254e63c7 : SUCCESS in 1h 45m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7d360311f1b4ecd8a5ad78622c57ce9 : SUCCESS in 2h 27m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0818bd0dd65d5163649a7b1caa5353e4d1e781d3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-29 13:51:13.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e68e16f9a9df68e545ee0cfefc429875e0a0e2c0","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-29 16:29:37.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b6b3adc7f9788d553e4b69a80c2c6bddb72da51c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 16:30:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dd67c2f8494527530a3b33524692c02d92df59e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 17:08:37.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea7c347db02042bebf4c179432c18b62\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfe3cfc3d58e43c38c4fbb68f8df3470 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8e10ff704a34ad581ea953b734e3741 : SUCCESS in 6m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d75eb4bf4f7948fb94d53de82662f0d7 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3f36350152843f5a6dc5c0c15fe11c3 : SUCCESS in 16m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d781b4c6e1354bf19581d29bc0bdcf2f : SUCCESS in 5m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f6bf9c44b2a04a2c81af6595f46cddf5 : SUCCESS in 5m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bfa8173bdd814b8a979777f0c6804b04 : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f0103e549e1a4c68b3a4bfb1d727fb80 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5f1f52f85678436381ff96fba92ce519 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/41eec63376444d2fb238fdf3cec618cb : FAILURE in 26m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/64ba98804a8344eba3d6cd9be952947a : CANCELED","accounts_in_message":[],"_revision_number":2},{"id":"e7e795be13a95867309be3f1d773a60f3c51af50","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-29 17:22:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0f2ef1356eacdc4121f45482382b1596e14c1d45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 19:43:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fa3063b1a6b47eaa83556aa609a6b7e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6670ed63c69d4f5193861510601ed3dd : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ab36187a18e4aafaf753d26ac14c7bb : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aacd34466b4e4f35b46c754b8ba1ca5a : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad873aeaceb74860a6cfd2d579acdcd1 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acccd27c3502461e948c17b04b76c217 : SUCCESS in 13m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2ac5e491be646f8a00456d849e1df09 : SUCCESS in 7m 42s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/cc58b44f6f9d4f6796d3214e5e2f9138 : SUCCESS in 4m 33s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/cb30ac3b80694fcdac3c6f086b2e433d : SUCCESS in 7m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f3ae2ad182bb460393d3f9602c415afc : SUCCESS in 7m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5d8c4bee1b254183bc01d0ff8b538025 : SUCCESS in 7m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/364bfc624be8470aa761bb9cc8a53595 : SUCCESS in 7m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0e695053fcd1453ab634e2bdeb64a610 : SUCCESS in 12m 12s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5704a5f09b2d43f49c74b9d5038717cd : SUCCESS in 1h 31m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cb39bcd2527c46aeb7b8abfb63fb623d : SUCCESS in 2h 19m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6df85d9ee5434e818d26901a2411a367 : SUCCESS in 40m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3512d202a771465ebfa2f40bedfadb8a : SUCCESS in 37m 09s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/270d6aea25e44042b24afa6305a59b11 : SUCCESS in 56m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a513148a3aa4edb886a05405714e75f : SUCCESS in 31m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/475f95d2fc064a54a28bffdfaefd46b0 : SUCCESS in 1h 42m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2f65804e0d4046239924e11f51b0c2f8 : SUCCESS in 2h 11m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/597891a9dacb495ba3a37e89c26dca39 : SUCCESS in 1h 48m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0be770ee0790471f86f424cf9ac5b6a6 : SUCCESS in 1h 55m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9884849d154e49f09a760597b0a99c04 : SUCCESS in 1h 58m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1581abf273c449c814bc7a3bc760bdf : SUCCESS in 2h 11m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fda833c41e183f2f27dc4a9ac434c434e9632e07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 19:43:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"41099dc99f98b85d76dc6096f9526fa537417958","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 21:46:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d0b66b602c048798a92cf5c14aba45d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23ff9cd324f944fa92c5034198376eb2 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1647b2627e634b2ab4852a8d56f02523 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78b9a1a29c6c46c6bac8af99726303cb : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c6ee2d9684b47d8bdd4e60e6f951a4d : SUCCESS in 9m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1b72b615d534a818efe2deee0338741 : SUCCESS in 4m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d78d953c90c2401287aa7457de21136e : SUCCESS in 4m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e392ee6a31fc4e79868ca57760ef25c2 : SUCCESS in 7m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b6d748a1d2ff4790892909d217758cf8 : SUCCESS in 1h 24m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/06249ac572df4b85b368b6202a4050a8 : SUCCESS in 2h 00m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bb46d706f9e34de8aadbe7e4ea66ba7c : SUCCESS in 38m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b43bc9940a4649dfa3d7e8aae7329f39 : SUCCESS in 48m 32s","accounts_in_message":[],"_revision_number":2},{"id":"3fd5571503175a617991ee87f8c0bffdb624d0a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 21:46:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"33ce1e554a8223cf9ddeb56283b993ce89e18874","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 21:47:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ffd018cf45141e1af93b1f8022b04df\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6485d8bf1273460ca32159a8e99eb5b6 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/497cf5290390434a947e26f7cf687c31 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2e4ee9c65e0fba660b0316f1f4203810bb8a67c0","revisions":{"dfd315ada8e80cbff3a9c1952a94e79da8c5f8e1":{"kind":"REWORK","_number":1,"created":"2023-03-14 17:14:16.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/14/877414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/877414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/877414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/877414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/877414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/877414/1"}}},"commit":{"parents":[{"commit":"5ed6f37519f27335b136340f14f658d49adfe540","subject":"Workaround an openstacksdk bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed6f37519f27335b136340f14f658d49adfe540"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-14 17:09:42.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-14 17:13:31.000000000","tz":-240},"subject":"Fix ORM caching for with_for_update calls","message":"Fix ORM caching for with_for_update calls\n\nThe SQLAlchemy recommends to use populate_existing() when using\nwith_for_update() [0], it fixed issues with the caching of the objects.\n\nThis patch precisely fixes a bug when locking of a loadbalancer in the\nbatch member update API call, the load balancer might not have been\nlocked correctly and race conditions could have occurred (processing\nsimultaneously 2 requests in the workers for the same load balancer).\n\n[0] https://docs.sqlalchemy.org/en/14/orm/query.html#sqlalchemy.orm.Query.with_for_update\n\nStory 2010646\nTask 47642\n\nChange-Id: Ibd4da09079e83789d6cfe3658fcf1f266f5cf8b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dfd315ada8e80cbff3a9c1952a94e79da8c5f8e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dfd315ada8e80cbff3a9c1952a94e79da8c5f8e1"}]},"branch":"refs/heads/master"},"2e4ee9c65e0fba660b0316f1f4203810bb8a67c0":{"kind":"REWORK","_number":2,"created":"2023-03-27 06:54:00.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/14/877414/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/14/877414/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/877414/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/877414/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/14/877414/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/14/877414/2"}}},"commit":{"parents":[{"commit":"5ed6f37519f27335b136340f14f658d49adfe540","subject":"Workaround an openstacksdk bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ed6f37519f27335b136340f14f658d49adfe540"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-14 17:09:42.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-27 06:50:55.000000000","tz":120},"subject":"Fix ORM caching for with_for_update calls","message":"Fix ORM caching for with_for_update calls\n\nThe SQLAlchemy recommends to use populate_existing() when using\nwith_for_update() [0], it fixed issues with the caching of the objects.\n\nThis patch precisely fixes a bug when locking of a loadbalancer in the\nbatch member update API call, the load balancer might not have been\nlocked correctly and race conditions could have occurred (processing\nsimultaneously 2 requests in the workers for the same load balancer).\n\n[0] https://docs.sqlalchemy.org/en/14/orm/query.html#sqlalchemy.orm.Query.with_for_update\n\nStory 2010646\nTask 47642\n\nCo-Authored-By: Gaudenz Steinlin \u003cgaudenz.steinlin@cloudscale.ch\u003e\n\nChange-Id: Ibd4da09079e83789d6cfe3658fcf1f266f5cf8b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e4ee9c65e0fba660b0316f1f4203810bb8a67c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e4ee9c65e0fba660b0316f1f4203810bb8a67c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
