)]}'
{"id":"openstack%2Foctavia~877959","triplet_id":"openstack%2Foctavia~master~Ib1e00a0b45776cfab80b9effb2690e12fea63e78","project":"openstack/octavia","branch":"master","topic":"ipv6-bug","attention_set":{},"removed_from_attention_set":{"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-03-27 23:25:41.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-03-21 10:32:40.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-03-27 23:25:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib1e00a0b45776cfab80b9effb2690e12fea63e78","subject":"Fix octavia to accept [ipv6]:port","status":"MERGED","created":"2023-03-20 14:12:28.000000000","updated":"2023-03-27 23:26:52.000000000","submitted":"2023-03-27 23:25:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877959-ipv6-bug","meta_rev_id":"6da67be917311ff10c63a34b67629fd7d4dd17f9","_number":877959,"virtual_id_number":877959,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-27 23:25:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-03-21 10:32:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2023-03-21 14:26:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2023-03-21 14:26:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-03-21 10:32:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-20 16:45:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-20 17:25:28.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-03-20 17:25:28.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-03-21 14:26:58.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"50179890bebc0698fd4d1f3e7c12895a8ef5651f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-03-20 14:12:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b851f30e7112c0d8eddfafb5afa46de39f88adb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 16:45:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c78507b179d943f4a15c56bce14762db\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ec54e7ce8a9462c87d2c2b38860e807 : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d23a88b15524dcf91acaea43a158390 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c5057b9848d4a0ba45ca2f9a5fdf1b2 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb9086e9bab146fd99013b3833ab97ca : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/836e9a0501d6484088e1cfb3c5d23eb3 : SUCCESS in 9m 30s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6af2d8cb70084dc6a4e5e57dffe9f4a9 : SUCCESS in 4m 41s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1caa048c49b44050b8e45056b5cb6611 : SUCCESS in 5m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4a567b0449c04fdea1acfb0070565f14 : SUCCESS in 3m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d443091878bb403abfd731b310d92738 : SUCCESS in 6m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c6517e2b2ade46c5bcf8afe29370039d : SUCCESS in 6m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5a7c903ff3e044738b7392b324e0661f : SUCCESS in 5m 53s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d1100b4605854e41863ad4d02cb7f3f7 : SUCCESS in 1h 03m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd06ec43955a4714b267d16cd6564fcc : SUCCESS in 1h 45m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0b5bd06492f244639c48cefeb2cc6e29 : SUCCESS in 37m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/af3038c7dc0e4949afaff3ffd6d95702 : SUCCESS in 42m 13s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bcccf233ad8c4d5bbf2a6ada01ff2fae : SUCCESS in 58m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc8115f9abec4bd5b793d5984c6c685d : SUCCESS in 33m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6025ae7e5b014c44a193ae439bdbf776 : SUCCESS in 2h 12m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7f9b90cccf5d44a7832235e93626b30f : SUCCESS in 2h 29m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1dee9efd619a42f8ac8dae997dc6d9f0 : SUCCESS in 2h 03m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b1db4d7ec0134dccb8dc95a639e1bfc8 : SUCCESS in 1h 52m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c015122d5e5b4fb8a8545593eee5f059 : SUCCESS in 2h 27m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf0f1e7ebc3c42bd818c60ed2c71c85f : SUCCESS in 2h 18m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"81590d5eb12c799c5f551ddf39285fe661e63d4b","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-03-20 17:25:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dcdbd8dd258e2ef646b77af394fdf5646597a94b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-21 08:24:59.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7e3a92878c5312fbc1cb2990c0192140a47b0db6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-03-21 09:45:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"76b239d6da2e89fb6255bf695ee595e2d6ef62c0","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-03-21 09:45:29.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e9145ed49d038e27715d8be6dc2cd2d83fe29cd1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-21 09:47:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7947f36a10c6adde84710757ebfdf668d24bfef0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-03-21 10:31:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dc60bdcb19e6f6a6279590fa30ac674bf56ec3ee","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-03-21 10:31:15.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9ffeae80e2bba92d979f487be2a9666615339e4d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-21 10:32:40.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"993f718621441481c90706cf50a04702d886aab1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 13:11:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/850d0828736844ed85f187204720605d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e90e2d653344c9999e737abb35a046a : SUCCESS in 11m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c41524f521740a1a4c1e071c6f55e40 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/702e4a1be07049a693a37cf369775b1e : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20d4c12c759046c3aa95ff7d107756d3 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e77ad825fc24703a7359f0a74250fe5 : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7c8cd1999eb4c8a9e37aff067562ed5 : SUCCESS in 6m 06s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/2114e45473f24bacbb8fe12e260e9493 : SUCCESS in 6m 10s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1aa24a4628cd41a68afd3d8d1f159e12 : SUCCESS in 8m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d21a75775f9b423db27ce1be106f1ebf : SUCCESS in 5m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d68de50afa3c45c2966af1b38734ce7b : SUCCESS in 7m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a3d401749c574a55bb74015ec8ed26be : SUCCESS in 7m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bb28b43f3d0e4a76baf2735138f01103 : SUCCESS in 8m 56s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/27d41f3a352840b6aaf525dabf66f014 : SUCCESS in 1h 29m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bae0ecbbd65c4eaf9110eba76e7cf35b : SUCCESS in 2h 27m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/262b8a8cc35043b59e37682e6e4636a7 : SUCCESS in 44m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f50aacd466134d008c8d7fc5842171d7 : SUCCESS in 53m 36s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/cfe1f2eed3004bbf9f8c956ceb04451c : SUCCESS in 1h 10m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/87de7f0fa339405c9c47d63fcd9d1cd3 : SUCCESS in 43m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a3f6272d7ec04cbfba3e4eca91743a52 : SUCCESS in 2h 10m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a69d7cfc170e4a529ab717ff227716bc : SUCCESS in 2h 31m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/00cbeeb2bd1d47b89c439b33b1096b08 : SUCCESS in 2h 25m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e147cc524a3645c888b60b2c48e5846d : TIMED_OUT in 2h 33m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/33231c3194494ff5829faf2f50e2b4fe : SUCCESS in 2h 23m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/66c53b2cc92c4da0bfa45c021b92e4d9 : SUCCESS in 2h 32m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4efcd8d9a0d8f8eab68c97c669d2e71c4a85af6b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-21 14:26:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"99859cec8152bf3ea5c631fa607b132a359a3c4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 14:27:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0cea0eeb1fd8afec18b22c222d02a01175e774e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 15:18:52.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/37b853a0bb3a487e9360e52d1431ff42\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5498ec7ccbb49c98c615c1d3ce8435b : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e23b73fa8c8475699fdae9a1d9bf293 : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1778e0c201464ee1b28b57512ca0a716 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89812defd7a4410790a89c087234f577 : SUCCESS in 12m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b0cadcc7802474695eb3fe0fe6157c2 : SUCCESS in 4m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2f49a46a03e64cfab7eac7e665ca99f6 : SUCCESS in 4m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ec458edb5d4f45e6b70b9d1497c6bb7c : SUCCESS in 7m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c446bc010b44438cbd2e4e68aba050dc : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8def7418311847bfbf46b688d1b291bf : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/db7c2dd5e6c44f1cade2061233d3a668 : FAILURE in 49m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5d2467290d56456ebeb64b29fbe91832 : CANCELED","accounts_in_message":[],"_revision_number":3},{"id":"7977af44da558d9133e1258e99ce0c7127f64c19","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-21 19:29:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1ef07b6cb68f6eb95c5ab588259e4361ec2f2de4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 22:08:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bdf8e10379a4fb8a39b6fbabd8ba558\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdfb9500b4b942b78a6bec0adf1012bf : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c70e19e294b3450498c5262e24a465dc : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/881f59521c444713a778e2eb4985f0f3 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b2c47def2884dc8afe9d6dd6bc467ee : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8250e9121f324c2486db72535be42e1c : SUCCESS in 12m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1974bc672c74404bb56294c1aaee247 : SUCCESS in 5m 21s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/35cbed98c39c4acdb455660d9ee6eade : SUCCESS in 4m 56s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fe8b098b502249f88e86e5295688eb8b : SUCCESS in 6m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3afaa4bf491f49dbae8c570a99d9ab11 : SUCCESS in 4m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e69798c09af14ec8b7716dbe6d2edbad : SUCCESS in 6m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/42a8568a217c421d81c89dfe945839bc : SUCCESS in 7m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e2a98b0ebfbc4e219e04d8cc8deb09fd : SUCCESS in 7m 32s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/31a52c08768946ed8fd6b9c3e41b2a47 : SUCCESS in 1h 05m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/012764747abb43198d82bd588b6026e5 : SUCCESS in 2h 24m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7c8d63839a874eb480a55c960ec8abf2 : SUCCESS in 42m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a367e10c9cfc42998a1d9be2b01d1f74 : SUCCESS in 44m 30s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d2eae276490c4bf3839bb0d11a092396 : SUCCESS in 1h 04m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e0ca0271a0834ee68cc7f61be72e9b4f : SUCCESS in 33m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d35c6d245c474ed1b7f9d4590de6e393 : SUCCESS in 2h 03m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b099546bbdc24883b23cd861f51d52ee : SUCCESS in 2h 36m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2903294848ee4b25a198e78bb7879271 : SUCCESS in 2h 05m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9295abbb01054a31b1849d90b2363124 : SUCCESS in 2h 25m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/584a77261d134fa3a33d3e9d8f2f4ec9 : TIMED_OUT in 2h 32m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3627ebbe60c4d8499a7d8b201c75fe5 : SUCCESS in 2h 12m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9a7d7a5874451ddb756287fe426d1a76bffdaea1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 22:09:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9368faa4b7e2b904fb6f4eeb6e6cd3c568491b5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 22:30:48.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b885d21eef9448f9d4ebd958b2eedc8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a902fa3ad22a4950aa6552d7c2112aaf : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b583dfb31dc64a60a3c6f2f67527ea3e : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d401a733bee14f9e829a6a0e92ecb053 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2273ec18e5e34affb6e2b2e8b45f7c7f : SUCCESS in 11m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3e1bc7271ee49e08ae2c885ed978ee6 : SUCCESS in 4m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d33301df794f42bfa7883f07969954fb : SUCCESS in 5m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/50be07a597474717807821b382bc29c0 : SUCCESS in 6m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/037092a75fd8410886b8b5db088ed11d : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3b78aff077647b9a99a00488654b80a : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/38cdfb01c4824d40ae41e01ca6c90845 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8c3050c98e8942d497c612efeaf1b454 : FAILURE in 20m 36s","accounts_in_message":[],"_revision_number":3},{"id":"2cf3f9088af3c656f9099183f5aadac31ad78bc6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-22 07:31:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5ed1697fc40f72d958f5185aaee347a382ef70be","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-03-24 09:13:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cd09c12d398f759cdef3916246b8bef2d6578ea7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-24 09:17:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dbcaa45bdb03a356a503e0061455855832200dc1","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-03-24 09:17:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4e35779328b6e3d347bdc7a5a6d0e9319e807733","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-24 09:23:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"49b687617945bea676fb18fc01f505b89b97a627","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 12:22:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3642994ff793427d956463bfa47ecb43\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db1d84d1c89845e5ae8e91cf7536d4e1 : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5701654ff9be4cc09b89a9f23413e833 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a39f8e922444d12963baeed0b01ad8b : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68f0649fdd29451b866649cad2fe751b : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db2351231d91473f926f4c29209ce43c : SUCCESS in 12m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b31914393a74313a9d55a5db9646157 : SUCCESS in 6m 07s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8810f9f7b0c447719d09d77f1b1cadd3 : SUCCESS in 6m 10s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e1026efa9df14faf88752da7cff0f436 : SUCCESS in 8m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/00e5053424e34f78a69ee7cee0d13151 : SUCCESS in 5m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2bc38571e39e4e96b09d0acf045974ae : SUCCESS in 8m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/18ffa55c476d4f0ab745ae3fcfc9e4d7 : SUCCESS in 8m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a39a8875ec914b4d999ffaa11f9d4193 : SUCCESS in 8m 18s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/68fb583b4bff49bc9e2450d244f9c7ef : SUCCESS in 1h 16m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/32d10210b36f4799a41b1570490b5a4f : SUCCESS in 2h 13m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5a0a52146f0444d9b5befecad892b7c1 : SUCCESS in 37m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2e2cc2f0554448af871e59af24eba479 : FAILURE in 20m 05s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/cf370d10f1c54b868ebd9160c8b9791b : SUCCESS in 1h 15m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/52289a6ad5224b41a6398d3ed626853a : SUCCESS in 33m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/30f2b144e3694982b8c1bbe700b11318 : SUCCESS in 2h 13m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/82719e96842b433896703c5fd8449067 : SUCCESS in 2h 56m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9b10d28c5cb74172a5d6811927491542 : SUCCESS in 1h 43m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/05d94c4fd9804e7d9bb6e3d91dad39db : TIMED_OUT in 2h 32m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d5d8a36b13eb4c90b86bb570e15ea0af : SUCCESS in 2h 11m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/84826484ab744e28b467c5e0b84bbf11 : SUCCESS in 2h 11m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4c86b66cd0fd5c79250f141909fe95fe3a9e0251","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-03-27 12:26:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d1bf723e63a35e62523cf4721e378455c6c819c5","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-03-27 14:48:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cdd4e645012ea7d0998286fefd33a79ee873f06c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-27 14:51:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5c05ca19d96a91cc740d01380814ef13c1bbe5ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 15:02:38.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e435e063485940bca45617bb5361c9e0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13f326648fa343de85cd97621f26e8ea : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa7d6ab7ed5246b8ab2d3aba0e6c5194 : FAILURE in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5580cbedd52d48c9ace84b81d26bf7fb : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9bf5bfd414494b7eabb08f1b0b96b875 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a5f16a613a241fe86757518abad0c93 : SUCCESS in 12m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3edd0ad686ce430496d49e62a1c08a0e : SUCCESS in 4m 29s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/2c2851cb5aba4a098482394307a6a2ec : SUCCESS in 5m 04s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/366d359a3ec14c1595cea37afee0ba32 : SUCCESS in 7m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1d0f97d02735442797eb2e28ea97aebc : SUCCESS in 3m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2ef2d0ec072142f3b83b150cb90ea2d7 : SUCCESS in 6m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8dacff3e2a1f4f6ea2824457020ca653 : SUCCESS in 7m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b4885cbf538447a782d8925cdcb4b3b2 : SUCCESS in 7m 08s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a80ea7bace604dac93d3cb794ded0bf8 : SUCCESS in 1h 54m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a9aa63f0209542adb42c04f0cb09c87d : SUCCESS in 1h 52m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b1020e4c355b448aa28b181a33f2551d : SUCCESS in 33m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9d0e757a7b254010afef1e1cb2c25750 : FAILURE in 28m 25s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/75da8064a86e48d9af6a5a3b41ab376b : SUCCESS in 1h 09m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f9829701ea7463a8f2422d94e1d82ab : SUCCESS in 30m 48s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e76e87995eae42e0abcb9ae5004707af : SUCCESS in 1h 56m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9f88276e55194c37be58c80f212cf665 : SUCCESS in 2h 03m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ff6e23aae67a44989560b3c676bc3575 : SUCCESS in 2h 23m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/328a3277565d4d9b864bd790718dd7b5 : SUCCESS in 1h 56m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4be76a2a68fb4619ac52094e93832eae : SUCCESS in 2h 26m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b31de1c1ef849e5adcd61e6cc771797 : TIMED_OUT in 2h 33m 27s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/plug.py\n  Comments left for invalid file octavia/amphorae/backends/utils/interface.py","accounts_in_message":[],"_revision_number":3},{"id":"0b90e532e5539d3824467c93beb64ccb046443ae","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-27 18:23:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"845c0e5ee867ca445fdb6f36fe49a6bfa9b81dc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 21:28:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5e516d4008e4179999e9fcf64b3b420\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9a60d7670ef4bf9ba51948852142312 : SUCCESS in 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2fec1e821154a12a061c1dfb50150f7 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9cf72e651d08474c9c22189b58fe71c8 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b125d5a3a28248f2abaa8beafa219889 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79d2ed77dc224f759e79c010839d042c : SUCCESS in 12m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d85c849f2b474cb599c5f7907cdfc48f : SUCCESS in 5m 15s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0bfd67d308a64d248e5e51dc9464fb53 : SUCCESS in 4m 49s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a36e1f9896c546468a14b9014f48b70b : SUCCESS in 9m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a7f6598011874155866924fe5609582f : SUCCESS in 4m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a172977b1c19479d95126ee284e06a96 : SUCCESS in 8m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cd2841e9de254f338ee8b6ce177fb86c : SUCCESS in 9m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e5cd11aff21040b5a676a45cca15f1be : SUCCESS in 7m 14s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/10e076a6ce1f421b973b8e69db420fa3 : SUCCESS in 1h 58m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4649231690c1430484db8b4faa478a37 : SUCCESS in 2h 09m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9a0b6c955a4c486e9626fe1301ca8cd1 : SUCCESS in 40m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/77979c5ad8e742c1b2501d84e6a6ea7f : SUCCESS in 45m 44s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/78294e1cff0246d58e5fdaaccc5c1b57 : SUCCESS in 1h 11m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8183544fb4a249bc88c9e79d84fb4b1e : SUCCESS in 35m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/96c74696cb474bb997a162882e16af38 : SUCCESS in 2h 33m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fa1c3e7155d049288ef208161e83c9a5 : TIMED_OUT in 3h 03m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2e5791fcd66a4fd889bc6bc74f9c74a5 : SUCCESS in 2h 34m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c8acad7dbf174fbab0c72e45a7e9acff : SUCCESS in 2h 07m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/de6059790ebb409288953975c09ae7b6 : SUCCESS in 2h 14m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/21327f99e08f4bc99212f8c060d85161 : SUCCESS in 2h 15m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"550796ca0e73d690d0235379176e34506189f3ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 21:29:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f8a178d7ae65ca6c04309967196861e15b77e507","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 23:25:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a87cd1563e294c06904165fe01e05958\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6cb68455f6d4593a48bd2c896e6ff1a : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea1eda1ff0534c9a84fd1ec11aba76ab : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f0575aba3104f7c8a657618bee760c1 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c365228e38d04685b5c760d6c9c017fc : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45c39f8018454faa9cfa9eb408657bef : SUCCESS in 4m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7427dd7646da4c27862e4f73f0a8934b : SUCCESS in 4m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/873fb5c27f1c4e7aa8d6223a4dd5fecb : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b41de3ad9ec24bcc8c6f61eee02d0dff : SUCCESS in 1h 30m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a697b20e88c44242a0cd57362d752880 : SUCCESS in 1h 53m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/89004988366a441faea8e1715998ba71 : SUCCESS in 36m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/77ef53a6416b4f6ca977bc4b8373c0f9 : SUCCESS in 39m 41s","accounts_in_message":[],"_revision_number":3},{"id":"6744742c5fd9d9515510374ae5b04bfedb4eeb80","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 23:25:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6da67be917311ff10c63a34b67629fd7d4dd17f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 23:26:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae0d46bc90894776b752a3b5c80a0b38\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acde2c611fe34c6386184010b4eb4566 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3076cdfe1664402882ba48680e9576ac : SUCCESS in 36s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a460698cd933112529d1eae27d483da6b3035eff","revisions":{"5d61d2bcb4d33349a26c476411a3d746fbcdc0a0":{"kind":"REWORK","_number":1,"created":"2023-03-20 14:12:28.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/59/877959/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/877959/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/877959/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/877959/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/877959/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/877959/1"}}},"commit":{"parents":[{"commit":"764cbf014f494f28de5dfdae3876afec2f5f705d","subject":"Purge some dev tools from the amphora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/764cbf014f494f28de5dfdae3876afec2f5f705d"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-03-20 14:09:52.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-03-20 14:12:26.000000000","tz":60},"subject":"Fix octavia to accept [ipv6]:port","message":"Fix octavia to accept [ipv6]:port\n\nA host identified by an Internet Protocol literal\naddress, version 6 [RFC3513] or later, is\ndistinguished by enclosing the IP literal\nwithin square brackets (\"[\" and \"]\").\n\nThis patch fixes octavia to be able parse and\nuse this format.\n\nStory 2010655\nTask 47687\n\nChange-Id: Ib1e00a0b45776cfab80b9effb2690e12fea63e78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5d61d2bcb4d33349a26c476411a3d746fbcdc0a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5d61d2bcb4d33349a26c476411a3d746fbcdc0a0"}]},"branch":"refs/heads/master"},"24e574f7b5331ac315e216ed7cf050aae822ba5f":{"kind":"REWORK","_number":2,"created":"2023-03-21 09:45:08.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/59/877959/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/877959/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/877959/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/877959/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/877959/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/877959/2"}}},"commit":{"parents":[{"commit":"764cbf014f494f28de5dfdae3876afec2f5f705d","subject":"Purge some dev tools from the amphora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/764cbf014f494f28de5dfdae3876afec2f5f705d"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-03-20 14:09:52.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-03-21 09:44:58.000000000","tz":60},"subject":"Fix octavia to accept [ipv6]:port","message":"Fix octavia to accept [ipv6]:port\n\nA host identified by an Internet Protocol literal\naddress, version 6 [RFC3513] or later, is\ndistinguished by enclosing the IP literal\nwithin square brackets (\"[\" and \"]\").\n\nThis patch fixes octavia to be able parse and\nuse this format.\n\nStory 2010655\nTask 47687\n\nChange-Id: Ib1e00a0b45776cfab80b9effb2690e12fea63e78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24e574f7b5331ac315e216ed7cf050aae822ba5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/24e574f7b5331ac315e216ed7cf050aae822ba5f"}]},"branch":"refs/heads/master"},"a460698cd933112529d1eae27d483da6b3035eff":{"kind":"REWORK","_number":3,"created":"2023-03-21 10:31:05.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/59/877959/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/877959/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/877959/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/877959/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/877959/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/877959/3"}}},"commit":{"parents":[{"commit":"764cbf014f494f28de5dfdae3876afec2f5f705d","subject":"Purge some dev tools from the amphora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/764cbf014f494f28de5dfdae3876afec2f5f705d"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-03-20 14:09:52.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-03-21 10:30:57.000000000","tz":60},"subject":"Fix octavia to accept [ipv6]:port","message":"Fix octavia to accept [ipv6]:port\n\nA host identified by an Internet Protocol literal\naddress, version 6 [RFC3513] or later, is\ndistinguished by enclosing the IP literal\nwithin square brackets (\"[\" and \"]\").\n\nThis patch fixes octavia to be able parse and\nuse this format.\n\nStory 2010655\nTask 47687\n\nChange-Id: Ib1e00a0b45776cfab80b9effb2690e12fea63e78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a460698cd933112529d1eae27d483da6b3035eff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a460698cd933112529d1eae27d483da6b3035eff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
