)]}'
{"id":"openstack%2Foctavia~878371","triplet_id":"openstack%2Foctavia~stable%2Fyoga~I53ed0425f00cfbbc964f5083b4f2942cf5bf565e","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-03-27 11:56:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I53ed0425f00cfbbc964f5083b4f2942cf5bf565e","subject":"Fix template issue with user_data_config_drive","status":"MERGED","created":"2023-03-23 12:51:28.000000000","updated":"2023-03-27 11:58:53.000000000","submitted":"2023-03-27 11:56:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878371","meta_rev_id":"7aa9284553915722c7950e057ab559ccd6460da7","_number":878371,"virtual_id_number":878371,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-27 11:56:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-03-23 16:15:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-27 10:16:53.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-27 10:16:53.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-23 15:56:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-23 16:15:06.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-03-27 10:16:53.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"17c015dbc6fabfcc5b37c01b07fa7d60f8f460d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-23 12:51:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1365fca8cb41332318403c4138ffcb48969f9d3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 15:56:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc142c3d6289448ab65dd19314128dea\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b910593b2a404832b9c0960ac4899c62 : SUCCESS in 9m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8662754e0680423a99e9c46788e96d3e : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d0ac3dd29c64e539e16adae3ba26f61 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e0fd244c23d47e0a504bb73400a9e8f : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26fffde99f2f4e6190f11cd71681559c : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f07579278f4e4cadb5047d76c9b806c3 : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d80b16160cdb42a98effb669b9e17e63 : SUCCESS in 4m 31s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b61d00e93285421b9b3585571dae63c3 : SUCCESS in 5m 20s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/86a785c2e2ec436e8a2d6a60cb752d47 : SUCCESS in 5m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7399fc1cebc24d459c9cbc0d23fe3c2f : SUCCESS in 4m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c0255ab31a374a76807399484848e402 : SUCCESS in 9m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/26a8e6994f63484fbc8b65f5bc73eec2 : SUCCESS in 1h 58m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ad1274fd33274e598e70d30bcefedd91 : SUCCESS in 1h 40m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b359169fc7784dd69d9ab94261d05401 : SUCCESS in 44m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/58dab019607a493988a425d4953cf316 : SUCCESS in 37m 47s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f2250fa0ae7e4efa8d04ccd475161251 : SUCCESS in 1h 07m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be6872d8fb5a4b4189855f489271c4fb : SUCCESS in 32m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/de8116625acb4bb0b808a13bf3759526 : SUCCESS in 1h 51m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1f6ad433141d4fd3bb3469a8dfc94708 : SUCCESS in 2h 56m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f175d2c5c33f4dd487a35f55965c22b9 : TIMED_OUT in 2h 33m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4849ca462e3843b5acf7951d98e035c2 : SUCCESS in 2h 19m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/86d8218eda204a27b87b1badabdd565f : TIMED_OUT in 2h 34m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b93ea5977fc4dab8d12f9ae08e6d51c : FAILURE in 38m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57da82544229c6c9b34a8337e63be9cea60908d6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-23 16:15:06.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba54d774069b59c19995d656368b8dd582a2ea2a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-27 10:16:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"99788f5576228af8648fa5b9e2caf663f33e5bb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 10:17:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"500cce1e143894dc5d117ea1e63659c66ae06deb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 11:56:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ad098ec7b5242a88e2ab1af499393e5\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd6141ec117f4210a1345d5ad04c574d : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f2349c16c884427bd75d9519f5e3143 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3819d85f3804b75bf9a6c0026ec3622 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9236b9766a6417596d8e374c435ba87 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03eb6377197c4461aaf5598e1d05c116 : SUCCESS in 9m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a63356b0d0543f5be22017ced6638e1 : SUCCESS in 4m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9e78a7d7025f4c419037af1a4798fd1e : SUCCESS in 4m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c40bb5aff04f4d04b8cdd349b52d2ce7 : SUCCESS in 8m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5f107ef3c972475193e42f25fe200f64 : SUCCESS in 1h 27m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ac1f6de5ad441dbad48078cafc8ae72 : SUCCESS in 1h 37m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/40ae06295b5d400184fb7784cd1ebaea : SUCCESS in 31m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2a6317ca0f9d44c6823469a25b0a69dc : SUCCESS in 35m 43s","accounts_in_message":[],"_revision_number":1},{"id":"2c872d1c62f06181d8cfde0793f84f6b0d0e4d84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 11:56:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7aa9284553915722c7950e057ab559ccd6460da7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 11:58:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19b1a9aeb66d4505bcb044f7d14009a2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15ba6fdc999f421b8049eb4ab348bb0c : SUCCESS in 1m 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd8af862f31e47c5b847d90cbf6b8f98 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b3458b7a92e70fb604032a8add72d16311936fc","revisions":{"3b3458b7a92e70fb604032a8add72d16311936fc":{"kind":"REWORK","_number":1,"created":"2023-03-23 12:51:28.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/71/878371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/878371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/878371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/878371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/878371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/878371/1"}}},"commit":{"parents":[{"commit":"cb8aec77ecb4783be9d48df091b767dd248a754a","subject":"Merge \"Fix PING health-monitor with recent haproxy releases\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb8aec77ecb4783be9d48df091b767dd248a754a"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-08 05:33:58.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-23 12:51:04.000000000","tz":60},"subject":"Fix template issue with user_data_config_drive","message":"Fix template issue with user_data_config_drive\n\nFix a python3 error that prevented to use the user_data_config_drive\noption when building amphorae.\n\nuser_data_config_drive has been broken since we switched to python3,\na mix of str and bytes variables were provided to the jinja2 templates.\nThis commit ensures that the type of all the certificates provided to\njinja2 are strings.\n\nStory: 2005553\nTask: 30705\n\nChange-Id: I53ed0425f00cfbbc964f5083b4f2942cf5bf565e\n(cherry picked from commit 64d8d991f16dd818a981daa34ce418265a907b28)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b3458b7a92e70fb604032a8add72d16311936fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3b3458b7a92e70fb604032a8add72d16311936fc"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
