)]}'
{"id":"openstack%2Foctavia~878372","triplet_id":"openstack%2Foctavia~stable%2Fxena~I53ed0425f00cfbbc964f5083b4f2942cf5bf565e","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-03-27 12:02:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I53ed0425f00cfbbc964f5083b4f2942cf5bf565e","subject":"Fix template issue with user_data_config_drive","status":"MERGED","created":"2023-03-23 12:51:59.000000000","updated":"2023-03-27 12:03:55.000000000","submitted":"2023-03-27 12:02:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878372","meta_rev_id":"a1ec008acda836c4deb8bc01e897552d7bd00003","_number":878372,"virtual_id_number":878372,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-27 12:02:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-03-23 16:15:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-27 10:17:07.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-27 10:17:07.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-23 16:02:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-23 16:15:30.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-03-27 10:17:07.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"048f8ca281c996ed4304b0e31e0708830b9e1b1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-23 12:51:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"100297701af4fd4328b03cc02b69b7263a493456","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 16:02:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a811d1e8705496e8a36c173096516f9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9c39b2fd4bb43a6b8fa29047e4ffcbb : SUCCESS in 11m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fce4b19610c24f43b91c933ce8fc24eb : SUCCESS in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55571acf9c4340f2a85b479c97a9c5ad : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88db29f96b4747b684805cf75ff02faa : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4315619f78f24df7b137124239b9a90e : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b29f94a7e6344b26a5ac6969c191e8ea : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce292e86aa6a4fd4ad2045e6fad9ce6f : SUCCESS in 10m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e529adac8bbc4751a6ff659a348f2197 : SUCCESS in 5m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/68c27986dd3c4b36bb43b1662617bc93 : SUCCESS in 5m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e1688a0d939f40359b22c4b8917425b4 : SUCCESS in 7m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/11515d87dfd749619f9c61e236e6adb4 : SUCCESS in 4m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/33d36212f7a246df96d493731a082bda : SUCCESS in 9m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/407bf92ce4e04271bcd44bbe8bef9ed1 : SUCCESS in 1h 15m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a2e5918f00d412fab1b6c74b2dd0c91 : SUCCESS in 2h 12m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7d26d94b3f034b5d8f7c54a6d69e8334 : SUCCESS in 36m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b6121c3ea94544899657678460c73dcd : SUCCESS in 51m 57s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3ab6d880334843b094d87ea200acd9d5 : FAILURE in 9m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4166c8f91ad64579b3a90333de14783d : SUCCESS in 32m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/02bb630c68ab4907a7959c2ff6b96782 : SUCCESS in 2h 16m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ca7375438dc44d1a9889c736265d3c5d : TIMED_OUT in 3h 04m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7c5dc1c08121411d8537700bbcd8e3eb : SUCCESS in 2h 11m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4edf1f9e04614e1aa6c57c6adab91339 : SUCCESS in 2h 23m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/142347fe47924bf2901f71f5cf4041f7 : SUCCESS in 2h 22m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fe202a768baf4e1bb7318371e67387d9 : NODE_FAILURE Node request 300-0020813609 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b5960441d65f24b065205208f0899afb3a37e463","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-23 16:15:30.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4cd6745de34cb6fe97cc6406673853dece12f5a4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-27 10:17:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d34936d7946a1684ed57ee86216843c28f7ec715","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 10:17:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"882f07c925a6964efa6f2d5f01182648644580ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 12:02:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f636b51cb8a4a0193b8675eaea873c8\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d683ab8d372c4663bdcbd0832048c5f2 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae65d43cf696493ca7fecdebdd0384c7 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abb8476b04c240bca16f4f90222eabbf : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73ddc0de91c0477f9369163d9f6a0d32 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9c9444745744af69ac2cbebeccfb02b : SUCCESS in 11m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c1dc7c631d44a36b54dc1235de3caf4 : SUCCESS in 4m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bdca83930f104458836a8347a92c73e1 : SUCCESS in 5m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6ef7ebd0d824444b8b4e83402c33f214 : SUCCESS in 9m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b255a1aadbe849e7b62e95923533ec2f : SUCCESS in 1h 24m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe4159f6aff043ec809b5b46dcecb72b : SUCCESS in 1h 44m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a003833029b84fa9a5007cc58b1bb295 : SUCCESS in 31m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a27d797a51684318bb834c64f1487203 : SUCCESS in 37m 50s","accounts_in_message":[],"_revision_number":1},{"id":"4a6755f1097c8b7449618fc90c8f7f7890a13050","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 12:02:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a1ec008acda836c4deb8bc01e897552d7bd00003","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 12:03:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/819977c921d548d9b45e27df872a06bf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c9a5881d4dc42a885779045a6566e95 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/712468f374524f819cd89b3f0b6100a1 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de74774df0eb818e5832195d28a288e927a5b93f","revisions":{"de74774df0eb818e5832195d28a288e927a5b93f":{"kind":"REWORK","_number":1,"created":"2023-03-23 12:51:59.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/72/878372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/878372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/878372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/878372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/878372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/878372/1"}}},"commit":{"parents":[{"commit":"61102e0a142d6d93ba1e643943a55c37df519617","subject":"Merge \"Fix disabled UDP pools\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61102e0a142d6d93ba1e643943a55c37df519617"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-10-08 05:33:58.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-23 12:51:52.000000000","tz":60},"subject":"Fix template issue with user_data_config_drive","message":"Fix template issue with user_data_config_drive\n\nFix a python3 error that prevented to use the user_data_config_drive\noption when building amphorae.\n\nuser_data_config_drive has been broken since we switched to python3,\na mix of str and bytes variables were provided to the jinja2 templates.\nThis commit ensures that the type of all the certificates provided to\njinja2 are strings.\n\nStory: 2005553\nTask: 30705\n\nChange-Id: I53ed0425f00cfbbc964f5083b4f2942cf5bf565e\n(cherry picked from commit 64d8d991f16dd818a981daa34ce418265a907b28)\n(cherry picked from commit 3b3458b7a92e70fb604032a8add72d16311936fc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de74774df0eb818e5832195d28a288e927a5b93f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de74774df0eb818e5832195d28a288e927a5b93f"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
