)]}'
{"id":"openstack%2Foctavia~878375","triplet_id":"openstack%2Foctavia~stable%2Fxena~I59664615e1ef5e46a99c12256dbe02f89ecf4c0b","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-03-27 12:26:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I59664615e1ef5e46a99c12256dbe02f89ecf4c0b","subject":"Fix update listener certs doesn\u0027t work","status":"MERGED","created":"2023-03-23 12:53:46.000000000","updated":"2023-03-27 12:27:48.000000000","submitted":"2023-03-27 12:26:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878375","meta_rev_id":"7be44ac69b57c67b11c47efa897b7736e9393e9b","_number":878375,"virtual_id_number":878375,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-27 12:26:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-03-23 16:12:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-27 10:20:02.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-27 10:20:02.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-23 12:53:46.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"CC"},{"updated":"2023-03-23 15:34:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-23 16:12:54.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-03-27 10:20:02.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"0df99da6cd852e7fdbb243c7a69a153f2c8f289e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-23 12:53:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ec54cc0c99b413d7151c339bee7e1690b833d4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 15:34:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93fb57ff874f4bd1b9b08c9e630d51e1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98a3702dd6e24f9db9c80358124a6e99 : SUCCESS in 9m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94a8f03029b44059a1f21e78dda11002 : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c8518fd561f4c338691887c09385225 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19943ac39fbf4db69316a3d8cfecdbc0 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd861b32ef7d4242a4d9cfb175e89fa4 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54278a4ecce6415ea8a9dfa00c74575d : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61cb976657204178afcb733446b02476 : SUCCESS in 14m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4084f75f0460460f833ef86adb761413 : SUCCESS in 4m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3855d53a5e3e44c283427d95a39811d3 : SUCCESS in 4m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c281cc40af514e1bbfad0acf203a800e : SUCCESS in 9m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/98ffcb1afb9449be9d50c002e68c8a18 : SUCCESS in 4m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b76bba08186944bd9c03240f29d2f596 : SUCCESS in 7m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1d5eb96026964058a26ae9c219191fb2 : SUCCESS in 1h 58m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5726a636af0b4a2380cc8b3795f68f9f : SUCCESS in 1h 54m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4404b6f98b074952acf9a51ada5f190f : SUCCESS in 38m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f712379b3b454d3ba358429e4148a014 : SUCCESS in 55m 47s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e358059de07c4f218eef14d15735b9be : FAILURE in 10m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/99a440914bd34cc996974620ddf92883 : SUCCESS in 33m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f4594acc026d4e5ead2f2e50c592747c : SUCCESS in 2h 18m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fe37b7239da84635af14e662c5016e64 : SUCCESS in 2h 24m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8aa8229d468a4987a93f836f39737ac3 : SUCCESS in 2h 23m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/636389d303034176af9c1167c5db063b : TIMED_OUT in 2h 33m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8bcd3e752d1c45c4a0e062181b052c81 : SUCCESS in 2h 12m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/491bd089d8ba4a9ba109f3ec3d5f156a : NODE_FAILURE Node request 300-0020813694 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d90bd7161e35406c400f96b1ad4cfbc8c82e6f85","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-23 16:12:54.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a8046d56c76cb74d4b59fa1bea80ab174334eb61","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-27 10:20:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0fd5e87508dba84b4213b20675c6d9e03c21b9b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 10:20:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95dfb43ed5b1c71b49772e4ad88a0c34a4328184","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 12:26:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d3c1139c7bd4eb2a1adbc4102a34c0d\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dcd512c892db45d39d3bd8dfb383745b : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95e5489787444df3aa2f0541563e38fc : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5730eb9dede43399e2b98939685dee6 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c854dd52521487ab3051a12c1f4b01f : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39073813bdb949c9b2819134673ea188 : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00904d95933c4847baef080109a8d9b9 : SUCCESS in 5m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3ff4691f9a394880b5d78aab20fa6654 : SUCCESS in 5m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c613f04dd1854f078c7ad28a36578f85 : SUCCESS in 7m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/127c86ef7f4549e2b12efd7c068a7d37 : SUCCESS in 1h 18m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/401ee5236ba442159a3b4bfa6ffcae0c : SUCCESS in 1h 46m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6a2561f000ee4562856b73b2f501121e : SUCCESS in 31m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a39a9df2be6c48ef86a0332eed6de384 : SUCCESS in 37m 21s","accounts_in_message":[],"_revision_number":1},{"id":"b26ba412cfa05b04ac641f2601f19dd3ab72066e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 12:26:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7be44ac69b57c67b11c47efa897b7736e9393e9b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 12:27:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee8e691b76c04c27bc025f650c2e5bcf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/280b8af383e146a4b102443a695a0a57 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2dc543bf2744ad9ad680117c359eda3 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"170df2ba57082e5a6100a3f11619378958e34ea6","revisions":{"170df2ba57082e5a6100a3f11619378958e34ea6":{"kind":"REWORK","_number":1,"created":"2023-03-23 12:53:46.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/75/878375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/75/878375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/878375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/878375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/878375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/75/878375/1"}}},"commit":{"parents":[{"commit":"61102e0a142d6d93ba1e643943a55c37df519617","subject":"Merge \"Fix disabled UDP pools\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61102e0a142d6d93ba1e643943a55c37df519617"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2021-11-17 14:43:42.000000000","tz":480},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-23 12:53:37.000000000","tz":60},"subject":"Fix update listener certs doesn\u0027t work","message":"Fix update listener certs doesn\u0027t work\n\nThis change fixes listener update sni-container-refs\ndoesn\u0027t work by adding O_TRUNC flag to upload_certificate method\n\nChange-Id: I59664615e1ef5e46a99c12256dbe02f89ecf4c0b\n(cherry picked from commit d340a1de50b55a73f0e3a4d398ee735b0ccc128b)\n(cherry picked from commit 524ce50564466dfd08f51fcfc0281d5b3850518c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/170df2ba57082e5a6100a3f11619378958e34ea6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/170df2ba57082e5a6100a3f11619378958e34ea6"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
