)]}'
{"id":"openstack%2Foctavia~878376","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I59664615e1ef5e46a99c12256dbe02f89ecf4c0b","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-03-27 12:28:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I59664615e1ef5e46a99c12256dbe02f89ecf4c0b","subject":"Fix update listener certs doesn\u0027t work","status":"MERGED","created":"2023-03-23 12:53:57.000000000","updated":"2023-03-27 12:29:39.000000000","submitted":"2023-03-27 12:28:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878376","meta_rev_id":"503f4f979afcda499f63349838859d29d7d2bb97","_number":878376,"virtual_id_number":878376,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-27 12:28:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-03-23 16:13:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-27 10:20:14.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-27 10:20:14.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-23 12:53:57.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"CC"},{"updated":"2023-03-23 15:36:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-23 16:13:08.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-03-27 10:20:14.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"7022420a189ffa609faf8d5410e7ce5933e90264","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-23 12:53:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30d1df815da650ee2b8dca34d65cffd373e33eae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 15:36:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b861ed2c216436dabfd95e29a6b8f71\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd8f318b9a7d442fad9a4556a14d65f9 : SUCCESS in 11m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d609a39ca2004211a83f88e25629ea38 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f960a6b4023e45d6bfa84640213fc368 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d7a4e40b2804f198250c6313c6de468 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/127959346bd642cca4c6887e42699a07 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20f7f9727b1e468db6c122084a8bb2ac : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb3899e34bbf47b78b7e146a1472c084 : SUCCESS in 10m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a9f8434fa964826987e437c49668c94 : SUCCESS in 4m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d2011e69341f496c87dc852449ad67e4 : SUCCESS in 6m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6a8a4f8141a54a899402a0899e73f0f6 : SUCCESS in 7m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8a4195916939470fa0e803cf6c7a8419 : SUCCESS in 5m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/962a50246f0d4e578c4cbd21d0635235 : SUCCESS in 6m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b95327ec3b2a4e41a5a87d61bf7639db : SUCCESS in 1h 04m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c14bb21b6dd749789c2e3ffb49b8f5bc : SUCCESS in 2h 01m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cb412f89772f478fbbddaba178c85de0 : SUCCESS in 39m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/45b79ad6e97049e398e6ebb0c628e4fd : SUCCESS in 41m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/206e6dc79048453394f1ac5a9d6dd9a8 : SUCCESS in 40m 14s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/659afc1e66704e828b0554f561cb4b34 : FAILURE in 10m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8833d8f64d1b4e1e8571998c7d9d1056 : SUCCESS in 1h 53m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/81a5f95089e94c0985b466b3667219cd : FAILURE in 32m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/290d0a5245a04ef9a186c46de35fd2a4 : SUCCESS in 2h 28m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fe8b5f8dfbdc4ecf81a2ff6691a7c037 : TIMED_OUT in 2h 33m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0c47d1421f2e4469b6242624355d5774 : SUCCESS in 2h 25m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/71d5d8e5b32d4ec7971874ac6bf609e4 : NODE_FAILURE Node request 300-0020813718 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"43c43c272ce250c1cb4949388ebc8472bec696c7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-23 16:13:08.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4efb397bfa95d6a6cb824937b68cbc028e4b7eb8","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-27 10:20:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e99951efad6f490506dea1df05bc6866e5579fff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 10:21:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"380c4238d0f1ed58d6b1a57a7a6b8d7371d6d738","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 12:28:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e8db71a5feb4cf18b908503371a2a76\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57b8bfe794174b2b934fcf327636d6e7 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7596a21fc1b466f9a37c7598cac626f : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8df3daf0e4574d4c881c153e6ca812fe : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/205bcf66c7b3466f99c612e5b79dd4e3 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b56e6b05729d42e1bbb13669db97ea30 : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4adad02650f430b906c56482707a748 : SUCCESS in 5m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/afa86955dfa447b2bfd78678c81ae674 : SUCCESS in 5m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e9de8dba39cd4e1eb37e83316efc92f4 : SUCCESS in 7m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f38c1b4de00b46b4830e6dbed4bcffcd : SUCCESS in 47m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a9a835562ce45218caf4fa469f29144 : SUCCESS in 2h 03m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/39a34912e53341bd993c0d25de7dccb4 : SUCCESS in 37m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/42fa5c4793284e4b98e9bc7939eea623 : SUCCESS in 33m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2d748612a23240368ca24c66a7c5ba0f : SUCCESS in 35m 21s","accounts_in_message":[],"_revision_number":1},{"id":"c778587736c71a8377e3878a9e5165abde27cd9c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 12:28:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"503f4f979afcda499f63349838859d29d7d2bb97","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 12:29:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2f78785c51a41d3850af5ac5bbf3a99\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d875f3a223a46ccbd4bbe6cc5b49efb : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5be39233eae4e199ac2231cc723086c : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fe91e1e7368641fad6627c830253defb320bba7d","revisions":{"fe91e1e7368641fad6627c830253defb320bba7d":{"kind":"REWORK","_number":1,"created":"2023-03-23 12:53:57.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/76/878376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/878376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/878376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/878376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/878376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/878376/1"}}},"commit":{"parents":[{"commit":"f6d2e59334aad55ab884e85a8cd697c7bc3bf861","subject":"Merge \"Fix PING health-monitor with recent haproxy releases\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f6d2e59334aad55ab884e85a8cd697c7bc3bf861"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2021-11-17 14:43:42.000000000","tz":480},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-23 12:53:53.000000000","tz":60},"subject":"Fix update listener certs doesn\u0027t work","message":"Fix update listener certs doesn\u0027t work\n\nThis change fixes listener update sni-container-refs\ndoesn\u0027t work by adding O_TRUNC flag to upload_certificate method\n\nChange-Id: I59664615e1ef5e46a99c12256dbe02f89ecf4c0b\n(cherry picked from commit d340a1de50b55a73f0e3a4d398ee735b0ccc128b)\n(cherry picked from commit 524ce50564466dfd08f51fcfc0281d5b3850518c)\n(cherry picked from commit 170df2ba57082e5a6100a3f11619378958e34ea6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fe91e1e7368641fad6627c830253defb320bba7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fe91e1e7368641fad6627c830253defb320bba7d"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
