)]}'
{"id":"openstack%2Foctavia~878378","triplet_id":"openstack%2Foctavia~stable%2Fyoga~Idd6480bc0a7de7c42b34601172c2dae50077aa48","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-03-27 12:28:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idd6480bc0a7de7c42b34601172c2dae50077aa48","subject":"Fix unclear error messages in the API","status":"MERGED","created":"2023-03-23 13:06:14.000000000","updated":"2023-03-27 12:29:41.000000000","submitted":"2023-03-27 12:28:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878378","meta_rev_id":"da75a5432db855fbf97bd4931b9df581b22e7ecd","_number":878378,"virtual_id_number":878378,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-27 12:28:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-03-23 16:14:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-27 10:26:20.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-27 10:26:20.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-23 15:47:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-23 16:14:52.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-03-27 10:26:20.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"6c3fb5fd846cae337004cf013529dccd40582039","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-23 13:06:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c01280551c65f222e2612661978bb5145ce9c66b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 15:47:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1244e4d6713a4d90a02d3b1a9060ec07\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/947b3e31313b4000b44d11385aeb1f11 : SUCCESS in 11m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1b932bbee2d48e2be2b9b6cca1e8c8b : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ece7f5a0381141359005e170e280482e : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bccf8a258084a098e5658b5cc74b51f : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1da1dd20d8f4bb0807089d5d1283284 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cb3d29625814318ae0b3f3290379465 : SUCCESS in 10m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/168bc12518c844a4912b9200525b5c16 : SUCCESS in 5m 22s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6d7906fe38884c98b61ee1dbcf42d847 : SUCCESS in 4m 48s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/20397618e0f446b193aa3fa012038222 : SUCCESS in 5m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a453f340c9804ea5834fea5660c7de04 : SUCCESS in 4m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a9a0b15206fc466585615cc7404eb6ea : SUCCESS in 7m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a13972e10be248b8b54a15bd575c7d81 : SUCCESS in 1h 29m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac9887f5abe1440f9a29c426e3becaf6 : SUCCESS in 2h 20m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/17b9abf6391f493bb7bf525cc032b86c : SUCCESS in 38m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c7266d01156449e2bd7c12d9dfaa5eb8 : SUCCESS in 59m 43s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/029880e5744c41cbb71b8d60c437828a : SUCCESS in 57m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a490b921dee54d68a4abe9df2b61a7c1 : SUCCESS in 48m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c920feef090e4a529e77e0a6423cfcec : SUCCESS in 2h 15m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a8b38152d739464b953721f021dee566 : SUCCESS in 2h 25m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/899268c68de842e98d21a8915aed4d6f : SUCCESS in 2h 29m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/00b35d2c5fa14af3abac81bd3a1a4e6a : TIMED_OUT in 2h 33m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5d269057450e492788d27b5687d01ba5 : SUCCESS in 2h 32m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/049cc565c704430b99fb62d6b9aed7f9 : FAILURE in 40m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a77685951ff3553e25b0e3320206de7fe2e5ef64","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-23 16:14:52.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1fc5a4ac1ef66e25f2dff53416ffc93bd6ccdf1c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-27 10:26:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4d2c58dd2f2d657e830e2df6c971cbeb4e477888","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 10:27:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fed344f0cfbd281de8f4ce13fbb73de8c9ab284f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 12:28:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa041e952a1b4fe69735135a93209809\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61781eeaed014e4594b9b865b4ff6035 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bd75bac968c404eb2d9812f9a00c250 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f49e215c5084ddfa6201e1286741c25 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/431bc13370dc4b6d842abc574dfc7584 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/705e852cd8dd417fb6842e6b25cc8666 : SUCCESS in 11m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c6113f5c04d44f1ab60f44e2706961c : SUCCESS in 5m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/76caf77f08f6408a9229671ce1135752 : SUCCESS in 5m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/128ab72b9ab641b0a98b9714cd47c2ff : SUCCESS in 7m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cb9aa3cf9bce4fe98eeec9442ccbcdea : SUCCESS in 1h 39m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b1e58674a8534cbf98249160fff29d9c : SUCCESS in 1h 56m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/853329701f2e43a39ae9e0f49d759a62 : SUCCESS in 36m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6b168e51bde6413f8ee1493a5eccecca : SUCCESS in 40m 45s","accounts_in_message":[],"_revision_number":1},{"id":"abff247c192de96369cbd7e9ceb9e3be65f24b2f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 12:28:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"da75a5432db855fbf97bd4931b9df581b22e7ecd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 12:29:41.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1fa14ae438a6420e8c1ed3249ab1536e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a577cf0209594ddca397016831394b28 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7116eb8fed634606ac8e218ceda58b69 : FAILURE in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d3191eb939e4b3810878755273654ca91b35107e","revisions":{"d3191eb939e4b3810878755273654ca91b35107e":{"kind":"REWORK","_number":1,"created":"2023-03-23 13:06:14.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/78/878378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/78/878378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/878378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/878378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/878378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/78/878378/1"}}},"commit":{"parents":[{"commit":"cb8aec77ecb4783be9d48df091b767dd248a754a","subject":"Merge \"Fix PING health-monitor with recent haproxy releases\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb8aec77ecb4783be9d48df091b767dd248a754a"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-04 09:20:20.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-23 12:55:17.000000000","tz":60},"subject":"Fix unclear error messages in the API","message":"Fix unclear error messages in the API\n\nSome API calls may have returned useless error messages such as\n\n{\"faultcode\": \"Client\", \"faultstring\": \" is not a valid option for \",\n \"debuginfo\": null}\n\nwhen a DB constraint failed.\n\nThis commit improves the error message by reading the name of the\nforeign key that failed from the exception.\n\nThe value of the parameters should have already been validated by WSME,\nthis exception is only thrown when there is an issue with the database\n(corrupted DB, empty table or missing fields in a table).\n\nStory 2009957\nTask 44919\n\nChange-Id: Idd6480bc0a7de7c42b34601172c2dae50077aa48\n(cherry picked from commit 57e006eac321b210c98917019572b3576aefd97e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d3191eb939e4b3810878755273654ca91b35107e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d3191eb939e4b3810878755273654ca91b35107e"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
