)]}'
{"id":"openstack%2Foctavia~878583","triplet_id":"openstack%2Foctavia~master~Ie26a154c320d8527e54b871a3b99bbbc3c2ee811","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-03-27 17:41:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie26a154c320d8527e54b871a3b99bbbc3c2ee811","subject":"Fixing octavia-grenade job and disabling pylint warnings","status":"MERGED","created":"2023-03-25 11:09:09.000000000","updated":"2023-03-27 17:42:37.000000000","submitted":"2023-03-27 17:41:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"878583","meta_rev_id":"1f0545f0207c234c73f0a73100a7d9ac56da82b0","_number":878583,"virtual_id_number":878583,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-27 17:41:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-03-27 15:28:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-27 15:10:09.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-03-27 15:28:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-25 11:57:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-27 10:45:59.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-03-27 14:19:12.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2023-03-27 15:28:33.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"f32f703cf47d7d64f48475050b7345bbc2df28e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-25 11:09:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c90e16ec2a51226998f740628eac162a0d1ba95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-25 11:57:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/efcce90b163c4bd8a928925c2ef90e54\n\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/76e1e05204344742b2faa1d9d018afbd : FAILURE in 40m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9de93c6d6c463a9415ec06b88f71d9af65921f95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-25 12:08:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bdfd4106d74dd9d7c9fc1f330e309a385c21fcc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-25 13:02:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82a654613e064b2f9df85f6ff6f97801\n\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0e136210bb254c6bbe2a51a59a03a4f6 : SUCCESS in 47m 14s","accounts_in_message":[],"_revision_number":2},{"id":"5bbc86a9afeaa247c1d5e6c15659495df9b5ad07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-25 13:11:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b37c2322e73f4521e73f40022eb7ef5a2e79ca10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-25 16:15:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1439a5a060df4217a6303af66158d112\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5635ae46a1074cd4ba59bdbee3f192a4 : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1586ec15b1db4810a7b63669c8e5ad98 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/700aecc8b7dd43679273704016ab319d : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9b0dabacfc1495b8238234966c8c85b : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fe2777d7a1f4b7e8004c35751c740ec : SUCCESS in 10m 09s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/937d6e1f30bd4044b2b4d9eace4f766c : SUCCESS in 4m 32s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0124e454b250438f9e3d012fee8d8e6a : SUCCESS in 6m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b15ff6e9e36d424693f964a9743ce474 : SUCCESS in 4m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eb63108fae6a4b79be717070a594bfd1 : SUCCESS in 6m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e70d9ba62dc849a688fe4cb190195bb6 : SUCCESS in 6m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c99fea5fb9ea49f6a1034b42b8e8e3a6 : SUCCESS in 5m 59s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e16f6b9c4114429a90aecefd1e032a29 : SUCCESS in 57m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/300dddced10345b19b72d073b48baef9 : SUCCESS in 1h 41m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1868132127bb4504afa6d170f2d66e2d : SUCCESS in 34m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a42546ab26904f83b543478e996f8a29 : SUCCESS in 40m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ae6b0bf331c343089dbc8f43783b5bf1 : SUCCESS in 52m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d5192bd6dc2546eb90d63168cb98786a : SUCCESS in 32m 33s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cbee11f4d2a844bd9b0980fa7a8df85f : SUCCESS in 1h 46m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3e60b63d42bb450aa1d53ce4656a87bb : TIMED_OUT in 3h 02m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2ec420f912404648b9a2f166e4e9e065 : SUCCESS in 1h 44m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a9fc7bc2e0214636908016800fce6af4 : SUCCESS in 1h 49m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9f507d9ed62f4fb183153504f8b77ce3 : SUCCESS in 1h 56m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f46e87743cd48f5b1d2b4b38a15f261 : SUCCESS in 2h 00m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0890370dba4f48e88b6ec38440b0a7cedb53f4c7","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-27 10:45:59.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"95e0871fb94c199e386bab6f7337a3cf3ec6b4ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-27 11:59:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4fe58d22abf098d3f3ad1ade10a18c9838371dae","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-27 14:19:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b340d794c72722dc5072c7600a8ae7e4d043d270","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 15:08:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5d5ab746d1140f091e56eff7c720f50\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50e49d19f27148dd8091780aeed74176 : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f1868d973a34feaa0c6cff00cb90b04 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fbe77f8b1f374db8947d6acffb17870c : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea90bcdf32f94441b6a1b4aae22e7df4 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9af340523a94418398bc6a93bc0aa3a2 : SUCCESS in 10m 47s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/11f9f2acf19e466ebe64f735fcc8d13e : SUCCESS in 4m 44s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b27be9ea1c854a569c397b3f1fd9db86 : SUCCESS in 8m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d21b2085ce35453d9748ac6d3db1bde6 : SUCCESS in 4m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/08fdf39382aa4365bd4228127bb82d68 : SUCCESS in 7m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/751ab0c5b9964ee38f33a436509971f4 : SUCCESS in 7m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5402b1676562469db9df16767580e46f : SUCCESS in 7m 07s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5aef1814051142009aa02e5adc769e68 : SUCCESS in 1h 11m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/39a5a9f08387494d9a5a4655b6dbc20a : SUCCESS in 2h 19m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b63178833cfb4f5cb2fcbac684c67dd2 : SUCCESS in 41m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5e1b43e5fa1646078e76a505c612987a : SUCCESS in 50m 19s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7117bbc3d058452bb6bd650203c0e973 : SUCCESS in 1h 01m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc815809037049458071b980fb0e9006 : SUCCESS in 40m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6236a2c6848247d69264ed48fbf64e2a : SUCCESS in 2h 03m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/87f409b0976d4b1c882fd8976c6d2a02 : TIMED_OUT in 3h 04m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b1658af96be4478d9fcf1fa44d15a6e6 : SUCCESS in 2h 09m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c485a41dba264110b53467a896116f10 : SUCCESS in 2h 17m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/53f712b3c9f34070acadf2eced21884e : SUCCESS in 2h 17m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e930ee9176184af391231b017517dc79 : TIMED_OUT in 2h 35m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5b75284b2da4898ffa8dcaeb67b6ab0b8ea5ad97","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-27 15:10:09.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"406e30d5843c8f8cbba931816128b4cd39b5ee0d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-27 15:28:33.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"133f722ad3f839cf6753b00436c0fd182680821b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 15:29:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b047aaaa16054003554cf5cfcff4aac3e9638215","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 17:41:23.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4eabab610914b2397d5379400c509b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5ae84a7ab5c4764a2a8e4a643b0e2e8 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c4e0e016377449eb35bf37a3cd454c0 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf3339f3f31644ae8fd0b59f57fd6d64 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71cca5247372423581c9f0549f06838f : SUCCESS in 9m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2ab6fc528dc945ff999c0959e8b3fdc6 : SUCCESS in 4m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0534b314573143a6972cbd0285025565 : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5c2aeb3a288f4adcb3fba9800518f2c8 : SUCCESS in 1h 35m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a2214da5f1bc4894aeefa4ee30a1534c : SUCCESS in 2h 11m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ad522dd92dfc4208a05d4a701cee2e98 : SUCCESS in 41m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/22d6cd19c6be414eba343b663df0228e : SUCCESS in 46m 24s","accounts_in_message":[],"_revision_number":4},{"id":"7090e0ed084f726bb23cae2d35db77ab0b2f2612","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 17:41:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"1f0545f0207c234c73f0a73100a7d9ac56da82b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 17:42:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/560abdb208c0478c8f28cc6c3fef227f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bf3d640f8584f8c84de031dbd59dba1 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"afb4d151f73043b592243359a393fe6c4cf8f410","revisions":{"5ccc0a0e0d3e659c48837fab96fe6c06e5dfdb13":{"kind":"REWORK","_number":1,"created":"2023-03-25 11:09:09.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/83/878583/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/878583/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/878583/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/878583/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/878583/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/878583/1"}}},"commit":{"parents":[{"commit":"ba92aeb946343c045e8363a2aae78e0a17a6cb82","subject":"Fix failover when the last listener is deleted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba92aeb946343c045e8363a2aae78e0a17a6cb82"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-25 11:07:44.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-25 11:08:59.000000000","tz":60},"subject":"DNM/WIP Testing grenade job","message":"DNM/WIP Testing grenade job\n\nChange-Id: Ie26a154c320d8527e54b871a3b99bbbc3c2ee811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ccc0a0e0d3e659c48837fab96fe6c06e5dfdb13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ccc0a0e0d3e659c48837fab96fe6c06e5dfdb13"}]},"branch":"refs/heads/master"},"ce3c98602a167f4b71cab75512175a56d552a2a3":{"kind":"REWORK","_number":2,"created":"2023-03-25 12:08:05.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/83/878583/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/878583/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/878583/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/878583/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/878583/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/878583/2"}}},"commit":{"parents":[{"commit":"ba92aeb946343c045e8363a2aae78e0a17a6cb82","subject":"Fix failover when the last listener is deleted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba92aeb946343c045e8363a2aae78e0a17a6cb82"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-25 11:07:44.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-25 12:07:59.000000000","tz":60},"subject":"DNM/WIP Testing grenade job","message":"DNM/WIP Testing grenade job\n\nChange-Id: Ie26a154c320d8527e54b871a3b99bbbc3c2ee811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce3c98602a167f4b71cab75512175a56d552a2a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce3c98602a167f4b71cab75512175a56d552a2a3"}]},"branch":"refs/heads/master"},"9bd24b353d5937dbd499a27c95164e59311b289b":{"kind":"REWORK","_number":3,"created":"2023-03-25 13:11:27.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/83/878583/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/878583/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/878583/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/878583/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/878583/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/878583/3"}}},"commit":{"parents":[{"commit":"ba92aeb946343c045e8363a2aae78e0a17a6cb82","subject":"Fix failover when the last listener is deleted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba92aeb946343c045e8363a2aae78e0a17a6cb82"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-25 11:07:44.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-25 13:10:57.000000000","tz":60},"subject":"Fixing octavia-grenade job","message":"Fixing octavia-grenade job\n\ngrenade now updates from an ML2/OVN deployment. Octavia needs to source\nthe devstack/lib/octavia file from neutron, which provides the functions\nthat manage the octavia management interfaces.\n\nChange-Id: Ie26a154c320d8527e54b871a3b99bbbc3c2ee811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9bd24b353d5937dbd499a27c95164e59311b289b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9bd24b353d5937dbd499a27c95164e59311b289b"}]},"branch":"refs/heads/master"},"afb4d151f73043b592243359a393fe6c4cf8f410":{"kind":"REWORK","_number":4,"created":"2023-03-27 11:59:48.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/83/878583/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/878583/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/878583/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/878583/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/878583/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/878583/4"}}},"commit":{"parents":[{"commit":"ba92aeb946343c045e8363a2aae78e0a17a6cb82","subject":"Fix failover when the last listener is deleted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba92aeb946343c045e8363a2aae78e0a17a6cb82"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-25 11:07:44.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-27 11:58:30.000000000","tz":120},"subject":"Fixing octavia-grenade job and disabling pylint warnings","message":"Fixing octavia-grenade job and disabling pylint warnings\n\ngrenade now updates from an ML2/OVN deployment. Octavia needs to source\nthe devstack/lib/octavia file from neutron, which provides the functions\nthat manage the octavia management interfaces.\n\nThe commit also removes false-positive pylint warnings, an issue was\nreported to the pylint project:\nhttps://github.com/PyCQA/pylint/issues/8497\n\nChange-Id: Ie26a154c320d8527e54b871a3b99bbbc3c2ee811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/afb4d151f73043b592243359a393fe6c4cf8f410"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/afb4d151f73043b592243359a393fe6c4cf8f410"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
