)]}'
{"id":"openstack%2Foctavia~878658","triplet_id":"openstack%2Foctavia~stable%2Fzed~Ib1e00a0b45776cfab80b9effb2690e12fea63e78","project":"openstack/octavia","branch":"stable/zed","topic":"ipv6-bug-stable/zed","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-04-05 13:14:22.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-04-05 13:14:22.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-04-05 10:53:53.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-04-04 12:29:25.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Ib1e00a0b45776cfab80b9effb2690e12fea63e78","subject":"Fix octavia to accept [ipv6]:port","status":"MERGED","created":"2023-03-28 10:15:56.000000000","updated":"2023-04-05 13:15:45.000000000","submitted":"2023-04-05 13:14:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878658-ipv6-bug-stable/zed","cherry_pick_of_change":877959,"cherry_pick_of_patch_set":3,"meta_rev_id":"8375ab5b5ab64b14407c3215b5abbd4594239542","_number":878658,"virtual_id_number":878658,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-05 13:14:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2023-04-05 10:53:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-04 12:29:25.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2023-04-05 10:53:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-28 10:25:05.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-03-28 12:51:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-31 07:10:57.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2023-04-04 12:02:27.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-04-04 12:02:27.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"553067cc63394a1431b9a7a721600ef737e81519","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-03-28 10:15:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"aaeee2823b93c841d40c81ca5133b4d89adc9abf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-28 10:25:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"00631dc21724448672cffa812cf8f387f4e38084","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 12:51:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c34779ce4b9a41e79709c71e1d67d932\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f745d7c6cbe5422998101dc1b32f5eb2 : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4311b40b5914fd6a09ae102f5c87163 : SUCCESS in 7m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9d17c8f1fa4432c80c3c82bc0aa2c57 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ed6436a47b34e7da515fa8334d943be : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/badc495450174b0b87a228811856492a : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62ec21036d164734a336706238a1be3e : SUCCESS in 11m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/930d7c4394ec4373a5aa61afb5996e1a : SUCCESS in 5m 28s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/343174f4ccdc444b88ade2a1d983a8b2 : SUCCESS in 5m 12s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/6b34ef521bda41d491d604b28c9a3864 : SUCCESS in 7m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/09dc40d8ef054b99bb80a04371a2f9c4 : SUCCESS in 4m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/731c85cfbad4484e85d82de8d0ac2175 : SUCCESS in 7m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5c45a72a9d79476e9b85b9067afcde5c : SUCCESS in 6m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/787cf8e72d6849c59bbc6aea9e214e78 : SUCCESS in 7m 56s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/caa3a625295e4086acfa5ad58826aa49 : SUCCESS in 1h 23m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/630052f3666645b1ad40424d3591eacf : SUCCESS in 2h 00m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/41dc95c51ec5400e97352ab71ada867f : SUCCESS in 35m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3b328fe70596499297d5b95d2be75095 : SUCCESS in 35m 00s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/19cf93ae81a841dd968ff2126b2f24db : SUCCESS in 1h 04m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/070db738f1454f46aad53bd1273db4e4 : SUCCESS in 32m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ad11b260c21e4165aefe7cd87ea34712 : SUCCESS in 2h 08m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e6c288beb1f44ca89df3dbc79e6411ea : SUCCESS in 2h 33m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d47467e5c68f48639d4137d4c09f6075 : SUCCESS in 2h 23m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8638d6a129ed476a8efa06c7be6f3d8e : SUCCESS in 2h 26m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2abdfc6e31b349fcb2c80fccc21cd876 : SUCCESS in 1h 45m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/daffbc8363804ea099daf397f234fc18 : SUCCESS in 2h 01m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"182b00d90006c49c783f5bd7ba88c74dea2b9760","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-31 07:10:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83a904145d827402eefca0340e9eabec7b944775","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-04-04 12:05:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"54da12821dd411fb25eb211768395fa351d04223","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-04 12:29:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"70465329a70e582bbbbc93e2899cf9d806fad6c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 14:42:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7498b4341ce141de9495ed2900a451b8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b570ecd3ec5f4584832c3a975749738c : SUCCESS in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c662ec5bbc104a6d8cbc27dbad76e346 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4ec8775da3745509c6257c47d92a977 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f89fc410dff4f8fb9f240eb7f5e6ca7 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09dc6a1a8c6b48c2965bb7aba05ec62e : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0369b7dd8d494a60a6bf1c0992d7c74e : SUCCESS in 11m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a164aa2b9d1d4a9da5c214baf861f10f : SUCCESS in 5m 20s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/36dac3b6c9ff4acaa55cc268f47dc3b2 : SUCCESS in 4m 47s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a32e7209a3a04cd6ae373911a7cb7f38 : SUCCESS in 7m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b1625f917d244f07b99dc3474d7d857b : SUCCESS in 4m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bd1cfa2b8d9347b48290c346c18c1dc2 : SUCCESS in 7m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a5a84a76f5cb44fcac988bb644e9f8ef : SUCCESS in 7m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a2feee47409a407aa1bb1fe6f5364341 : SUCCESS in 7m 01s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1121abc87b204f51a4e349d9ab0e7dca : SUCCESS in 1h 50m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5373b980b95d46de8e9cbe6e8afb9212 : SUCCESS in 1h 43m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0bfbd634682544cf936e40062c02dd2b : SUCCESS in 33m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/44336a211b854abcabd88a8d33c1375f : SUCCESS in 37m 09s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d86c44e2f45a44fe86b33d2dcf756c24 : SUCCESS in 59m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a163fc7537b841da86d006e998d479cb : SUCCESS in 29m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/70f2edb32520475b831c6807f36bbcfd : SUCCESS in 2h 02m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/693b7aef463a42c7b610bebddbefd3d4 : SUCCESS in 2h 07m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c64fb65d12474a449e6f999436a6d061 : SUCCESS in 1h 44m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0d8e818f18524eb8b95a049814f26147 : SUCCESS in 2h 27m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5cd1f28e7efd4dd3959cd740a1518d7b : SUCCESS in 1h 49m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/18cda52955c446f382a8d538b9666248 : TIMED_OUT in 2h 34m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b34eb5163499f4eed48e9934a371c477e0c89d3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-05 10:53:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"519ea24c1550e8931f505fbe5b2dac28d229538c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 10:54:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0194b2caaccfe37146fa5d0b859256548d5c10c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 13:14:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7525b39f9b8c42a7a6f5ea074ced7ab6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d4e6741ac0e4bd39347d3eeebae8f8d : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6cbfb6863be4d4da0463efc671b4f1b : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a6dab3803f44675979a3a455564ff8c : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc38984fdadf474ca161ed793a0925dc : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbe961d6edf9404580fe9f07502f8447 : SUCCESS in 4m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/006be6b557ef445a82e86252a0bd907a : SUCCESS in 5m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f12e673a222e4a28ac34fcd10bf78cc1 : SUCCESS in 7m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e17e281caa774d6f81f4a0de2404741f : SUCCESS in 1h 45m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0eabc3c85ab8407c894e25c1b1611251 : SUCCESS in 2h 15m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c506d8bbd9ec47cca6d21a4477fba0f5 : SUCCESS in 37m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cc1fe57beacb47bb8cf4e33ec7decb00 : SUCCESS in 50m 20s","accounts_in_message":[],"_revision_number":2},{"id":"515944c8c28d9e8eb69dda99e064f9e144dff6ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 13:14:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8375ab5b5ab64b14407c3215b5abbd4594239542","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 13:15:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed55a68e9dbc4e3aa629ea00377c19e0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5921178fc994550990d944ee0355892 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cafc732a1d0846b48da7b39c8cc8eadb : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"332d7deeba740b9fd8a67065f5818010a5d2990b","revisions":{"c11ba4a0f731ef1b7267d8b3d44158c3055f0a57":{"kind":"REWORK","_number":1,"created":"2023-03-28 10:15:56.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/58/878658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/58/878658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/878658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/878658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/878658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/58/878658/1"}}},"commit":{"parents":[{"commit":"1584ff1e66889d828fc80850399061b2f72cb526","subject":"Merge \"Fix listener creation with fully-populated API\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1584ff1e66889d828fc80850399061b2f72cb526"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-03-20 14:09:52.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-03-28 10:15:56.000000000","tz":0},"subject":"Fix octavia to accept [ipv6]:port","message":"Fix octavia to accept [ipv6]:port\n\nA host identified by an Internet Protocol literal\naddress, version 6 [RFC3513] or later, is\ndistinguished by enclosing the IP literal\nwithin square brackets (\"[\" and \"]\").\n\nThis patch fixes octavia to be able parse and\nuse this format.\n\nStory 2010655\nTask 47687\n\nChange-Id: Ib1e00a0b45776cfab80b9effb2690e12fea63e78\n(cherry picked from commit a460698cd933112529d1eae27d483da6b3035eff)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c11ba4a0f731ef1b7267d8b3d44158c3055f0a57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c11ba4a0f731ef1b7267d8b3d44158c3055f0a57"}]},"branch":"refs/heads/stable/zed"},"332d7deeba740b9fd8a67065f5818010a5d2990b":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2023-04-04 12:05:24.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/58/878658/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/58/878658/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/878658/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/878658/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/58/878658/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/58/878658/2"}}},"commit":{"parents":[{"commit":"b2437857c84f46ba8d1e3ae95f09e6e53ed01b66","subject":"Fix ORM caching for with_for_update calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b2437857c84f46ba8d1e3ae95f09e6e53ed01b66"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-03-20 14:09:52.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-04-04 12:05:06.000000000","tz":120},"subject":"Fix octavia to accept [ipv6]:port","message":"Fix octavia to accept [ipv6]:port\n\nA host identified by an Internet Protocol literal\naddress, version 6 [RFC3513] or later, is\ndistinguished by enclosing the IP literal\nwithin square brackets (\"[\" and \"]\").\n\nThis patch fixes octavia to be able parse and\nuse this format.\n\nStory 2010655\nTask 47687\n\nChange-Id: Ib1e00a0b45776cfab80b9effb2690e12fea63e78\n(cherry picked from commit 4bbe88ee78ad05f76dab2068d15de9b50d3811aa)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/332d7deeba740b9fd8a67065f5818010a5d2990b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/332d7deeba740b9fd8a67065f5818010a5d2990b"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
