)]}'
{"id":"openstack%2Foctavia~878838","triplet_id":"openstack%2Foctavia~stable%2Fyoga~Ibd4da09079e83789d6cfe3658fcf1f266f5cf8b4","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2023-03-31 09:09:31.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-03-31 09:09:31.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-03-30 13:01:23.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Ibd4da09079e83789d6cfe3658fcf1f266f5cf8b4","subject":"Fix ORM caching for with_for_update calls","status":"MERGED","created":"2023-03-30 07:03:22.000000000","updated":"2023-03-31 09:11:08.000000000","submitted":"2023-03-31 09:09:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878838","cherry_pick_of_change":877414,"cherry_pick_of_patch_set":2,"meta_rev_id":"035a4234093bafc348496e4705fb56c73223bce8","_number":878838,"virtual_id_number":878838,"owner":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-31 09:09:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-03-31 06:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-30 13:01:23.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2023-03-31 06:41:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-30 07:03:22.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2023-03-30 10:03:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-30 11:57:28.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-03-30 11:57:28.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-03-31 06:41:24.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"74c2e5abac1928f430d6f6136b303270ac026ce5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-03-30 07:03:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* octavia/db/repositories.py","accounts_in_message":[],"_revision_number":1},{"id":"29114e760b05911de0eecaa18ff9aa02a99da264","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-03-30 07:38:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5cbf92a5680d0cb4bf18ba63f5943a247a74ef17","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-03-30 07:39:28.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"1a2e61494b2c71c7f1b83d913abbc22154d4e507","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 10:03:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cc072f27e7c48ed97bb6bdbe92635ae\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b7ef88d64764938b315ba8546404aa5 : SUCCESS in 12m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45c60b73f013417392c0b169395205be : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e12d5ebd99534dceaa2886356a7d8ace : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77a74f27eeab436f9b40f1391baa6eaf : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc8b98bb1aef446cbeb59e1a0595a80a : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b41f8f9b51e343a9afe62494e7126fbf : SUCCESS in 13m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fde661f58514d61bc4ad5b9499bd908 : SUCCESS in 6m 37s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/eb4fc259ce2c49e98b60909e0a86e01e : SUCCESS in 5m 33s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/25d794677d0d45f5823259b21a64fbbc : SUCCESS in 5m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7d665ed9110e4e4594031d948983327e : SUCCESS in 4m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eeda97add6f142098d18f1aaafadb189 : SUCCESS in 7m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/53125b5748b04094b4c317798b639e57 : SUCCESS in 1h 31m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8fde4946a381403da85f08239cc51aa3 : SUCCESS in 1h 31m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e21f25e4fb9b4a52a824f014eb24ad94 : SUCCESS in 32m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/00c5e57ab70b4175b4e10f22714fcbeb : SUCCESS in 35m 35s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/de22c3a33b2149a3b24ae3a012d10961 : SUCCESS in 1h 06m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd6d099aa8934e90a868f031c86feeff : SUCCESS in 31m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ced9e311cb9046abba8bd2298b596001 : SUCCESS in 1h 53m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/de39dcd14fb2494da2fa9c9888352b77 : SUCCESS in 2h 18m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8468b6594a814fe982b78205e6423418 : SUCCESS in 1h 38m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/25fd1ddbc5f446ed8d1c3041045cc303 : SUCCESS in 2h 22m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/22a21355ac944219ba53ea03d233b5a1 : SUCCESS in 1h 48m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/e0451bd18d0f439c9d14d941da83e138 : FAILURE in 42m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d544e7598df9433f5b5339e4faa7209fbc269e97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-03-30 13:00:10.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ed21d0c2f895b6ace7958d177557dfcc3fa67f98","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-30 13:01:23.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c637d826187f857fa029dd4f32e3ab430b3d20f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 15:37:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c5006694f1844868adbf85d0e3556f4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f9dddff4a8e467fa5f87398096e09e0 : SUCCESS in 12m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19f2938b7fb946649b2507e61fcae0e6 : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97dcba4a9bf54f118530fd60e1f4a09a : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/993f96e295224f3da4ed3dd423b7e50e : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5753de313c8b4d64bfc66a2fe37bb01f : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e8e15ebfc7142919a04b1d4c732232d : SUCCESS in 14m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8565b562d6347fabcace2102d247a1e : SUCCESS in 6m 17s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/75a83bb94e224088a9ac7fd33b78b9f4 : SUCCESS in 4m 40s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/6233cebdc8b140b1922bec9ebe3ab639 : SUCCESS in 6m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e4c40f2746904737828eb231c52880de : SUCCESS in 5m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2ab4c534adaa4e439c458cfaa2916a18 : SUCCESS in 8m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/91a93ed2932147c8895bfa169cbfbbd7 : SUCCESS in 1h 19m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e779d005bb3542728f116aba421709d3 : SUCCESS in 1h 44m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ea131484359e4b6eb0558ddf1fc5780e : SUCCESS in 52m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/50cca4df52ee4f40822f12a79a282e79 : SUCCESS in 59m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/31a80bd603704e91b42145b96939b532 : SUCCESS in 53m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95109a06bec74212967af7f1acfe0f85 : SUCCESS in 29m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/282e1c076fe6473496f9e8a5b000eb1a : SUCCESS in 2h 24m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/27b3594cb6854a82bca7aacab04564ec : SUCCESS in 2h 04m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2c2cecc21a5644e1a90675f3baab1fa9 : TIMED_OUT in 2h 34m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/60adf162d4b1473c9f6b194b888f08b0 : TIMED_OUT in 2h 32m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/f02607ce60df46ada0fdf098c065fbb1 : TIMED_OUT in 2h 33m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/60c24797b5e646529741fd9be4048c7a : FAILURE in 38m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3561ca9db67a3bce0c52882cffefebed638937cf","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-31 06:41:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e1dafa617b6e4d7377e103ecc3d3299631606e21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 06:42:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b8aa3a0c5dfaa675ee27ac7027345f047eea3ab8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 09:09:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83dc0ced1d334f0091f67b6df307a669\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/acb2514c1034469ab546bc44129fd2bd : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e82d20c444554aa08cf70dce5d080d02 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b884a98e40f8429c844958b25739d6c6 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c85128ab8ff7450ab6c7bee52541f9a1 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b46491e05ba74c1eb3259cb07bb05665 : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b6c87cbadac44e3a4ed3d1dd2e45d18 : SUCCESS in 5m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bcfa059822234b67a147c3366e27191a : SUCCESS in 4m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/07f314e8b0924e57a4f07c5b7375689f : SUCCESS in 8m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d2e5bba928f347aca9b5ed31842e0466 : SUCCESS in 1h 11m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/874b650734034c598c24cf072555bb41 : SUCCESS in 1h 52m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2257aec3272746488d3761a6c0efc1f4 : SUCCESS in 36m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fc1886fba63b492b8e165836de721dc9 : SUCCESS in 39m 35s","accounts_in_message":[],"_revision_number":3},{"id":"668dbf1efaedc3194922fa0795b67ac2da7ca47c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 09:09:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"035a4234093bafc348496e4705fb56c73223bce8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 09:11:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3480238d8064929aef7b74d8bd6294e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0af67891e4fb4142ba5f0aa2dfd40af9 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66296dc9cd274eac9107eb8c25d57a79 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"df2a5275eb5f9eb043f1ff9418353ec5e1fcaa96","revisions":{"f294a359b1b531602117151937c15201c7d756f7":{"kind":"REWORK","_number":1,"created":"2023-03-30 07:03:22.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/38/878838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/878838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/878838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/878838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/878838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/878838/1"}}},"commit":{"parents":[{"commit":"12b452f6e1391b3ad5c2e44bdd983632350965af","subject":"Merge \"Fix ignored [nova] service_name\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12b452f6e1391b3ad5c2e44bdd983632350965af"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-14 17:09:42.000000000","tz":-240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-03-30 07:03:22.000000000","tz":0},"subject":"Fix ORM caching for with_for_update calls","message":"Fix ORM caching for with_for_update calls\n\nThe SQLAlchemy recommends to use populate_existing() when using\nwith_for_update() [0], it fixed issues with the caching of the objects.\n\nThis patch precisely fixes a bug when locking of a loadbalancer in the\nbatch member update API call, the load balancer might not have been\nlocked correctly and race conditions could have occurred (processing\nsimultaneously 2 requests in the workers for the same load balancer).\n\n[0] https://docs.sqlalchemy.org/en/14/orm/query.html#sqlalchemy.orm.Query.with_for_update\n\nStory 2010646\nTask 47642\n\nCo-Authored-By: Gaudenz Steinlin \u003cgaudenz.steinlin@cloudscale.ch\u003e\n\nChange-Id: Ibd4da09079e83789d6cfe3658fcf1f266f5cf8b4\n(cherry picked from commit 2e4ee9c65e0fba660b0316f1f4203810bb8a67c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f294a359b1b531602117151937c15201c7d756f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f294a359b1b531602117151937c15201c7d756f7"}]},"branch":"refs/heads/stable/yoga"},"ffe2eb4a22a6701ba26e216d31505040e911506e":{"kind":"REWORK","_number":2,"created":"2023-03-30 07:38:54.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/38/878838/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/878838/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/878838/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/878838/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/878838/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/878838/2"}}},"commit":{"parents":[{"commit":"12b452f6e1391b3ad5c2e44bdd983632350965af","subject":"Merge \"Fix ignored [nova] service_name\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12b452f6e1391b3ad5c2e44bdd983632350965af"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-14 17:09:42.000000000","tz":-240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-03-30 07:38:39.000000000","tz":240},"subject":"Fix ORM caching for with_for_update calls","message":"Fix ORM caching for with_for_update calls\n\nThe SQLAlchemy recommends to use populate_existing() when using\nwith_for_update() [0], it fixed issues with the caching of the objects.\n\nThis patch precisely fixes a bug when locking of a loadbalancer in the\nbatch member update API call, the load balancer might not have been\nlocked correctly and race conditions could have occurred (processing\nsimultaneously 2 requests in the workers for the same load balancer).\n\n[0] https://docs.sqlalchemy.org/en/14/orm/query.html#sqlalchemy.orm.Query.with_for_update\n\nStory 2010646\nTask 47642\n\nCo-Authored-By: Gaudenz Steinlin \u003cgaudenz.steinlin@cloudscale.ch\u003e\n\nChange-Id: Ibd4da09079e83789d6cfe3658fcf1f266f5cf8b4\n(cherry picked from commit 2e4ee9c65e0fba660b0316f1f4203810bb8a67c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ffe2eb4a22a6701ba26e216d31505040e911506e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ffe2eb4a22a6701ba26e216d31505040e911506e"}]},"branch":"refs/heads/stable/yoga"},"df2a5275eb5f9eb043f1ff9418353ec5e1fcaa96":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-03-30 13:00:10.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/38/878838/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/878838/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/878838/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/878838/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/878838/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/878838/3"}}},"commit":{"parents":[{"commit":"12b452f6e1391b3ad5c2e44bdd983632350965af","subject":"Merge \"Fix ignored [nova] service_name\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/12b452f6e1391b3ad5c2e44bdd983632350965af"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-14 17:09:42.000000000","tz":-240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-03-30 13:00:10.000000000","tz":0},"subject":"Fix ORM caching for with_for_update calls","message":"Fix ORM caching for with_for_update calls\n\nThe SQLAlchemy recommends to use populate_existing() when using\nwith_for_update() [0], it fixed issues with the caching of the objects.\n\nThis patch precisely fixes a bug when locking of a loadbalancer in the\nbatch member update API call, the load balancer might not have been\nlocked correctly and race conditions could have occurred (processing\nsimultaneously 2 requests in the workers for the same load balancer).\n\n[0] https://docs.sqlalchemy.org/en/14/orm/query.html#sqlalchemy.orm.Query.with_for_update\n\nStory 2010646\nTask 47642\n\nCo-Authored-By: Gaudenz Steinlin \u003cgaudenz.steinlin@cloudscale.ch\u003e\n\nMerge conflict:\n    octavia/db/repositories.py\n\nChange-Id: Ibd4da09079e83789d6cfe3658fcf1f266f5cf8b4\n(cherry picked from commit 2e4ee9c65e0fba660b0316f1f4203810bb8a67c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df2a5275eb5f9eb043f1ff9418353ec5e1fcaa96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/df2a5275eb5f9eb043f1ff9418353ec5e1fcaa96"}]},"branch":"refs/heads/stable/yoga","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
