)]}'
{"id":"openstack%2Foctavia~878839","triplet_id":"openstack%2Foctavia~stable%2Fxena~Ibd4da09079e83789d6cfe3658fcf1f266f5cf8b4","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2023-03-31 09:09:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibd4da09079e83789d6cfe3658fcf1f266f5cf8b4","subject":"Fix ORM caching for with_for_update calls","status":"MERGED","created":"2023-03-30 07:53:26.000000000","updated":"2023-03-31 09:11:09.000000000","submitted":"2023-03-31 09:09:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878839","cherry_pick_of_change":878838,"cherry_pick_of_patch_set":2,"meta_rev_id":"3b1434f01b51c53d79614fa0d531848eb10a2050","_number":878839,"virtual_id_number":878839,"owner":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-31 09:09:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-03-31 06:43:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-30 13:04:20.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2023-03-31 06:43:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-30 07:53:26.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2023-03-30 10:40:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-30 13:04:20.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-03-31 06:43:23.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"c8e889f7a76ddcccffa45753de617e5d7ecd8b03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-03-30 07:53:26.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"8cabb31d309b85bb5407af82c549022413740196","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 10:40:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9874f5aa5fb84df4a3c3fac548e98458\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43383b70382d44ec8a31d163634da9de : SUCCESS in 12m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4424682fce714c24965ee4d0aab62206 : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea123f85f2c64666ada35af7e574de25 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30a39d83d6044241826b92cd943c4417 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dafadb99db514fa593567806a86c61dd : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07700ec1c7a945aa8840658921a06c96 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fa2ed6c1eaf4ea9a9a7ce6d039d5949 : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c92060fab9ef4075915126b10abc54b1 : SUCCESS in 5m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a3d1c2cec31f4a95b228a8b1a3ed23dd : SUCCESS in 6m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/845c4ae66e5845188df8b7ab4943243f : SUCCESS in 9m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/97d9ae789783400597b4746c1b9e7da4 : SUCCESS in 5m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a1f8dcb2b8e94d119f9238fbd27d1db5 : SUCCESS in 8m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6921cf5c572b4392a1babdba79e87d54 : SUCCESS in 1h 23m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a9b0c4f4e2eb4845981ac3fa3b6d24ee : SUCCESS in 2h 09m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ef2ce2cb1df44e1da35f46820866e09f : SUCCESS in 36m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2a8b3eb367774dde9a1246866b4eba92 : SUCCESS in 50m 04s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d1a5d82d41d24ca09bbcefadc10eff99 : FAILURE in 9m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8d3ac8d008b74c19923a029419348b37 : SUCCESS in 38m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/20bbbc1ecbc84484af0270533b7127d5 : TIMED_OUT in 2h 34m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2ffbb4facb7b429ca6412ca34d844128 : SUCCESS in 2h 32m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a60d75d003cb4378985cb630b0c95466 : SUCCESS in 2h 10m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8fcbc0cb9f87492fa70f7315a1b3f003 : SUCCESS in 2h 09m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/91927bcc09bc4da4830b79429f9544f4 : SUCCESS in 2h 01m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5c0609e0d6bf494f9ac4a3772c4fff8a : NODE_FAILURE Node request 300-0020865347 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7e5f3d7901bfa1a8a7ea3a1cc7dbdd6429f2a4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-03-30 13:00:33.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2d2d6b98bc7ecb2764ae9958585652c52a2fdf2d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-30 13:04:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d540bf0105f1abf38bf2219b98a5d6351289a828","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 15:34:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44b42c600c8a42c088ae7f487bab9b65\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/341a51a836b04746b7b7adc0ffbeba13 : SUCCESS in 8m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd895ddcf27d4d4fb3d5d413ff4fcf3e : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c45adad8e7a04fdbacc603f3f63a7941 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a763cfd05984319bd76df8e8b81b172 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80c5994e50f24bd99fbcabcbffa53d2c : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2adec2e08a784607ab5720a1fe372e94 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c3f5f9b7bbf417dab31a7fce53e55fc : SUCCESS in 13m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cadc7cd4777f45c49e8bb2e818336b55 : SUCCESS in 5m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5c6342258b7141e495ae2a8e44fa605f : SUCCESS in 5m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c22ab5becb784a3b98c74ea9ab1305c5 : SUCCESS in 12m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1af8c65b8c58480f8cc45debece8da06 : SUCCESS in 5m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3c2218d552754aee8e67038e8b2d5dd0 : SUCCESS in 10m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4488df13a4564a03a0725e154c382040 : SUCCESS in 1h 22m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/99bf57e6631d42e19969d15d63c03a85 : SUCCESS in 2h 03m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3c78705761ad4e46bc7f0373feb9787c : SUCCESS in 40m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/99a0d4f098e3440e84323f0b1fd6777c : SUCCESS in 48m 34s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1a19ea978a2e47419f3c662c24110400 : FAILURE in 11m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5c90918b70e94098832760267c2f2db1 : SUCCESS in 35m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/96370991ae0943d59bfc7b33f47783c8 : SUCCESS in 2h 05m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1fc00f8e148449198d75dc7d3697096a : SUCCESS in 2h 14m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4e9720e2a6a14f9da41f1e18e33427c1 : SUCCESS in 2h 25m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4c698c2272734948984e6d2edd964a58 : SUCCESS in 2h 26m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/caaea719d11242aa949439a1086e12fa : SUCCESS in 2h 31m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/cc17d38c00904d1caa0e12de09e73a87 : NODE_FAILURE Node request 300-0020867645 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"29f11545f346f5c02cafbe24c20aef7e89a21b95","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-31 06:43:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"90e565c14190a35e02f037915943e98b3697c83d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 06:44:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"866486d85be9d9641c22ca92e4df45cf8b73235b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 09:09:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cef74546f6046618eefc76ad3ac530b\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be3f638ac5204b31a4e8b80c284d03d8 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/555e191f8a404891823a6e857bce92c9 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5467a2b6dfe403788094949767e51d7 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce155b39bb0e4d17974783aaa2dac547 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4151db9366a409f885b8e1629d29ab5 : SUCCESS in 11m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3419020ddc6451285c39e28e000b037 : SUCCESS in 4m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f20e3d62797f4f6bb8cb1577777329fc : SUCCESS in 5m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0d662af421f14c978fb7ece30c0aec2f : SUCCESS in 6m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/09fd1151973c445192e887d4c73c85e3 : SUCCESS in 1h 25m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d2807b3bec264d6a84608e83e9a81c21 : SUCCESS in 1h 52m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d2dae794f7b4492e8aa17c80306bfc90 : SUCCESS in 31m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/86800457971e4095b2f26c501a3c8e30 : SUCCESS in 38m 48s","accounts_in_message":[],"_revision_number":2},{"id":"b9f347328652c7f3b9672fbb1d714a45816d1535","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 09:09:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3b1434f01b51c53d79614fa0d531848eb10a2050","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 09:11:09.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/306e62faa2614ca2b2b11c023be89163\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05a52706ea1b4fd991ce73c546eb4ad2 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce214c1d086741e798da07bb122c7258 : FAILURE in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"499bbaef3d30fa70cc3a04634c14db14cd28a642","revisions":{"c0cf6444baf071353d31ccf4614f8e57006346ad":{"kind":"REWORK","_number":1,"created":"2023-03-30 07:53:26.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/39/878839/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/39/878839/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/39/878839/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/39/878839/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/39/878839/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/39/878839/1"}}},"commit":{"parents":[{"commit":"9d4a38a53f8d5f168e20e9fb367175b408f70a30","subject":"Merge \"Fix ignored [nova] service_name\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d4a38a53f8d5f168e20e9fb367175b408f70a30"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-14 17:09:42.000000000","tz":-240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-03-30 07:53:26.000000000","tz":0},"subject":"Fix ORM caching for with_for_update calls","message":"Fix ORM caching for with_for_update calls\n\nThe SQLAlchemy recommends to use populate_existing() when using\nwith_for_update() [0], it fixed issues with the caching of the objects.\n\nThis patch precisely fixes a bug when locking of a loadbalancer in the\nbatch member update API call, the load balancer might not have been\nlocked correctly and race conditions could have occurred (processing\nsimultaneously 2 requests in the workers for the same load balancer).\n\n[0] https://docs.sqlalchemy.org/en/14/orm/query.html#sqlalchemy.orm.Query.with_for_update\n\nStory 2010646\nTask 47642\n\nCo-Authored-By: Gaudenz Steinlin \u003cgaudenz.steinlin@cloudscale.ch\u003e\n\nChange-Id: Ibd4da09079e83789d6cfe3658fcf1f266f5cf8b4\n(cherry picked from commit 2e4ee9c65e0fba660b0316f1f4203810bb8a67c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c0cf6444baf071353d31ccf4614f8e57006346ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c0cf6444baf071353d31ccf4614f8e57006346ad"}]},"branch":"refs/heads/stable/xena"},"499bbaef3d30fa70cc3a04634c14db14cd28a642":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-03-30 13:00:33.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/39/878839/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/39/878839/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/39/878839/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/39/878839/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/39/878839/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/39/878839/2"}}},"commit":{"parents":[{"commit":"9d4a38a53f8d5f168e20e9fb367175b408f70a30","subject":"Merge \"Fix ignored [nova] service_name\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9d4a38a53f8d5f168e20e9fb367175b408f70a30"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-03-14 17:09:42.000000000","tz":-240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2023-03-30 13:00:33.000000000","tz":0},"subject":"Fix ORM caching for with_for_update calls","message":"Fix ORM caching for with_for_update calls\n\nThe SQLAlchemy recommends to use populate_existing() when using\nwith_for_update() [0], it fixed issues with the caching of the objects.\n\nThis patch precisely fixes a bug when locking of a loadbalancer in the\nbatch member update API call, the load balancer might not have been\nlocked correctly and race conditions could have occurred (processing\nsimultaneously 2 requests in the workers for the same load balancer).\n\n[0] https://docs.sqlalchemy.org/en/14/orm/query.html#sqlalchemy.orm.Query.with_for_update\n\nStory 2010646\nTask 47642\n\nCo-Authored-By: Gaudenz Steinlin \u003cgaudenz.steinlin@cloudscale.ch\u003e\n\nMerge conflict:\n    octavia/db/repositories.py\n\nChange-Id: Ibd4da09079e83789d6cfe3658fcf1f266f5cf8b4\n(cherry picked from commit 2e4ee9c65e0fba660b0316f1f4203810bb8a67c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/499bbaef3d30fa70cc3a04634c14db14cd28a642"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/499bbaef3d30fa70cc3a04634c14db14cd28a642"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
