)]}'
{"id":"openstack%2Foctavia~878928","triplet_id":"openstack%2Foctavia~stable%2F2023.1~Ia8c35071f54e6988d9b2ff682a529bd477b93b74","project":"openstack/octavia","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-04-05 07:57:39.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-04-13 20:18:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia8c35071f54e6988d9b2ff682a529bd477b93b74","subject":"Add octavia-grenade-slurp CI job","status":"MERGED","created":"2023-03-29 16:25:09.000000000","updated":"2023-04-13 20:20:40.000000000","submitted":"2023-04-13 20:18:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878928","meta_rev_id":"e6b229a2162b7065013e5ebc86551f31c3f45bdb","_number":878928,"virtual_id_number":878928,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-13 20:18:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-04-06 10:18:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2023-04-13 18:10:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2023-04-13 18:10:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2023-04-05 15:58:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-29 19:19:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-06 10:18:34.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-04-13 18:10:38.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"58c780249e80522036a258db6f7843fc9a6a46ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-29 16:25:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"035610ce05479de06b67b01ba98bf5c8b692bb3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 19:19:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f7c0c8ce94a48d1903eeff6eb1cd796\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2419abbea38849598a225cbfe96e1b0e : SUCCESS in 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae9ae024ecee4969a645c8a8ba741af7 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a8c199a55494cee936ad33a318ed119 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26b1230bd78f47d48017664bd787e2df : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f8bdfeabf784ecf8b7c278a990620a9 : SUCCESS in 12m 01s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/52f6565166da426ba716a3687fe4cf90 : SUCCESS in 5m 55s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c582f50bb3004ba29e100f53aa678c77 : SUCCESS in 7m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f9cf8c45fe744e6eb516fb3660aae884 : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fd59ad8c1d9140b6bc8823eb6b8f3453 : SUCCESS in 7m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1d62a0dc165b4b329e31964c2b00eb17 : SUCCESS in 8m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b15f07e76e02403f8744731ff05d05c7 : SUCCESS in 8m 13s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/af42ec30fa494243b65b49fb275735ce : SUCCESS in 1h 30m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19a054fc909a467594c341bb73a1cbea : SUCCESS in 1h 49m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e226115bdcce497287d26fc80f8e723f : FAILURE in 27m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/90575b59bc2344b384ddab1bc13a96b3 : SUCCESS in 52m 55s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/c4ee0a1e29e148daa8addfd3187b7a63 : FAILURE in 23m 51s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5ebf1467e5ba4ac89a5669dcba901084 : SUCCESS in 57m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f8f41cc39a2c4a25a60b9b56ee497544 : SUCCESS in 33m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5295344979e24a6dae25a37cc0dbc5c8 : TIMED_OUT in 2h 33m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/61103387da9442349147c5f82443f263 : SUCCESS in 2h 24m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/689c3e9c496943d188d503875d294069 : TIMED_OUT in 2h 33m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a8fe4007d5dc438aa8b3c8fd2f16fb18 : SUCCESS in 1h 53m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/427a1b26f0524c32b82a0807ee9a87dc : SUCCESS in 2h 32m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/662fa143c69447d6a4af0d63abae30b8 : SUCCESS in 2h 32m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be497aaba4a9b05ded8aaedff4601315d2ce986c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-30 12:07:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e6918c13c956f4bbb2f402b0a346aabe54bb0791","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 14:55:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/62a9f5639c924a2e974cd5609bf7f64c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccfa8aae7eb040768b03aaa0ea29aed0 : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2012b4da088843b4ab3b1e604afe040b : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84d68d0eeb15486f9ecd94369ab9416a : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4b5cfcebb7e49cbbbc93fb99e48c366 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5069e705c87475da4c9f8cc82c90edf : SUCCESS in 10m 39s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/eae766ee877a471d91e43e4b0e3b23df : SUCCESS in 6m 27s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8a66abfa8db5457d8ae9464f285da430 : SUCCESS in 7m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/375250e7fc1b4a4da526256b723ea6b8 : SUCCESS in 5m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/61d72bee80ca4c4f95de3d6a9e33a9a0 : SUCCESS in 7m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5d8d7242c338418790667e0713dce89d : SUCCESS in 7m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/878558d401d34da8a551edc85134ea9e : SUCCESS in 8m 31s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f895ef5df8514352bc083fb151f6597a : SUCCESS in 1h 28m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f6e7c905102d4f7189a5f6ad2c2e68cb : SUCCESS in 2h 25m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/488fec39e1e645f38e01eb35defb1d55 : SUCCESS in 39m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/70c44425f3d043bdb34434764e1a67c9 : SUCCESS in 46m 26s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/1dc703091dc04c53a048305ac6c0985f : FAILURE in 25m 39s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/557cf62451864cec9588c3d6d0738704 : SUCCESS in 1h 11m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/12f55698c68e41a085756b632d37e664 : SUCCESS in 34m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7744ca8a0945431692f27c8f6bd4ce8b : TIMED_OUT in 2h 34m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/35686fcd365348d2aba009cce9c631a7 : SUCCESS in 2h 43m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/107440e42bb24c14b1626cfe163a8a41 : SUCCESS in 2h 16m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4b851e0492cb486dbf6a9f354927a212 : SUCCESS in 2h 10m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0185f354a2bb4bd29195b7447fe80ea3 : SUCCESS in 2h 30m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f69f29c2c805463ebc38990b54d0acb4 : SUCCESS in 2h 17m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ee17493cbe3f338fb738de6c92b2c53446d985d7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-30 16:34:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"19b76bd726274b9e5d47ce16613160da7b906567","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 19:14:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d0b7bfb3544426dadde6d6b633f6145\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d843bf5b884146969744b26e34308ef2 : SUCCESS in 11m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f19db0e185249a2920873b54dcafd3b : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f3094d0640a47a28cb9930966c74329 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c27f4713d58445c59c7da3bee9ccdb00 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/208b209315ee405cab96a1427a40d1a1 : SUCCESS in 13m 11s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/be3efe214170404b9b133a38fb41a7a9 : SUCCESS in 4m 32s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/00ba2816da114de783cdeec0454a27cf : SUCCESS in 7m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1e51d2e1ffd147af81154bab035a285b : SUCCESS in 4m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3d349b4a4ab64a73808c705ce6e86435 : SUCCESS in 8m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/37821a636c6f48c18953c894d90fdfff : SUCCESS in 7m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/92bfab669a2b4ed98f30324a376422dd : SUCCESS in 7m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2de864fd7cbd4a2cb6efc7055ada3830 : SUCCESS in 1h 11m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/17a0a80970dc43b6be7364ae1f5f0f65 : SUCCESS in 1h 55m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9e413f770dca42ff8cb8e37fb80f7c11 : SUCCESS in 47m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/eac1018337b341a9ac05b360dfd70ac2 : SUCCESS in 37m 46s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/61684fa96a7248e5b8b086ad6db4c068 : FAILURE in 23m 38s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/16e0d98d37bc40d691e25186b6c60fd5 : FAILURE in 1h 05m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/30ffd82eb91f4f18bf6a02bb002f9d3d : SUCCESS in 31m 08s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9d6c80f9f9c24c94b252fd7e0f50fcd8 : SUCCESS in 1h 46m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/83e8a4d359c9457a8f93130576f4851c : SUCCESS in 2h 16m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5b17afa704f348188c664d565961b54c : SUCCESS in 1h 55m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4ca18da3040c4dde90cd3e410246a365 : TIMED_OUT in 2h 34m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/eac6b49e0a2e43ab8c54cd6c3a8b707d : SUCCESS in 1h 57m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3cf076747b1d43d18825e8a153f59b51 : SUCCESS in 1h 59m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"be9d6627ea3056af893e17fe2cd1f312534968cd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-31 07:46:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b368ee7741039fc86e626606df67aff31ad4943a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 10:55:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/10031a0f620f470d905582b7ca7495b3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7506cda266de4804836dfcbd57b14084 : SUCCESS in 14m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c05a1eee6a9d41008fc91260655e749e : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00fb935f70af4683bfeede51766c8bed : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/108b254e57914106b28f8652abc23e20 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6f35154b2a542a49e7598ffd2fcf7f0 : SUCCESS in 10m 50s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7bce589d8b12446c8e37879649483864 : SUCCESS in 4m 52s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/dcba27fba391415988dcc15f1f264abe : SUCCESS in 8m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/998ebd7c164e4eb792e89febbc322fcd : SUCCESS in 4m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0e58bdd574424529bd37a52d6e14b6c7 : SUCCESS in 6m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fc167ed1a92c451da455ff141aa21ccc : SUCCESS in 6m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/56c03299e7cd47dabc41e4f941121b41 : SUCCESS in 11m 00s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e4711d87da13455f8182082efe4e09fc : SUCCESS in 1h 10m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c8cd11a9fba460a93ae41d6823550c1 : SUCCESS in 2h 06m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/955951a4f73a4e908d8f21103d950ec1 : SUCCESS in 38m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2c4c45fd15f54ad28e0d31283e71dc88 : SUCCESS in 43m 41s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/2a251da267a848748e3c8b689808811f : TIMED_OUT in 3h 03m 27s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fd19ad3d944e4aa4a3c638f6d117ee20 : SUCCESS in 1h 05m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4b1c27a605c44d94a2787122924920dd : SUCCESS in 30m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/54e9df4d171140edb76fdd6376dc2d90 : SUCCESS in 2h 17m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/15aaa82ed5f843f68a7493f9d82f85c1 : SUCCESS in 2h 25m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/df7bc577953e465eaa8065602e4dd345 : SUCCESS in 1h 50m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1ac06f0639074fd1bc1a6304a7ce6972 : SUCCESS in 2h 10m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3601d6af426f49b487c14733283ad0a1 : SUCCESS in 2h 03m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c05b43733ce945e3b4aa01e6b981b6e3 : TIMED_OUT in 2h 33m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d9f8c9dfcc7cdcc6e2c9f680d635187bebf7a978","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-31 12:28:15.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"479020ab4fd038eee39feb622cd1c65ea0937ba1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 15:39:17.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5fe5a76bfcce4628a51f60e244a2b15c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8ad44f50925454a814d3fa96e85ae89 : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c024ce1ce0b84f9a9e48078a1ab2ae39 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/049bb433870c4825805340ffbf254da6 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/127c4155a6724bcb8b1c73ea0beda559 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a7fb9261a9149eaa3a1d66f8baa2bf5 : SUCCESS in 11m 15s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d5a43d95859a4b6fbfe22a1ceafccd14 : SUCCESS in 5m 44s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/2aeb86222c434f36bbe77d2a7cb6bfc0 : SUCCESS in 7m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c9893c7becef4a00adcc5b6e82f261ea : SUCCESS in 4m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b4a6d74e9a4d4b0ea268c5c8a12d2cee : SUCCESS in 7m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3eebaaba23e84878bd842772bee1a946 : SUCCESS in 7m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d2eae793cda74b1e84e63bf04e666abe : SUCCESS in 7m 07s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/58a0d17df22840d9b7fd189a76071563 : SUCCESS in 1h 31m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d609bbd2075d4d6d9e5594b6da94e138 : SUCCESS in 1h 40m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/690f14c8d45140a7b2633bcefbb8bf4e : SUCCESS in 32m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cb851d51da3d475bac9d6cdb532a0512 : SUCCESS in 37m 03s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f956b1d3969044fbad6bb1ad77d08cde : TIMED_OUT in 3h 02m 23s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c5adc76812ab4fad8661e14a198ac391 : SUCCESS in 1h 01m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4d8ae3f06924d0b88762fd702c8114c : SUCCESS in 30m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/272a21e5547a43bdb9055ae4985e3346 : SUCCESS in 2h 07m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3283bd0b3b80484595c847d058a2981f : SUCCESS in 2h 08m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d64feca668204a889d5cde393b394c01 : SUCCESS in 1h 45m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c8b065490c274ee18a4f1f20b95cb9ee : SUCCESS in 1h 53m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1a766344f0d14f0898782da360f3d79b : SUCCESS in 1h 47m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7268cebb41774347b3a46084eec0379c : TIMED_OUT in 2h 34m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5a231cf852f4d70fe7240bcb0034efdee2d42b9b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-03 07:30:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fbf58934cc51968ea2cc30f68c9b3531517bb82e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 10:34:18.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c463157cd4d54bf9ad5b44b089725ada\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84b647fdce3f4600af917edaf5e89eac : SUCCESS in 12m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fd94b83ca63443f82df18441d954a8d : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a06d250554bd4ac49358fc45bfd090d6 : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8799cdbed00e4d71aac379b93700a6ab : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd08e886149c484dbc15e9f954141ecc : SUCCESS in 13m 10s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5c890cc4c3974e1cb357eb82f07271e9 : SUCCESS in 5m 56s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5b1f1139946f4beeba196882fa4de607 : SUCCESS in 8m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d2f36c21f4fa434980f68867e87c236e : SUCCESS in 4m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a6d614a1d28148a4a1c660da096dd52d : SUCCESS in 8m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5f38d3cb8637407aa63ff6d607dadfc3 : SUCCESS in 7m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e933ed020c134ccf9d00cb45481a50be : SUCCESS in 5m 44s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8e47982b67ff437a8bf00d11ee594ce8 : SUCCESS in 1h 28m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/affed2d644934a2e8951dd1914f4ab6d : SUCCESS in 1h 58m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f8dc1a5e0571489592f57537626da38d : SUCCESS in 33m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3b4ec05a4aa841528e7f369174553304 : SUCCESS in 37m 20s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/8cd4a1ea0b094a5ba42d562a3eafa306 : TIMED_OUT in 3h 02m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c0e5c75c9256443a8e9a9e3d2cbfeab2 : SUCCESS in 1h 09m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/776a530787044e289256deab01d0de0e : SUCCESS in 29m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/65c5b281b5a142ae99597e1879e0cd18 : SUCCESS in 1h 45m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9ce310a48b954dd48b082077207d9cc7 : SUCCESS in 2h 18m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/97f22859a3f64c5392603c481f3567fb : SUCCESS in 1h 58m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ebfd33e55abd47d791d7ffc81a39763c : SUCCESS in 1h 45m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0f090a7f66954fdb8b05316301d685e8 : SUCCESS in 1h 58m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9db9e2cd3224d6bba914fa858fdd32d : SUCCESS in 2h 12m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8dfa2360b280647c4b765d505598cb2d88d8f630","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-04 11:04:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e03d1bffc4517eed5c2454cbd216d1200ca416ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 11:43:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/915f151be6ae4b0e99d33b6937e8b810\n\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8da394dd275d425fb92875a82b9480cb : SUCCESS in 37m 21s","accounts_in_message":[],"_revision_number":5},{"id":"10aa28ef02561999617fb2be202eae3656d75882","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-04 12:34:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ccf31cbb826bcebb53cd133e4c818d69aad1bc79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 13:21:58.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9c139e8e9aa4825b4f669f51b0f2283\n\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/869743b4ad5a42e4a90cd44ace10d5c0 : SUCCESS in 45m 10s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/7adebe56652040fea1ba5ace72572634 : FAILURE in 40m 12s","accounts_in_message":[],"_revision_number":6},{"id":"de16da32ebd83cc0aa81d0231649aa2e54a52031","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-05 06:07:58.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d21510005511490c5e5088dda6a02de502570beb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 06:56:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f40704effb6c4ecc992043ebf54f0b51\n\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8207238320c543dc9b771c7b0108f14d : SUCCESS in 43m 37s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/faadeebfa5b54b2f87119fb8b351991e : SUCCESS in 42m 48s","accounts_in_message":[],"_revision_number":7},{"id":"8a09bc5feeb7ebae450fc0296cc933572d4f61c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-05 07:56:48.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"78a8516a28db24a5a4536bce3a8878ea440f8a4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-05 07:57:39.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"7ba6197e70540efd37d4a45fed451d655a63d0ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 10:42:18.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b441bcf81de34b70ace76dae61a8bea3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/109fb8eaf60645f48edf6f7b26199e8b : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f9ac614023f42c3a589be7745e839b8 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb386215a4c94caab17b7909eca94b3d : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d5823d5ed084a6d8b726397b063c949 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d4deabe23d348b095ac19615dc5d635 : SUCCESS in 12m 14s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/77938d2dfc814f24b1cee06ffd9fd7a0 : SUCCESS in 5m 11s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f2bb965c634d4e7893776121ba9dd3f8 : SUCCESS in 7m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2393a7e2be454f0ca26ae88e8dd8731b : SUCCESS in 4m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/22d132a587924f8b879ecc1759c4f232 : SUCCESS in 7m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/44a44445c88f485a8dfa6ba433e8c036 : SUCCESS in 8m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c6f6e54abdb14c809d232060e8763ffa : SUCCESS in 7m 27s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4637ea9924434e6abecd8d12aa29a25f : SUCCESS in 1h 33m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/94dd5f47729b441e9d2267fe7808675b : SUCCESS in 2h 24m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/271473e3a1df4649ba301c164cc03808 : SUCCESS in 46m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/89a56ae558644cef81b436c6cb9e7524 : SUCCESS in 45m 23s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/794a6de762df4ac4b678a7f1260259f2 : FAILURE in 5m 45s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d21586291dd24bc9b4b1147174a15c01 : FAILURE in 37m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e7bd47d18e784c2b8a624788c5ae5177 : FAILURE in 5m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b83cd59f95a74cc79170ced3409040e1 : TIMED_OUT in 2h 33m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f3897c185c8940f09bb6aee77f5fe931 : SUCCESS in 2h 42m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1d523ece5d1c4c5db0991bb152014f20 : SUCCESS in 2h 15m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6aecb9ed423f49c2b2bebae8babf9ac5 : TIMED_OUT in 2h 33m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a93dfb850a9e461496c133a76a86763b : FAILURE in 1h 48m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/60ed3dc172b04655bd3e2907190a32a1 : FAILURE in 25m 34s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"63549f0c48fd9f0ca0f159bfbd65c2ed3d93b65c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-05 15:58:41.000000000","message":"Patch Set 9: Backport-Candidate-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0cb310759e10e081aa11d71347011ecf38b25148","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 18:25:28.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b88056fe533040199f0f998a27b7207e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13dc7d9ca7934692948d7870d2635d6a : SUCCESS in 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c40d01d2e0b4db9baf1aa815c73cc72 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07069f20796f40d584a2b9ddb0a4be9e : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/faf283921c564a12b247697d0e303d86 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34ed21f4b4c9446db0526f89c1ff5333 : SUCCESS in 10m 55s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3efad20f286048398ff3223018feadac : SUCCESS in 4m 49s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/cbdcdf676e1a4673a9b3e40f481ec088 : SUCCESS in 10m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/00c1ad4dd69b4145ba2f1666a495ad6a : SUCCESS in 3m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8899300a04424cfd9f633eb7081b39c1 : SUCCESS in 7m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/00124bb3f500436c9338385af84998cd : SUCCESS in 7m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6edab96e348a4e938d2c3f4d7515578b : SUCCESS in 7m 13s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/67514b54f1704edb986cf0f62f3e3416 : SUCCESS in 56m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bce0fd3a90fc44d4bdeb2966dc4f89d4 : SUCCESS in 2h 22m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dd7d85eae1094f75a03b404da215d4d0 : SUCCESS in 46m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/99014453a6fa40a7af05f19b5b7f2202 : SUCCESS in 48m 29s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/cbb2dfb354144bc5ab250505cbe7f13e : SUCCESS in 44m 11s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b0f77b19eb93494aa8b289f8aa8c1a09 : FAILURE in 30m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67ecc65808e64282b2dd1a6245835d19 : SUCCESS in 38m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/aa73913ab3514e33bcc8432363bd57ab : SUCCESS in 2h 15m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aa4071e5f1d84c3b84d51dc92a59c742 : SUCCESS in 2h 19m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bdfff458fe464d97bcc2981a296a0166 : SUCCESS in 1h 59m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/316d2163cd3543d1a053e2df84d4c17e : SUCCESS in 2h 23m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a14c935639fd428d96dc0cb3d7079ee9 : SUCCESS in 2h 22m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2b68fd6043842bd9bd30ea1f1555d23 : FAILURE in 25m 23s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"143e7bbc0180532c6ab4c4252bfcc6380a95c42d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-06 10:18:34.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"bc15d49f254161dfea068beb281e86071c3833de","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-13 18:10:38.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"92cdf1299a2205128a7cea5fcfb03da8e7792c29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 18:11:19.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"1a2cf3a1e5da9f79f25067a58e731e39c654b1fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 20:18:21.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1921985c59a4e2f81636b33f116289b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66e42943dd004bceaee0774de3d6e4d0 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/805d17bbfbcf491f8e179216f01170f3 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/503ca84175e042e28682676a64029391 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5074f36c52f046aea3d6565e5c5e1954 : SUCCESS in 13m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2a5e148a29ec4128adb0e703c96c03cd : SUCCESS in 4m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5edd849f89e44260b3e7f2c8afaf4591 : SUCCESS in 6m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1f91ce773d9a444c91f7ffa427181293 : SUCCESS in 1h 51m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/04b42c2d51fa49d5a0bad880733aada8 : SUCCESS in 1h 55m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a329c001f1aa4539a4fcdee6a602165c : SUCCESS in 1h 48m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b087ab3db63c4160a38e461fb1dd388d : SUCCESS in 38m 45s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/ee1e3e675a6e421d8dcda4f7f8017021 : SUCCESS in 40m 05s","accounts_in_message":[],"_revision_number":9},{"id":"6fd0833d6531d7a19314c64e384738230a5f4505","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 20:18:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"e6b229a2162b7065013e5ebc86551f31c3f45bdb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 20:20:40.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c865f85d628a44f780312963f8e9a98d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f2f0024c3944b3ebdd10c6aa595a8d8 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"487938f2c06b71eb411e734c86c3810a363370bb","revisions":{"c8314948a70dfd8d98c7be85b308c1478ea197b4":{"kind":"REWORK","_number":1,"created":"2023-03-29 16:25:09.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/28/878928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/878928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/878928/1"}}},"commit":{"parents":[{"commit":"78dfb08f070c8445b328fe0e38a1900debd2dae6","subject":"Purge some dev tools from the amphora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78dfb08f070c8445b328fe0e38a1900debd2dae6"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-04 08:49:05.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-03-29 16:24:58.000000000","tz":120},"subject":"Add octavia-grenade-slurp CI job","message":"Add octavia-grenade-slurp CI job\n\nAdded new job according to [1] to test the new SLURP (a.k.a tick-tock)\nupgrade strategy.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-September/030654.html\n\nChange-Id: Ia8c35071f54e6988d9b2ff682a529bd477b93b74\n(cherry picked from commit 5eb234b827fc630cbb19c432513601d3971cff34)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8314948a70dfd8d98c7be85b308c1478ea197b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8314948a70dfd8d98c7be85b308c1478ea197b4"}]},"branch":"refs/heads/stable/2023.1"},"d101ad57a5c53155fcb8c4f240e1e234ed1cb2c5":{"kind":"REWORK","_number":2,"created":"2023-03-30 12:07:26.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/28/878928/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/878928/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/878928/2"}}},"commit":{"parents":[{"commit":"78dfb08f070c8445b328fe0e38a1900debd2dae6","subject":"Purge some dev tools from the amphora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78dfb08f070c8445b328fe0e38a1900debd2dae6"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-04 08:49:05.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-03-30 12:07:00.000000000","tz":120},"subject":"DNM test Add octavia-grenade-slurp CI job","message":"DNM test Add octavia-grenade-slurp CI job\n\nAdded new job according to [1] to test the new SLURP (a.k.a tick-tock)\nupgrade strategy.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-September/030654.html\n\nChange-Id: Ia8c35071f54e6988d9b2ff682a529bd477b93b74\n(cherry picked from commit 5eb234b827fc630cbb19c432513601d3971cff34)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d101ad57a5c53155fcb8c4f240e1e234ed1cb2c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d101ad57a5c53155fcb8c4f240e1e234ed1cb2c5"}]},"branch":"refs/heads/stable/2023.1"},"4e94977657912bc623ba57faaff5559b795668a1":{"kind":"REWORK","_number":3,"created":"2023-03-30 16:34:51.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/28/878928/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/878928/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/878928/3"}}},"commit":{"parents":[{"commit":"78dfb08f070c8445b328fe0e38a1900debd2dae6","subject":"Purge some dev tools from the amphora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78dfb08f070c8445b328fe0e38a1900debd2dae6"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-04 08:49:05.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-03-30 16:34:41.000000000","tz":120},"subject":"DNM test Add octavia-grenade-slurp CI job","message":"DNM test Add octavia-grenade-slurp CI job\n\nAdded new job according to [1] to test the new SLURP (a.k.a tick-tock)\nupgrade strategy.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-September/030654.html\n\nChange-Id: Ia8c35071f54e6988d9b2ff682a529bd477b93b74\n(cherry picked from commit 5eb234b827fc630cbb19c432513601d3971cff34)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4e94977657912bc623ba57faaff5559b795668a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4e94977657912bc623ba57faaff5559b795668a1"}]},"branch":"refs/heads/stable/2023.1"},"cb22325b195f325b80711a702860324fe269e2d0":{"kind":"REWORK","_number":4,"created":"2023-03-31 07:46:33.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/28/878928/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/878928/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/878928/4"}}},"commit":{"parents":[{"commit":"78dfb08f070c8445b328fe0e38a1900debd2dae6","subject":"Purge some dev tools from the amphora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78dfb08f070c8445b328fe0e38a1900debd2dae6"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-04 08:49:05.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-03-31 07:46:24.000000000","tz":120},"subject":"DNM test Add octavia-grenade-slurp CI job","message":"DNM test Add octavia-grenade-slurp CI job\n\nAdded new job according to [1] to test the new SLURP (a.k.a tick-tock)\nupgrade strategy.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-September/030654.html\n\nChange-Id: Ia8c35071f54e6988d9b2ff682a529bd477b93b74\n(cherry picked from commit 5eb234b827fc630cbb19c432513601d3971cff34)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb22325b195f325b80711a702860324fe269e2d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb22325b195f325b80711a702860324fe269e2d0"}]},"branch":"refs/heads/stable/2023.1"},"ee7bb3924db43877a2075232d5112834361fef21":{"kind":"REWORK","_number":5,"created":"2023-04-04 11:04:21.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/28/878928/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/878928/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/878928/5"}}},"commit":{"parents":[{"commit":"78dfb08f070c8445b328fe0e38a1900debd2dae6","subject":"Purge some dev tools from the amphora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78dfb08f070c8445b328fe0e38a1900debd2dae6"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-04 08:49:05.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-04 11:04:14.000000000","tz":120},"subject":"DNM test Add octavia-grenade-slurp CI job","message":"DNM test Add octavia-grenade-slurp CI job\n\nAdded new job according to [1] to test the new SLURP (a.k.a tick-tock)\nupgrade strategy.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-September/030654.html\n\nChange-Id: Ia8c35071f54e6988d9b2ff682a529bd477b93b74\n(cherry picked from commit 5eb234b827fc630cbb19c432513601d3971cff34)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ee7bb3924db43877a2075232d5112834361fef21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ee7bb3924db43877a2075232d5112834361fef21"}]},"branch":"refs/heads/stable/2023.1"},"367d429a2fc0a63458ec099dc0e71edfb669dfb6":{"kind":"REWORK","_number":6,"created":"2023-04-04 12:34:49.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/28/878928/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/878928/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/878928/6"}}},"commit":{"parents":[{"commit":"78dfb08f070c8445b328fe0e38a1900debd2dae6","subject":"Purge some dev tools from the amphora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78dfb08f070c8445b328fe0e38a1900debd2dae6"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-04 08:49:05.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-04 12:34:43.000000000","tz":120},"subject":"DNM test Add octavia-grenade-slurp CI job","message":"DNM test Add octavia-grenade-slurp CI job\n\nAdded new job according to [1] to test the new SLURP (a.k.a tick-tock)\nupgrade strategy.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-September/030654.html\n\nChange-Id: Ia8c35071f54e6988d9b2ff682a529bd477b93b74\n(cherry picked from commit 5eb234b827fc630cbb19c432513601d3971cff34)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/367d429a2fc0a63458ec099dc0e71edfb669dfb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/367d429a2fc0a63458ec099dc0e71edfb669dfb6"}]},"branch":"refs/heads/stable/2023.1"},"efc6c490045d1e6cbdc3bceb7f21e06f798c2205":{"kind":"REWORK","_number":7,"created":"2023-04-05 06:07:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/28/878928/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/878928/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/878928/7"}}},"commit":{"parents":[{"commit":"78dfb08f070c8445b328fe0e38a1900debd2dae6","subject":"Purge some dev tools from the amphora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78dfb08f070c8445b328fe0e38a1900debd2dae6"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-04 08:49:05.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-05 06:07:52.000000000","tz":120},"subject":"DNM test Add octavia-grenade-slurp CI job","message":"DNM test Add octavia-grenade-slurp CI job\n\nAdded new job according to [1] to test the new SLURP (a.k.a tick-tock)\nupgrade strategy.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-September/030654.html\n\nChange-Id: Ia8c35071f54e6988d9b2ff682a529bd477b93b74\n(cherry picked from commit 5eb234b827fc630cbb19c432513601d3971cff34)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/efc6c490045d1e6cbdc3bceb7f21e06f798c2205"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/efc6c490045d1e6cbdc3bceb7f21e06f798c2205"}]},"branch":"refs/heads/stable/2023.1"},"8ccdc7e6eac6a2b99706aa93c5fc405f1bd6df75":{"kind":"REWORK","_number":8,"created":"2023-04-05 07:56:48.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/28/878928/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/878928/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/878928/8"}}},"commit":{"parents":[{"commit":"78dfb08f070c8445b328fe0e38a1900debd2dae6","subject":"Purge some dev tools from the amphora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78dfb08f070c8445b328fe0e38a1900debd2dae6"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-04 08:49:05.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-05 07:54:10.000000000","tz":120},"subject":"Add octavia-grenade-slurp CI job","message":"Add octavia-grenade-slurp CI job\n\nAdded new job according to [1] to test the new SLURP (a.k.a tick-tock)\nupgrade strategy.\n\n[1]\nhttps://lists.openstack.org/pipermail/openstack-discuss/2022-September/030654.html\n\nCompared to the original patch modifications to the job vars were\nnecessary because devstack\u0027s default is ML2/OVS in yoga\nand ML2/OVN in 2023.1.\n\nChange-Id: Ia8c35071f54e6988d9b2ff682a529bd477b93b74\n(cherry picked from commit 5eb234b827fc630cbb19c432513601d3971cff34)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ccdc7e6eac6a2b99706aa93c5fc405f1bd6df75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ccdc7e6eac6a2b99706aa93c5fc405f1bd6df75"}]},"branch":"refs/heads/stable/2023.1"},"487938f2c06b71eb411e734c86c3810a363370bb":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2023-04-05 07:57:39.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/28/878928/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/878928/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/878928/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/878928/9"}}},"commit":{"parents":[{"commit":"78dfb08f070c8445b328fe0e38a1900debd2dae6","subject":"Purge some dev tools from the amphora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/78dfb08f070c8445b328fe0e38a1900debd2dae6"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-10-04 08:49:05.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-05 07:57:12.000000000","tz":120},"subject":"Add octavia-grenade-slurp CI job","message":"Add octavia-grenade-slurp CI job\n\nAdded new job according to [1] to test the new SLURP (a.k.a tick-tock)\nupgrade strategy.\n\nCompared to the original patch modifications to the job vars were\nnecessary because devstack\u0027s default is ML2/OVS in yoga\nand ML2/OVN in 2023.1.\n\n[1] https://lists.openstack.org/pipermail/openstack-discuss/2022-September/030654.html\nChange-Id: Ia8c35071f54e6988d9b2ff682a529bd477b93b74\n(cherry picked from commit 5eb234b827fc630cbb19c432513601d3971cff34)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/487938f2c06b71eb411e734c86c3810a363370bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/487938f2c06b71eb411e734c86c3810a363370bb"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
