)]}'
{"id":"openstack%2Foctavia~879323","triplet_id":"openstack%2Foctavia~master~I7f43b30de9accfbf78a9d272fecd571de3b1f90d","project":"openstack/octavia","branch":"master","topic":"pydev-trace","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-04-26 10:00:06.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I7f43b30de9accfbf78a9d272fecd571de3b1f90d","subject":"Change Pydevd settrace configuration","status":"MERGED","created":"2023-04-03 10:09:36.000000000","updated":"2023-04-26 12:08:51.000000000","submitted":"2023-04-26 12:07:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879323-pydev-trace","meta_rev_id":"ba6f49bc35a717ba44791beb41c4b1050301712a","_number":879323,"virtual_id_number":879323,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-26 12:07:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2023-04-26 09:19:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-04-14 09:42:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-26 10:00:06.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-26 10:00:06.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-03 12:47:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-14 09:42:34.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-04-26 09:19:16.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"b5904e36f6f9c9c106ec8f7bec9e085fb4e3ad0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-03 10:09:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"112bf83f4afd09c9e34b5559add9107e8e78cf31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 12:47:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/798ca80ad91c419aac7efa933be1dac8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c16d0ab965964eeb9aa087f065587277 : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5069669640d940f2a6e3a3e760f19344 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b60b5662f8c4a63aa227b29184478ec : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da22455c3f794b0ca6402c5556a20d10 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3d01445333c43b283e8a0b47c8e353b : SUCCESS in 12m 02s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/811f4ce69b5c45e397f465d92d308bc4 : SUCCESS in 4m 56s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0d2b915d6e154b37983b72f837adadd3 : SUCCESS in 8m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/316756575b6344859f2b2c17eab700b7 : SUCCESS in 4m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1c2ed993afb24d4eb65e80d9b06a8ef7 : SUCCESS in 8m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/151547f600e940feaf902abebddad95f : SUCCESS in 6m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b1bea3713c0b4a809da2117f9346eae0 : SUCCESS in 7m 34s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6cb74072399a44b2b598a7c1638bea79 : SUCCESS in 1h 32m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6e9d56def91446daa8e6bd52b2e24662 : SUCCESS in 2h 01m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d6d4c6ecc23b41e99a3f1496737a5b3f : SUCCESS in 32m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/84f62943acc94b569e5f1c06a786ceb5 : FAILURE in 27m 40s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/416350ebd3784049ba3ea0a90ac38011 : SUCCESS in 1h 03m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/472eea377c734778b5b6e452d3d6db2a : SUCCESS in 40m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9396ea22ffc4445f90c7186665ae76cb : SUCCESS in 2h 08m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a244c0ac5ee24e089016d179ea4a2526 : SUCCESS in 2h 34m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/25c965e6eae4406b9e5d363351a6b25c : SUCCESS in 2h 08m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fb4f0f9c1e864c72847eb9630e35443a : SUCCESS in 2h 03m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2b7792cab1cf4378af006c30f95babb8 : SUCCESS in 2h 12m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5762cab67f394f08a9cff0d5fc58f394 : SUCCESS in 2h 24m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26c0c6f7218f7ae822d1640013ccabced751000c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-03 15:59:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aa44036f17ac2177815a6aff35b2f3c42ff5eea4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 19:26:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96dcb554af30440492a4f5a59057c6a1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e6fea2a37774bea81ffc8514aa2fa52 : SUCCESS in 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b513f6689e24ff5891a6018fc11b9ac : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4aef8d09d1344d82be7044698d9e28a0 : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e8624ee29f094c8e9dcaa1b8399a2b87 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6b2933d024b45a0986b7b7af026bb9e : SUCCESS in 12m 01s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5889daaf4b5b4b4bbb8892d11e6c517f : SUCCESS in 6m 22s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/bde693a0796d42af9b305af975cd626e : SUCCESS in 7m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1c382e52954742b0b3861870e1cf9684 : SUCCESS in 4m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fed69138acbd47cd9d23ad8300f83fa8 : SUCCESS in 8m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7a466b91bb1245b68158b6288b4e8f64 : SUCCESS in 7m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/27f2cc43b62b4f269eede277ebb6f53d : SUCCESS in 8m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/795001ae7d7b43f8a0a38447133e4f50 : SUCCESS in 1h 21m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aed65fbf982d4fe89a7d0252c28b3f38 : SUCCESS in 1h 35m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9a1a05624e324831869025b967e20474 : SUCCESS in 36m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6b2f5669788d4c1fb8fbaa916d6331f1 : SUCCESS in 40m 01s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/675be36840444104b45e086b98df897b : SUCCESS in 1h 08m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d3b18f76401447a8b580d43b741dbcc4 : SUCCESS in 33m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a552674f5e5648cbbe87f877c7472512 : TIMED_OUT in 2h 33m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/09527aef68824900923187a00798097f : SUCCESS in 2h 42m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2d546d055425416d90e71dacfbc7bd49 : SUCCESS in 2h 00m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1c80f4f7f6474e04be2af6f3e83a6a2e : SUCCESS in 2h 18m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/df71a203cfef470db3fb57376fa84cb7 : SUCCESS in 1h 45m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/92c0068ab5ac4187a5a388b7c4e07bdf : TIMED_OUT in 2h 33m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ad0b65c9cc6ed81e9c07f1089375053588e6d89","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-14 09:42:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2c2b2f8a7d201c7325de218482bd506c7b6f0b12","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-04-26 09:19:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c0466b6bf30b5e070bd7ed13c94327978ca6c8d5","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-26 10:00:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a514fd8c7380b9b1da839a10c8d75463e1310ebe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 10:00:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4d3e2bb413bd8283c4c0bdb759199145025d8fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 12:07:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/645336f030db4fb4b49000c9a7935ad6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65181b3869674eae94d7819726efff4a : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e28a4356f2fa4eaa9462cb9386f71e5c : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6130e35d1a9840b983a6d2ee8d3f0133 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c0fb7260b0a48da89aa37c5ab88130e : SUCCESS in 10m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a5b7b2a84ed54cdd92f8239f4b0568a7 : SUCCESS in 4m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1ae2e8ff83de475b9fc22079866354bd : SUCCESS in 7m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0cc48bfa46dd4fe8a6aff0690bc451c4 : SUCCESS in 58m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/13a9bee0b8f9491ca5c574bb967be315 : SUCCESS in 2h 05m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ec392b232ee945be9724a739b1f757d0 : SUCCESS in 36m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9b1d631562f14fa98df7f8377873aa67 : SUCCESS in 46m 27s","accounts_in_message":[],"_revision_number":1},{"id":"7c5e4c49cf69fcaeda50221494282f1a907d4949","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 12:07:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ba6f49bc35a717ba44791beb41c4b1050301712a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 12:08:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94d04870a31d4a42ad9f38b0000c2c95\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d568fc1527c45b2bf98cb83efd5a046 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9284b74069a969e3e3908a5f9a3ef8c1e55b5a76","revisions":{"9284b74069a969e3e3908a5f9a3ef8c1e55b5a76":{"kind":"REWORK","_number":1,"created":"2023-04-03 10:09:36.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/23/879323/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/879323/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/879323/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/879323/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/879323/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/879323/1"}}},"commit":{"parents":[{"commit":"b0c9cff6b0ec25b02b41f59486ae6e2bf9a4da7a","subject":"Merge \"Fix ORM caching for with_for_update calls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b0c9cff6b0ec25b02b41f59486ae6e2bf9a4da7a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-03 10:04:34.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-03 10:04:34.000000000","tz":120},"subject":"Change Pydevd settrace configuration","message":"Change Pydevd settrace configuration\n\nBy adding *suspend\u003dFalse* the debugger will not suspend execution at\nthis call already. This is consistent with the behavior of\ndebuggers in IDEs like Visual Studio Code and PyCharm (which is the IDE\nthat uses pydevd).\n\nChange-Id: I7f43b30de9accfbf78a9d272fecd571de3b1f90d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9284b74069a969e3e3908a5f9a3ef8c1e55b5a76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9284b74069a969e3e3908a5f9a3ef8c1e55b5a76"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
