)]}'
{"id":"openstack%2Foctavia~879871","triplet_id":"openstack%2Foctavia~stable%2Fyoga~I174e63cfe176e1792d0bbcd348dbc9115544e297","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-07-03 11:19:54.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I174e63cfe176e1792d0bbcd348dbc9115544e297","subject":"Enable taskflow retry feature when waiting for compute","status":"MERGED","created":"2023-04-07 09:05:40.000000000","updated":"2023-07-03 15:22:06.000000000","submitted":"2023-07-03 15:20:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879871","meta_rev_id":"9df43ee71f80f550acae6fba1037faa34420f9f5","_number":879871,"virtual_id_number":879871,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-03 15:20:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-04-13 18:18:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-13 15:51:08.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-04-13 18:18:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-07 11:32:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-13 15:51:08.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-04-13 18:18:31.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"217ca019a877017d1cb871f7c6c0e921a2f89660","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-07 09:05:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"011677dab9bd15228cecbfba57fca8af9f31ca31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 11:32:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b374561095f4a50b4819cee971411e9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fd3cb4dd7fd4b03ad7132e1756dc500 : SUCCESS in 10m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/babc2d299c0c4976b573270ca3e2b5fa : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6caa0e2013c046c2870fbfe5c0df477a : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d08868a926ea4d07950269563394bfa7 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df33927f247c4958a3b2949793f49124 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed5f398c4ecf4d3ba39ab13e178f4e7a : SUCCESS in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67464636373a49daba51038479bce4d9 : SUCCESS in 4m 57s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e5b08c6b0ab949aab13fca41cc3d0373 : SUCCESS in 4m 54s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d6717e3fba0b4dc8a14c09831add9842 : SUCCESS in 5m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5f8e435396374467a21980c6eb5c9d72 : SUCCESS in 4m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1f858d7417f74ec0b5a5ed939c633bce : SUCCESS in 9m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f31bef48ab7c45209000aecccbecf4ed : SUCCESS in 1h 32m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ff3772eee4440b9a76b377f42d7b114 : SUCCESS in 1h 49m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/56553d4e140148ea987d0104add79229 : SUCCESS in 40m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9979baefaafc4d77a6a92aa6a3131f4d : SUCCESS in 40m 27s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/db633fc08f9747b0be73a05fe1f9f5f7 : FAILURE in 35m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f66395345f934caa81b50cf196f308af : SUCCESS in 36m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3a57ec1e5e454fc18dd503dfd99fdfc3 : SUCCESS in 2h 08m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5760ed09853040e99cb6b919275654cc : SUCCESS in 2h 17m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/45a4e704212949a3a95359f1e3607b2f : SUCCESS in 2h 03m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/30aa52ba512d4a5aa23963139a5891e8 : SUCCESS in 2h 11m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/acb76d310e084a368832d9e1e6cb569a : SUCCESS in 2h 24m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c7f2e8ca43849f5a1fa11007f6db618 : FAILURE in 49m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45e424c5bae0cb5b42b9dee0c0a5054352358542","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-13 15:51:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ad68d9c5548a1e950e2a28e9d24a13558b0e2dc3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-13 18:18:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed2fb885910ed4d98afecb17c3dca87c12299b95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 18:19:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e2ee5a1144cbf6e1c2990137f121685a43f7368","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 22:31:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc823ab8d08e4d9b976634b9b686b836\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4755be6afac4fa19319e5aa086bf9be : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2169fd75d9a491faac70e5a1b02539f : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/470aca439cc846f095bcf140dda201f9 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/262d7e7df9804b6ea224186b5aa0d622 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2f29ac992664e7cb719a3901c429f93 : SUCCESS in 10m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b0e664e6ba5485ba520347ae0f86623 : SUCCESS in 4m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ce648d2e65cb4e918a9724fa2c777b5f : SUCCESS in 4m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5bddc1d348f843248a9ef1929212f054 : SUCCESS in 8m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5b2ac2daf16343dea53415d6bb73b5d9 : CANCELED in 16m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/86614f1f31314cd48934cea9e7cd7753 : FAILURE in 20m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e0d45751f19f4ceb924c53d295dd996d : CANCELED in 16m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e2d656b66e164612bd506baa1253f4ee : CANCELED in 20m 07s","accounts_in_message":[],"_revision_number":1},{"id":"342929ae087089da3f5d0ad6992b1010624a5299","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-22 18:32:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24133ab2759fc3ebea85578c8ceecbd5dd133c86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 20:38:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9840ce0edef6416fbb32cdc76171b6d2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df647315798f411d991517458b5dc85a : SUCCESS in 9m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0608248cdbfc438bbaad077cafe13039 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c66f0f45ae5043daaf8cbfb46836d371 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdc214c1969047e599653d68a756cbe0 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11275ce4714e4943bd689e32193e6912 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53ea2b9de7894f92a9248cadbbf91a42 : SUCCESS in 12m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36ba015a464c4f378fa027273a617cb3 : SUCCESS in 5m 44s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b668419648504a8b988da11bc811248c : SUCCESS in 8m 07s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/76c52b99f8ce4a4c9d74f6511cb57d93 : SUCCESS in 5m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9627fb6d33574f07bca79eed9b069c40 : SUCCESS in 6m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c89141814e194abbbd84a99b2391058a : SUCCESS in 7m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5e400bb8288a40398ca1679104cf223c : SUCCESS in 1h 39m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/25c53565b48044b5bef6129ff96c7647 : SUCCESS in 1h 44m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/acf2ebce37c8434ab1f16fb0fcbde2d9 : SUCCESS in 36m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ab907a3ea0c14ceb9bdd879d1a63ebe9 : SUCCESS in 38m 01s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2665037fb89943f48ae22f2d798f9d54 : NODE_FAILURE Node request 300-0021520522 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/184b6b689f844ee799f17d8485911810 : SUCCESS in 34m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/130fa162e3274e009006a91cb27f9a70 : SUCCESS in 1h 51m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a796617c46e141f684bfd546fb62f665 : FAILURE in 10m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2f1ea22ac1a144b484acefd6a610d69f : FAILURE in 7m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/16c58324d9ec463c88a3f31bc57d036d : SUCCESS in 2h 03m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/809071e56fa64f6e94cba7060a5a3432 : SUCCESS in 1h 38m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/60ba91230c56417b86c129979ea28986 : NODE_FAILURE Node request 300-0021520529 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da15475addf0d161e08010cda7a85534a57a786c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 20:39:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ffb46e9e7efb67244d32f17d977ef017f05c0f36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 21:07:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/34c4ab785f3843c889d6807c79e9edc6\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d3f29ad5b604366afaacbdd398d2dfe : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/997476a639cb4ed284cfdf895050740f : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc301451ef104e97bfc93646a1f5d6b0 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9815f0d63549478bb8d41933a18c0700 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/683796ad241b4b8a971151742ffeb871 : SUCCESS in 9m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a1302388db445579fc95de3448c8287 : SUCCESS in 4m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7a20375045b34933903f93106ef49909 : SUCCESS in 4m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c2d8d0134bc54c97ac39779d295d9218 : SUCCESS in 7m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/267791824463422f8a043ebd7ed1b2ed : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b4846ff008fa4b6b90dc140a015fde58 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5277d8c4dd674b6f960910706f77eaf4 : FAILURE in 28m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b6621e5bf2e6477f8c8c8c94723ad7e8 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"d0351b3888a000f9e4b7728e5291be4b9c948075","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-03 11:19:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b738bc9fb241a6222441c8884846f2ece097113b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 13:10:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9b3240cdf0b44c9ae4e36a7b25916ba\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11f212baf7f040be9c47c8998b7d70b6 : SUCCESS in 11m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9a6e49a94bf48d690e4a8038f10c62c : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87e2332619a94bb4bb0bd449eaa1a556 : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c4b8996c60b4f6d8419b76566e42298 : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43d8cc483ee245e586e6de683071fd0e : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ba6ae33f2e243e6a92b310b556995bc : SUCCESS in 12m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/878355e2f3444d8eb9e4fa36502c379d : SUCCESS in 5m 15s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/46fc5f481e894c6699974c2731aacdf7 : SUCCESS in 5m 25s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b7034d682f62458ca18583c16d68db20 : SUCCESS in 5m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cee37b92ad6748a99f043516dfae565a : SUCCESS in 4m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5108845874cd4687854e3b1680117bea : SUCCESS in 9m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f849729d8b984defb461ce9b915e572d : SUCCESS in 1h 41m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/33a33f984b9c492b8551beae58ad0283 : SUCCESS in 1h 44m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/67a1612d3ea442ec9bc416bd9387d68f : SUCCESS in 36m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/19aa8b920c074dc0ac4acf00c5f56d1b : SUCCESS in 38m 00s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bb48ccc9386945b390b9d1497e7f442e : NODE_FAILURE Node request 300-0021602368 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19669b626b2941f9870857f14a3eab9f : SUCCESS in 32m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/47fd6430359a4755b8d98a54e7bf0b14 : SUCCESS in 1h 39m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8de0c05085dd4709971064c676603395 : FAILURE in 10m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fcbee41345c84607b89bc50713c7dca6 : FAILURE in 6m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/01294ee59e6640729d8a98b47d8f8dbb : SUCCESS in 1h 44m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/45252e7e59fe402a831513a6cb99a096 : SUCCESS in 1h 45m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/be2769c9840d42298a5ffded93c6697a : FAILURE in 37m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82449643008e75b200d5bd4138470c1c702ff81a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 13:10:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"378a9958139e6d1f38adde330ea7693195ca9d5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 15:20:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52fe4d1da7e545269d8c64709447b7b6\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cacd976b3af477cabb2694167692518 : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dca74a3a27634dd182d279c7170a5885 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b8183f16efc47cab0ad14e1fd47ad81 : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/400f381e91884cdab0c7acfcf775a45b : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/675e4484a2bc45e4bd768a387817a6da : SUCCESS in 13m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/089d460753644346a3ce29fc7817d079 : SUCCESS in 5m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/30cc2b5a0cdc44be9ffa80f3f99afaa4 : SUCCESS in 4m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/591b2d0087334e699b4f86603b415077 : SUCCESS in 9m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/27e672a390014c69b921bfcc046d9331 : SUCCESS in 2h 02m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/69cddd2034254bacbc4d9f513b8c592d : SUCCESS in 1h 41m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d66a6a5dfb2b4587a7a21da9046629aa : SUCCESS in 35m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/79a1ae79157a4bab9d0b8e0decffe6c0 : SUCCESS in 37m 53s","accounts_in_message":[],"_revision_number":1},{"id":"1894971caa5974ffc8c3972e7aa15f50f5873ccf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 15:20:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9df43ee71f80f550acae6fba1037faa34420f9f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 15:22:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a106b75cf33f481899f8e6edb3829ca7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/266cc8e14ff44017bfe527bf6709599c : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e13ed7e248304d76bd16a29cf6cd7ecc : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ae3cc5b34d5ded02d1015c05a9158f394ca3ec89","revisions":{"ae3cc5b34d5ded02d1015c05a9158f394ca3ec89":{"kind":"REWORK","_number":1,"created":"2023-04-07 09:05:40.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/71/879871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/879871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/879871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/879871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/879871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/879871/1"}}},"commit":{"parents":[{"commit":"0c8c5037ff63f465588df16f0cb11f8ff2f03f49","subject":"Fix octavia to accept [ipv6]:port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c8c5037ff63f465588df16f0cb11f8ff2f03f49"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-03 08:16:01.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-07 09:05:29.000000000","tz":120},"subject":"Enable taskflow retry feature when waiting for compute","message":"Enable taskflow retry feature when waiting for compute\n\nThe ComputeActiveWait task may be blocked for 300 seconds (with the\ndefault settings for amp_active_retries and amp_active_wait_sec) but\nthat triggers trigger issues when persistence is enabled if this delay\nexceeds the jobboard_expiration_time value.\n\nComputeActiveWait is renamed to ComputeWait and now uses the retry\nfeature of taskflow so it would not block taskflow for a long period.\n\nStory 2008956\nTask 42584\n\nChange-Id: I174e63cfe176e1792d0bbcd348dbc9115544e297\n(cherry picked from commit a9ee09a676074eeb619a4d7c3d9912114de50e88)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ae3cc5b34d5ded02d1015c05a9158f394ca3ec89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ae3cc5b34d5ded02d1015c05a9158f394ca3ec89"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
