)]}'
{"id":"openstack%2Foctavia~879872","triplet_id":"openstack%2Foctavia~stable%2Fxena~I174e63cfe176e1792d0bbcd348dbc9115544e297","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-07-03 11:20:05.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I174e63cfe176e1792d0bbcd348dbc9115544e297","subject":"Enable taskflow retry feature when waiting for compute","status":"MERGED","created":"2023-04-07 09:05:57.000000000","updated":"2023-07-03 15:22:16.000000000","submitted":"2023-07-03 15:21:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879872","meta_rev_id":"410a5ce2c466f9ec1eac763ee947713fc9e0ddb3","_number":879872,"virtual_id_number":879872,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-03 15:21:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-04-13 18:18:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-13 15:55:01.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-04-13 18:18:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-07 11:07:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-13 15:55:01.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-04-13 18:18:38.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"dadc0d564fdfb690fddc9679e06b54e1b0fcd49c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-07 09:05:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2895921effdb0a3837382253fdf2016af329bd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 11:07:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f1a85f13e794edcb613b8050356c094\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/808d3167e6bf49ae9ccdf5645be2bb3a : SUCCESS in 10m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9ce75cca11b4ef08dcf3a861a6537fa : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fd4da44f86c45df91ecf529722115a3 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0415d9415b53467392fc1fe51009e9ca : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cd58e940bc341a084a05d82efda77f4 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1981469e8224771ad5058278c6d89bb : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/977f662da59f42a48eabcfb4d7b959a1 : SUCCESS in 11m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55b242c55ca94a05a36f269873530710 : SUCCESS in 4m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f321b58b28b64302bf1ecfa0f7f3de91 : SUCCESS in 7m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c8e4fe60c2694a9184d4d313ca8150c8 : SUCCESS in 8m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/83ddeea0c6ef44ecb823db63b695a740 : SUCCESS in 5m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/751819ccb84b4fe8be9700f03828e830 : SUCCESS in 10m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/348acdece5dd40cfa5c0f8e127583f9e : SUCCESS in 1h 20m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d88f34e71f664349a5887319006106c0 : SUCCESS in 1h 53m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5235dadff10a4854ab1e3dec76bf8168 : SUCCESS in 31m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e79bc757aecd4558928581f449e93731 : SUCCESS in 40m 56s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ad789a1b30d94b65b84c3aac4f1a2ce7 : FAILURE in 9m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c006093df5fa4024a05b3389b5da3a8a : SUCCESS in 30m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b5287c9da98e49be97c9261f3d4bcebb : SUCCESS in 1h 49m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e520fed005a1444da585f1fff08d4c09 : SUCCESS in 2h 00m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a064edccad4f4cfc9e8dfd56e7d1bcc9 : SUCCESS in 1h 54m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0bc79768c5af43df86611de434e2056b : SUCCESS in 1h 53m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c58c63ff20bd452eaa36fa088bce5b92 : SUCCESS in 1h 39m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ef4a48c3af654433af2fcc8829cda6d3 : NODE_FAILURE Node request 300-0020926012 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2bfa6ec45f85d9c3c570bd2e2a4d31ce048d34f1","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-13 15:55:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"199da769ee204c18e8c9889a513bf838a4cc10fa","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-13 18:18:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a63e1eb117c0f74e9286cf8f2ec88a2ad94db70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 18:19:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c33256bd789d40957b65020d60690e564b6ebe9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 22:31:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e143005bfcf485296104dc9df7c30cf\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be7bdcce6516477baee309f068bfe72a : CANCELED in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77cb962525e749ed99efb92484a7fe89 : CANCELED in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b23097e482bb4a759cf85f5413d3d948 : CANCELED in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c488b6b197d43639090bf580707340a : CANCELED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6807f40348914babab207067fac4ba7f : CANCELED in 1m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf26cdc119184bc5af1fa8774fd4d7fd : CANCELED\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/44962e1375d04793b843612b965041d8 : CANCELED\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3f46350c732944e7ab06d53d4a4959da : CANCELED in 1m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/87b8b09b046c495587f98804a8e73f2f : CANCELED in 1m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/77e56f0783ff4f878e918c68cc5c28dd : CANCELED in 2m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/54f7d3cdcab04ac899b10bf9add581c4 : CANCELED in 2m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a64de58a88df4c4e94afe5dd68909139 : FAILURE in 5m 53s","accounts_in_message":[],"_revision_number":1},{"id":"666d248067072cbd9518d3078154534f30558a55","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-22 18:32:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"09259e7bb9b9fe5f976e67287238ee12b658b144","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 20:39:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b983f03509c4ad2bbf8c369f57e1b7b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41cd2a9cc69a459cb37d84381ce3bbb9 : SUCCESS in 9m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51d7bbf54a51487785a33382bc63662c : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d63194b855904d679d1583b1b92ebf04 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a8b942cba674b13bb5595dd3b3b0eb6 : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df2fa93f41294b97ba977971efe339c2 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43a6a30ab5b744cab0bf1e3d055b7367 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b5925941a344826badc5b1e1ee61a6b : SUCCESS in 12m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfa2d4c7732146dfafa0d57feabcc4cb : SUCCESS in 5m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b37d1a91fecd4b70bf076698c4769c4d : SUCCESS in 5m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/61e6bcfdb22c465e9f90f297dd53becb : SUCCESS in 7m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/69c9f82fa8114fa0a2f178022a51b479 : SUCCESS in 5m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/35cac546aeb542a38977d7238b462e41 : SUCCESS in 7m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b50d8deac0fe412a88dddc3f6dc9ad5b : SUCCESS in 1h 23m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e38646c7380b4f69bc5e7d58ce45e1c9 : SUCCESS in 1h 41m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/74fdb046848343b78c1d7e3f6f28f54b : SUCCESS in 36m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/093954e2f9c24aec86c0d604e2b1dbca : SUCCESS in 44m 29s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a4a87e134b564c56b2e4b00006c1caa3 : NODE_FAILURE Node request 300-0021520547 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b2d7685cd674944bee123205aaf7ff5 : SUCCESS in 33m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f9f441a2e0cb475fb6d6ae7f188a185f : SUCCESS in 1h 55m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9e916925cb294381bb09aff598a09d91 : FAILURE in 18m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4945551f1fb544828d295229260a4ed0 : FAILURE in 13m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/aff7727c0fc642b5ba0e15b138b367b1 : SUCCESS in 2h 02m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/304b6de2f0d3475bbaa30c17f2964eae : SUCCESS in 2h 04m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/af6ddd2e9cf14b6ab902a1d34da7936b : NODE_FAILURE Node request 300-0021520554 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ca57ce324ef069c03c3e17c56fadcf3fe7d1c7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 20:39:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"976f3ccd753d97e1fdd2fe1ad21e3326f5079a1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 21:42:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8c66e24c95c4246bec04c29bd6395a3\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a7ab8f698014e5c91c3316836ae8814 : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dd5589568b343cd8f8fa83b4154165b : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87646dc5dc3e41e884b6810722578943 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28de3380505b461eb0cc64dd81c09e91 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c712b71f7a804d56b446af043fd5f7f4 : SUCCESS in 12m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7586d026c20b4e1e891187c13ff1565c : SUCCESS in 5m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/163f9535315d4ab8bfb691c392fa259a : SUCCESS in 4m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/72d01c046bb84a0c913c0cb8e9322522 : SUCCESS in 9m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4d7866d9d9134189a06e8c8a808452c4 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/977ba296e4824035a46f5a900f823c8f : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/92ef27cb39924e0bb80835ca36bd59ad : FAILURE in 28m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/211059e7746844a4a640287125a3dc44 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"832855d2c42df872c50ef92ce9419961ac0f25d7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-03 11:20:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"17b918c50f9aa1ac28773e2bb1e2f01204929089","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 13:25:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5a5fd95bc2b4a9b8ce0aeb0ffc0115a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bee7733e01564b65ab0068e704f5ae49 : SUCCESS in 8m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99b71fe17e954c65aab7314a1ccb3bc0 : SUCCESS in 9m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3673a0bc047a4a1997b98ca76db55f7d : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc8be458ed0048a48c4cefadc8be50b2 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b85242903cc4287a357f455fc379a89 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41d08cd492c349bb9bd40796a7a97fc0 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2d4a91b3905467eb86ccef265fe2fc3 : SUCCESS in 12m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcf5cc5f84184807942d48a49ba6984b : SUCCESS in 4m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d333a0146c71444b924088ba43998abe : SUCCESS in 5m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/db71616c340c442882b60b3c2039eeb2 : SUCCESS in 10m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f9808720a14245d39a0b1ddc7a7c4a29 : SUCCESS in 5m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/10c031e8e9cc4ece9295b8b22d5351f7 : SUCCESS in 8m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8c373d5cda7b45eda7e24f4938d63071 : SUCCESS in 1h 16m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2af8efbe69dc4386bf4b72902ac996ac : SUCCESS in 1h 42m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/23b2dedebe924ca19f975c868d169ee3 : SUCCESS in 32m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c738a3ca3de249aba2ef55ac11768afd : SUCCESS in 39m 20s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/50761a08a8ad4d0ca80fbdca9c58e98e : NODE_FAILURE Node request 300-0021602393 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d95fd4b4dd8451d9a37709a80562bae : SUCCESS in 30m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/816b89e9d51342f3b126b7beb011a7b7 : SUCCESS in 1h 54m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8c15c9e6e473424381c9a2e597e38032 : FAILURE in 9m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fc1f2d1f242b44e99a7e36de16d41aa1 : FAILURE in 5m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e2037dad79cf44c2b15b2617d9234105 : SUCCESS in 2h 04m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/282ce98c4c3640b8908a811eec602994 : SUCCESS in 1h 36m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/04aa117c478946ffaefff251c4609054 : NODE_FAILURE Node request 300-0021602400 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d00bf4284867bae2ecf3cadfd088b0408616fec9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 13:26:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7cf8030db11d83005bfd7f5e80a34487aaff0a38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 15:21:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01679cd0644c4c96a9bf3c439b72dc86\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ef87e97189f4267af4f81630045075e : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bf98f0205804150a9508703b458cd6b : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d631bf06510146a3936d579e62004b48 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97973c15c7604214bf64722d8e5ed6da : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6bc5808d21c48118dde266587683172 : SUCCESS in 12m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45d9812c21b8488f8b8845d335d91e73 : SUCCESS in 4m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/25c2c623a4ed4537bc3537c3934c5384 : SUCCESS in 4m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3cfcedd653c14080b8cef1d13f94053b : SUCCESS in 8m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9f4ae4cb814d4724b3f0ec6c2a372123 : SUCCESS in 1h 30m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/315399da13494b788323821c4de931d1 : SUCCESS in 1h 23m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b02b5153e71141ec9f512786a1d16c3c : SUCCESS in 32m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0ef81a756be645eeb31960d393fcf9dc : SUCCESS in 38m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c419f1aab1cdbc36c4766fd2d192cbcf5cad1ba6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 15:21:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"410a5ce2c466f9ec1eac763ee947713fc9e0ddb3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 15:22:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b04030d763e4187b23af97c13cd8598\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f63c8f213ed449d5aff340c5811ca815 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8dbc89244b3b487d8ac239f7b1b80efc : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa238045e23ecacf42d639b0550c763d0ffc2cc6","revisions":{"aa238045e23ecacf42d639b0550c763d0ffc2cc6":{"kind":"REWORK","_number":1,"created":"2023-04-07 09:05:57.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/72/879872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/879872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/879872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/879872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/879872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/879872/1"}}},"commit":{"parents":[{"commit":"f66f3cd855ab7616225cdfc7e93d993e95eb8442","subject":"Merge \"Fix full graph loadbalancer creation if jobboard is disabled\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f66f3cd855ab7616225cdfc7e93d993e95eb8442"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-03 08:16:01.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-07 09:05:50.000000000","tz":120},"subject":"Enable taskflow retry feature when waiting for compute","message":"Enable taskflow retry feature when waiting for compute\n\nThe ComputeActiveWait task may be blocked for 300 seconds (with the\ndefault settings for amp_active_retries and amp_active_wait_sec) but\nthat triggers trigger issues when persistence is enabled if this delay\nexceeds the jobboard_expiration_time value.\n\nComputeActiveWait is renamed to ComputeWait and now uses the retry\nfeature of taskflow so it would not block taskflow for a long period.\n\nStory 2008956\nTask 42584\n\nChange-Id: I174e63cfe176e1792d0bbcd348dbc9115544e297\n(cherry picked from commit a9ee09a676074eeb619a4d7c3d9912114de50e88)\n(cherry picked from commit ae3cc5b34d5ded02d1015c05a9158f394ca3ec89)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa238045e23ecacf42d639b0550c763d0ffc2cc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa238045e23ecacf42d639b0550c763d0ffc2cc6"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
