)]}'
{"id":"openstack%2Foctavia~879873","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I174e63cfe176e1792d0bbcd348dbc9115544e297","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-04-13 22:31:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I174e63cfe176e1792d0bbcd348dbc9115544e297","subject":"Enable taskflow retry feature when waiting for compute","status":"MERGED","created":"2023-04-07 09:07:24.000000000","updated":"2023-04-13 22:33:01.000000000","submitted":"2023-04-13 22:31:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879873","meta_rev_id":"b251bb67616e4611343ebaadb336f96f35961bd2","_number":879873,"virtual_id_number":879873,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-13 22:31:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-04-13 18:18:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-13 15:50:44.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-04-13 18:18:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-07 11:40:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-13 15:50:44.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-04-13 18:18:07.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"c5eec0f7aa339e133ebb704c48da1d1a8c54747a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-07 09:07:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79031537f36b51e6ab388096bf2d0325bf6edfaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 11:40:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9ae0c16c4a147c482c3a5eacbc1d4e4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7da1ab6b2c44a8c99f73863a9c15450 : SUCCESS in 9m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/930f430deab74fb3ab13f6c90a281926 : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7be2ae7de23341eca0a5ce660be0943e : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5156b74087b24b91b1c73b4dc7c00bfc : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05a2729affbc429889948e347a743e9e : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/327e14cc6e724a26b78f8d2db45d16a4 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d027285ed8b445fd9936bc0759d9347f : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/363227a73add464fae761031b9fb070e : SUCCESS in 4m 36s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0c1956bf05f44f229876d4e88d47232b : SUCCESS in 6m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d99f980b7b4149fba6960204bf3a760b : SUCCESS in 7m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fa732f09bebc410ab90c73e4e396c379 : SUCCESS in 4m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5b38dbf51bfb4271b756ded01c7edad4 : SUCCESS in 6m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ad5f1f18acd54b3c8dfdf8fe13927f8a : SUCCESS in 1h 04m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27612811cbc34d2aa8ea7c6282025217 : FAILURE in 5m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/54a23f5d6fee4f508014fff32b6e1ce9 : SUCCESS in 33m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7b22826ee8a948e9ad5a2dc6ce60f1a0 : SUCCESS in 29m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/220c2543c2c94414be7855647f3e5b6a : SUCCESS in 30m 45s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/87a8eed31f3841b4a6f590f9178ea614 : FAILURE in 12m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/342be1d39b7c4e979474fd45435367ba : SUCCESS in 2h 27m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5d433c69e3c144bb8ca9399076c5915d : FAILURE in 29m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cf335a420c42418a8e579d91677c56eb : SUCCESS in 1h 41m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1d7a965029704a2bbca497230c0bf291 : SUCCESS in 1h 49m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/92650589bfb247a89362051f2fc80317 : SUCCESS in 1h 34m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/806dd6077bc646f0a3eb8ce05f3e9281 : NODE_FAILURE Node request 300-0020926039 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed914fad5d8dd1a7192d7a5aafded6a4c72fae08","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-12 11:58:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e2e18d06f50d786123c665b6ac15aa5526a26a81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 14:19:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eb61776952d496492f199eb7bb9f430\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96114d12a31741baba27011d710b4efd : SUCCESS in 8m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/867ecb28783246b6ac54862392d8f579 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33ecb9a68d9343c39e0c5661d7279cf9 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ed3ded14f4944fdb31bbbfe1ed97225 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c815fdafe6b144d7acefa85f7666b1bb : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65667944d4b745898a3eed9718011888 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b377bfecaf274222b87f21a151a74b08 : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff5be2357ee1409ab73c54d6fcc9b739 : SUCCESS in 5m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7ada922ade0c4f79a011197e60ce6cdb : SUCCESS in 7m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f1b42fd80c654a059bcee38b65a32c36 : SUCCESS in 9m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/aeb1e8abdffe4f2ca2cf277b781d26e0 : SUCCESS in 4m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a3505b5f303c491991c81f7299f5d8ab : SUCCESS in 8m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/02fb8111f0324c7092fa5d425e980bc8 : SUCCESS in 1h 06m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/915c935091e347008decb4d022c91852 : SUCCESS in 1h 57m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3b72eb77cfc04c45ad3795cf29ca0cc2 : SUCCESS in 44m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cdd75b494d0249e8a5d71733bf00c246 : SUCCESS in 39m 24s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb11f877bd1d454da4cf26759179764d : SUCCESS in 38m 58s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9406bd7625bd4d71845337f7d270c176 : FAILURE in 12m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/98499a7262e9422bab5850a61a90ce77 : SUCCESS in 2h 08m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6b3bd2f5f0c94d46867fe70817c6860f : FAILURE in 33m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/26911875ebe049a79af80aa66846e131 : SUCCESS in 2h 13m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e16db3576cab4c14a0ad9898e7edada6 : SUCCESS in 2h 17m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a6a26e87b9374630a3aa4fbfec25731e : SUCCESS in 1h 45m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f0e16ebce6604db1b09503de3ccc9cff : NODE_FAILURE Node request 300-0020956323 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e883dfee35af24c8fbe02a4012c1706c88c6d90f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-13 15:50:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bca7c7f97c4f79b0f60cb85f1a03d9e11fb7d359","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-13 18:18:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bcffb1de8a7c5f6bb7f86c1ef6a4aaa58d2a7e87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 18:18:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fcb9b18ec1e313da205da00aa039d3044fa295c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 22:31:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61ba246883404e2f96b4d98671876dcd\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/360d0506a7a543e184e2993e5d07ff9f : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b819c9e7ed8d43c6a825195f8e26c81f : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a840a567e8b4f419e43e730d31f2e95 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3738d257c23d41c58ccd7cb84906d788 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cadacfd8f80496393d40e32f359415b : SUCCESS in 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/033efbbaa76348c296f38f6e438358cb : SUCCESS in 5m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bd80914099204145ac4892dddb27c037 : SUCCESS in 5m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1bcb001a9d784272b36ba23e665c269c : SUCCESS in 7m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/60feddcc50a64d0f8d41edfd0e971303 : SUCCESS in 1h 09m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28669564d50f4029bb8fd046e40079ad : SUCCESS in 1h 54m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d55dd8e8f00c4236983eae00231ec4c7 : SUCCESS in 37m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/efce4bae1bad41639d4418523e062b41 : SUCCESS in 35m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/444d7c15877d476884b16d9ce77b55bb : SUCCESS in 37m 04s","accounts_in_message":[],"_revision_number":1},{"id":"00504e6145763576054d41c4eada8a40caf0f64e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 22:31:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b251bb67616e4611343ebaadb336f96f35961bd2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 22:33:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48a8b06b342b48f281c3cf8709e04b6d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78f0bafcdf95427896b371e632070afd : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed0bd699efd94aba8028ed9b09ed545f : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1c40d42ddc327fc8dfdc06b48a49f640ed0e4c6","revisions":{"a1c40d42ddc327fc8dfdc06b48a49f640ed0e4c6":{"kind":"REWORK","_number":1,"created":"2023-04-07 09:07:24.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/73/879873/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/879873/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/879873/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/879873/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/879873/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/879873/1"}}},"commit":{"parents":[{"commit":"0fc4123f313acc99a5466215fedfaec7e659ddee","subject":"Merge \"Fix full graph loadbalancer creation if jobboard is disabled\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0fc4123f313acc99a5466215fedfaec7e659ddee"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-03 08:16:01.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-07 09:06:50.000000000","tz":120},"subject":"Enable taskflow retry feature when waiting for compute","message":"Enable taskflow retry feature when waiting for compute\n\nThe ComputeActiveWait task may be blocked for 300 seconds (with the\ndefault settings for amp_active_retries and amp_active_wait_sec) but\nthat triggers trigger issues when persistence is enabled if this delay\nexceeds the jobboard_expiration_time value.\n\nComputeActiveWait is renamed to ComputeWait and now uses the retry\nfeature of taskflow so it would not block taskflow for a long period.\n\nStory 2008956\nTask 42584\n\nConflicts:\n\toctavia/controller/worker/v2/flows/amphora_flows.py\n\nChange-Id: I174e63cfe176e1792d0bbcd348dbc9115544e297\n(cherry picked from commit a9ee09a676074eeb619a4d7c3d9912114de50e88)\n(cherry picked from commit ae3cc5b34d5ded02d1015c05a9158f394ca3ec89)\n(cherry picked from commit aa238045e23ecacf42d639b0550c763d0ffc2cc6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1c40d42ddc327fc8dfdc06b48a49f640ed0e4c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1c40d42ddc327fc8dfdc06b48a49f640ed0e4c6"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
