)]}'
{"id":"openstack%2Foctavia~879874","triplet_id":"openstack%2Foctavia~stable%2Fyoga~Iccc50867b28dd20ee45cad623ab19f23ff865682","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-07-03 12:03:22.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Iccc50867b28dd20ee45cad623ab19f23ff865682","subject":"Remove incorrect info message","status":"MERGED","created":"2023-04-07 09:08:41.000000000","updated":"2023-07-03 15:27:46.000000000","submitted":"2023-07-03 15:25:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879874","meta_rev_id":"739d904a8f54390cd663f8b6b47e1b3378ffd1b4","_number":879874,"virtual_id_number":879874,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-03 15:25:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-04-13 18:11:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-11 10:51:30.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-04-13 18:11:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-07 11:42:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-11 10:51:30.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-04-13 18:11:51.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-04-20 06:42:36.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"CC"}],"messages":[{"id":"140e783d213c47de2164318f5a32b4fc33b53e25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-07 09:08:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be39bd25bf530da99a6f617f213fa05409712e22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 11:42:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a25d433cc8e48efa1b8fe4163258890\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c0f4d32e80945339590c05a8be3d3b2 : SUCCESS in 9m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/837aa2368c7145a58b23a1ddb84f75df : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6b43bcfb97d48ccb207bcbc5b49c076 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c8ac1bf4fa74d519ebfe5dfd15f5577 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/345dd16919a042fb8fcd0bff459be7e6 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4138ffddf68941f4a04f88f30e2c1470 : SUCCESS in 11m 43s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5390cbdf2a634ad3910fa78ce4fd14e8 : SUCCESS in 5m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8b71bc34eeba46978c07930f4342a798 : SUCCESS in 6m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fcd6a4c2fa4c4f7fa5fbc60b9f327145 : SUCCESS in 4m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fdaaab37802f45a4a3cd5933d1eafd17 : SUCCESS in 7m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2e61ac0839744558b8446c37cfdc9dff : SUCCESS in 1h 26m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75980072b87b445785b615602bca72b6 : SUCCESS in 1h 45m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/60a6c18b828844d8902f6350df2415c9 : SUCCESS in 36m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ad142809be8d419d9371258fe5d5ea48 : SUCCESS in 35m 00s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8d8f1949ad71457dac1f4a887827b17a : FAILURE in 29m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8dbfcc849ac4425699b83d00101de458 : SUCCESS in 31m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7e532112ab9c4602b46e055e4f1ab558 : SUCCESS in 1h 40m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f7210d931f74493c82e6e008ead1247c : SUCCESS in 2h 18m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c5ee81ba70834986a3375e7d4f9a2feb : SUCCESS in 2h 20m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/32d96c0f75fe49c9a7b3558dbceced4b : SUCCESS in 2h 31m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1101c9be4f9a4c76bf437300a0b95a13 : SUCCESS in 2h 10m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/defd7d6936a84d9e9011a9806c43c079 : FAILURE in 41m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e4228f189012415b41146a0b0ab849076a9bc8aa","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-11 10:51:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"087c0733cb84a6b8d077df6985388e80ef7b1273","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-13 18:11:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f719a5e8d9629b83b42b93eb06629433f59672c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 18:12:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"79ecb137dda8f915c5b36caff89f09113d3f0f2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 20:18:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/743d446de74e446e8efc2606f53bf421\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03f6868b180d4176b18d1c83ba12db05 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/018519061d544670bcaf999e9142281c : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a25e1eefab974d12afb1fde579cf16b2 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d0255b9d2644be1bc952f62628f2ae7 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/728fc6a6361443399ee9ee8b27b0c033 : SUCCESS in 11m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a4dee8012d1c4bd0ae659c2de554a2a3 : SUCCESS in 4m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a4a9d8d3cd7d42b785fe610572a6b9fc : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/30052c040a1b4bcfb928ed5804221ece : CANCELED in 13m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c6d122695a14fb4a0d4e049ff783bbe : CANCELED in 17m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/187280d0bcdf4e599551228e66113e4b : CANCELED in 17m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e3cffa6995c5485bba850f5e5589c0f2 : FAILURE in 7m 36s","accounts_in_message":[],"_revision_number":1},{"id":"94791b04b499bf5b8ac409d7f56de86c1da9b7b5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-14 05:35:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eab71c85ddb853acaf97571be81cd9358bf1077d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 18:14:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f1f0646028b4dc186f56bb59f41a0a1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee3b26e5248d45f59986634c27808ea8 : SUCCESS in 9m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98a546cf01514a3381415ec5d9869f7c : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dce0273d9c043a0ab93be482680f65c : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d8159e0660b426e952167eeada91fd7 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5329ad052fb4a02ad5095686201b773 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85102c61b8dc425baf816feb98d56642 : SUCCESS in 12m 23s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c69ef67cdec24f99b384ea6317bfe5c0 : SUCCESS in 4m 50s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/72059bd99345415986b1d207cfd4634f : SUCCESS in 5m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9d8ba5f84e0f48e7839b5b6d37e48aec : SUCCESS in 4m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/65b721abca5343a7b92d06f3c0759840 : SUCCESS in 10m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6c9aeb10be074cad8221fd3b26782484 : SUCCESS in 1h 43m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/309007c1bdda4635be0d8df5454c16c4 : FAILURE in 8m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/94fc14d9b05c450b9063d886f0243caa : SUCCESS in 39m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dab31c712d6d4fc29a7f5a0a95d0c7af : SUCCESS in 41m 31s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/eb442f2c5b3d446589a8f3f8968f968f : SUCCESS in 1h 01m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ed1488cfaed04d6e9fbe4b5a4d4de5f5 : SUCCESS in 38m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d1d9800bd37d416eb4d036243e00c9af : SUCCESS in 1h 47m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/31f0dc7aa6c646f598a12e84d3b3882c : SUCCESS in 2h 17m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5ab6b3ff18484b2fae588e083ab1f117 : SUCCESS in 1h 57m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/720e5d46be9f42c38a8ef3cf6b1abc1f : SUCCESS in 2h 17m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/146d88328d884005ad254aa0416a0508 : FAILURE in 7m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/633efbe6b36d4765aebcb82db3267394 : NODE_FAILURE Node request 300-0020974851 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"438039b53d4c43561e09b3b34af0f4712d7d3211","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-04-20 06:42:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3431cf5e4d67896066b91748c4aae91c1e3707be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 09:46:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1001cb2279064863b63d38066991f158\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8862326f7a84bcab3977f9f11aeaa41 : SUCCESS in 11m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0442b8d1ba4d4840bb6f057645cd20d1 : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c22eb949cb134cb59ca688670418e1b5 : FAILURE in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c82192e804748bd81a392ecbdd80c28 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/813435801e984ab5b610fb9b901fac90 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f593a574a6c54f6e831052ecbe354230 : SUCCESS in 11m 56s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9a54230e969147ab9095abd8f78cd4fd : SUCCESS in 5m 06s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/83ca9693b1574009aaa053ed9be85d0e : SUCCESS in 5m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/59e8ba5431024e73a15a963f77d96486 : SUCCESS in 5m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/db920b2d76474152900d61d8e7d03fe5 : SUCCESS in 8m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/503f30880fe941158b656de883bb8bcc : SUCCESS in 1h 25m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c7affba0cf284386b46e81b96c4e8a2f : SUCCESS in 1h 48m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ca47752e849b4d5a95dca6872c5f36fc : FAILURE in 32m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2007377376134785b090bd3b6b49d525 : FAILURE in 7m 41s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e1e6c03ed8984724ad16933370c27f77 : SUCCESS in 1h 08m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c231a6dc2e9c4073b8368a3537c7360f : SUCCESS in 34m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/287d14eacb784f9892381546578656bf : SUCCESS in 2h 05m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/507f94955d22406c9a33bd3d5a3643ad : SUCCESS in 2h 45m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7cc86fdb82e6472d8fa71c8b14c61ca0 : FAILURE in 18m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f88c51b6fcb94ee88965ae758f4bd034 : SUCCESS in 1h 37m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a7defa97c9404313b054265b56afdcee : FAILURE in 7m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/5f69072b697f4b7eb41b84de869a7c39 : FAILURE in 41m 24s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/hacking/checks.py","accounts_in_message":[],"_revision_number":1},{"id":"997eb74d9c9a02a16ec6eae86c079f3bdf1446e0","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-20 16:13:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ffe4224acc489ae547d213cd3edc641e7070c41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 18:25:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e182373c79b4044ab82d81b419d9a89\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74ed75bdcc44482fa6904227434d65c9 : SUCCESS in 9m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f3bc2a90c3c49b09f8d4629cffc45f0 : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f50aea4eccf743c5a71e83e212cf60a1 : FAILURE in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d88afcffbeb94e0baf182b56dc4a5466 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3648cdda2913457b94a8393085d7de26 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/436df24b84aa4ee3a139b4e776f191b2 : SUCCESS in 12m 30s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/16f98b97f58d4e489d72ff08a442d693 : SUCCESS in 4m 57s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3bce4ee66d1149589a78d3bab7c0e298 : SUCCESS in 6m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/04edc4f602dd41eebe88a1925f937dfa : SUCCESS in 5m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e74e4383ee674307909ed4a034d4bdf3 : SUCCESS in 8m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/df9ae3d7ed334f9aaea4fb3288d0a5ce : SUCCESS in 1h 19m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a68856cff19349d79f18fa1a379c8f18 : SUCCESS in 1h 52m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fb2ad721c4a04a6fbf86cd2a71cdfb91 : SUCCESS in 37m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8adeeb0082274084a30b0e50761a0073 : SUCCESS in 34m 59s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2d4e275157884b08ac2e82d45b17e42c : SUCCESS in 1h 06m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c1917c73c10d4e96a287636e97eac674 : SUCCESS in 30m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5ca20eff775c4dcfa2ab7cf7b745d37e : SUCCESS in 1h 45m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/46c63988e6014d6d9a919b30fa5829fa : SUCCESS in 2h 10m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/430c2e791b8d4bed9f24a14e4c96819d : SUCCESS in 1h 48m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3fa8c87123284180a840c6d33fe375ad : SUCCESS in 1h 46m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1d8e0683e00748a182f741a77b31de5d : SUCCESS in 1h 46m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6a6f563ee81477198f95ef64519c638 : FAILURE in 42m 02s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/hacking/checks.py","accounts_in_message":[],"_revision_number":1},{"id":"1bae7667cef19a0d7258fb24ae26ec5702b25448","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-21 08:30:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a36dd84e6137b84d22e4eb8ad66db6a6c9703557","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 10:52:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/17c2d463568d4919817cfc89290be8dc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b77d38bbcd7141729a650745caa6f6e9 : SUCCESS in 10m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48db6ad231514c548c13c8f9a7bf515f : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b85f393db29645e6a1be05456129d9b4 : FAILURE in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76c34caddba64469ba0b7cbc9ad76b10 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/067a478c90c145e9ad95252f396396e2 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/780bb2981a054a779ebe413f311c4a35 : SUCCESS in 11m 08s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4b984972ebd148398e28977736bcc6de : SUCCESS in 4m 51s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3fac85135e4a45b9b26620697ef18bc4 : SUCCESS in 4m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d606d9f4a0af4ea1aa2129cd70a6bbcb : SUCCESS in 4m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/af34d99c4f0f4d2991ec8db1a90a14b2 : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d6362e62a4f244d7ba3440a9aa2fdefd : SUCCESS in 1h 35m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/abf5ea95f06745d38d696a22cc48b62c : FAILURE in 6m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bc6293fb61794ea9bcc137818d457016 : FAILURE in 6m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fd3fca42e7c54acd880759800a073da7 : SUCCESS in 44m 30s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/acdf46c8229043f3a11df3ff7554adae : SUCCESS in 52m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b118275f2fe44c56b4647036ad85f500 : SUCCESS in 32m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d04faf26f92c4fc195c70c91c38b3d5b : SUCCESS in 1h 53m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4a017f7557224d5d9b241bb993234fc2 : SUCCESS in 2h 15m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8571a6ec5ae74e448a2e706443f5d1bf : FAILURE in 36m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/89c1ac8ca2b64896af45dbc995db8279 : FAILURE in 6m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2c69c198882e446c838ce147855224df : SUCCESS in 2h 08m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/07a145eee04a408da66662dce4caa31a : FAILURE in 41m 45s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/hacking/checks.py","accounts_in_message":[],"_revision_number":1},{"id":"d22bdee26d3649daaec23dcdab8e2bdf7a7c0cb7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-05-05 06:09:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c4be2dfcd9af3f28e30914f2a7b02f127ca3ff26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 09:19:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/63d29642a80345ae935dd2cf7023e35c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5de9c262cb64c81a3fe25baca2c7e0b : SUCCESS in 8m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ca504b8e5594e129e13516af45b82bf : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c739cd3f500348739b00b822465cac74 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6639c95625fb404eb003913f3b91afc4 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5572f65c07ca416cb948751eb9b12f41 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/314d3ec7138546dc9f76872af6084fe5 : SUCCESS in 9m 43s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/748af77688744d8f8c0ba5e485774cfd : SUCCESS in 5m 09s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fe310408ec93443fbefc5bb09aa86f04 : SUCCESS in 5m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/82ee294c5abb4acb810931cf87bf07df : SUCCESS in 4m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ffab3892d8554e7385192855e82253c3 : SUCCESS in 8m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/77db0d28831048db8b924c79ab572044 : SUCCESS in 1h 31m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c73ee6d468ef4b03a6b69340530fdef8 : TIMED_OUT in 2h 33m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/51dee07902ec4737828811ed740fb212 : SUCCESS in 40m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7d0d854a0a884261be775a9be6d82cf1 : SUCCESS in 48m 36s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6c26e5ee54894ff59e5ce1590e7c2ce2 : SUCCESS in 57m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c52fe075a6ea4c4f95aac93b93466dd6 : SUCCESS in 43m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d1c925761c0242fb8feb26c4f28df1c9 : SUCCESS in 2h 20m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4537050547c34ff584e2c7718b258525 : SUCCESS in 2h 53m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/885dd8369759429fb29ff608cdb744f5 : SUCCESS in 2h 28m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/06f19ebe2d5f434695d158cf625f304d : SUCCESS in 2h 32m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d9081f51e803445f9c1baa9f94f38077 : SUCCESS in 2h 06m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a4c798b612249938c68ab0a070a4157 : NODE_FAILURE Node request 300-0021143227 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b3f6bfae25113d13f6502c996d5fd7424d1608a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-08 07:24:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f6b7332f716236f143e6c3549936ca405e2c4c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 08:36:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d96d4d5de4d44fa1a355f1fe445619e5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf4bbb8f0b3b4894a421fc8f2f50ff10 : SUCCESS in 9m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9d7b17eea4c4f6abe4117bdfa026dbc : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fb2a33db9754030b41edfd06830b457 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a74fd0e115664b48bea7a9b3c2044b2d : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8103b2b3ab6408bb66dacc386a8d71a : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b9eb4158b0a45db89fd7a978d74ddf7 : SUCCESS in 13m 09s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d61785777a374484a6601a62524cb393 : SUCCESS in 4m 32s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/88ce7321f9cc4cada8b218d6b34cc6f5 : SUCCESS in 6m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/85d56c939b564dceba87c9203dbd7d88 : SUCCESS in 4m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b53b7bf4b991464cba2d60b5e1056827 : SUCCESS in 8m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/82fe18e1dc0c4fad80b326f70c01e062 : SUCCESS in 59m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e197c731067440fa3d7ea96b16bfcfc : FAILURE in 7m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6a425bdfb532476e8d06142e11fc32c6 : FAILURE in 5m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f59f1acc7e6a4d549ca7ce95d721204b : SUCCESS in 48m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2e865b7c3ea54dd1a7db427533097ef2 : SUCCESS in 1h 07m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f8e70e4c350c4558a21296488bd110c9 : FAILURE in 7m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6f895592f5d24d3a930b9d3aa9ee0b2e : FAILURE in 7m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/84cbf74b5e48419d91a77e70349c8a53 : FAILURE in 11m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8b6a6033a0c94b399495c1eff8106d83 : FAILURE in 6m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4176eded7efe4b60ab3580edffe9b6f7 : FAILURE in 6m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a65896cab9db4e8a98dc005e26d9173d : FAILURE in 4m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/2ca9d1b2cf5246ee9bdc5eb76e0f0a44 : FAILURE in 29m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"225e9f93e6c8cf3f11225d575768f1b62e302c2f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-08 08:50:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9800e467757df5e8c1f8ef8ba1865da5929d9151","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-08 08:52:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1829849bfb50b70d7a4fe668af0cfbee8b77a35b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 10:54:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb26bc8e55bc4dc3bb060b77e4336d0f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5d8c3a0f32b4d3cab0e58947ddea59b : SUCCESS in 10m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23bc1814572549b4a5400b6ddb8194a1 : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6732bcd85fc4778914b7206b62985f0 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fc1487997154ae4ab5b34373c7ea046 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23b807770cb44e14b65448d46a09214f : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b029d38779b148cfba01597dc77961a7 : SUCCESS in 13m 21s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/387a0d88901645b78ca5fb48b88bc358 : SUCCESS in 7m 30s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/892ad32790a640fc853ee055d4ccb561 : SUCCESS in 5m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d0886c26f15440978c36353d27fe2962 : SUCCESS in 6m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/94d5fa73edda4c289fefa2f0a5e0d7a9 : SUCCESS in 7m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/385af54af203425b919b12de2f8763e2 : SUCCESS in 1h 56m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b063e315e43744309db823ba0fb3c8ac : FAILURE in 7m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/270c505278804ea385ede8a8d58d3102 : FAILURE in 5m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/757aba53750a4fdb9a7555d054ac90f6 : SUCCESS in 46m 46s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f0cc4302728b48f48c12eb552ed21968 : SUCCESS in 1h 08m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21f9513d93fe407996f602982eb0e628 : FAILURE in 7m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/73e591192e6c44aaac58210c1edbc036 : FAILURE in 5m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e5cf419500a04605a02dbac4edf0c0f3 : FAILURE in 9m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b24c0793d7e7442d99d25d88bed9c757 : FAILURE in 7m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fecec45bab9a4e5fbb19ce6f31b27150 : FAILURE in 5m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e79e2e3148634248b3799fa307a4dd5a : FAILURE in 5m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/7d77145004f04835b73aa0e2ec330544 : FAILURE in 36m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e26d9a65be3d6a58ce22897b327028f70d531d79","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-08 16:21:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2fd771fade026c3faaec5e3bd024ae78740713de","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-03 12:03:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1b81fcd40f2677eab2b05305d1ab710a346be421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 14:00:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdf569662a4b4b85b09683695e74cb6c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2597e5c684fd40778a0d3e821b680b6b : SUCCESS in 10m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a39e6e9e3f3b4d06906c2209bfa0c221 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6d123dff1844e458472ff0e98e239d6 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce214d81fbf342a9bd5d5dfa458515c8 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a691737291684799ae1c9213da131443 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67535feb45864b688524a68eb8edb77f : SUCCESS in 12m 33s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c7bb30aaec7f4198bcbec8ee03a10b9c : SUCCESS in 5m 30s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/04a0dd1d3fea4fcfb3a3c137b433410f : SUCCESS in 5m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ab922289bcdd443f8dc51485ca911ded : SUCCESS in 4m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d7c1ecd55a3f4e2aa070c93dc8aa9923 : SUCCESS in 9m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/94ddbb90d5c3490e8c75239fd47e0025 : SUCCESS in 1h 34m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1fd53cc7d7094ab4b4886a51f469f7df : SUCCESS in 1h 44m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/057d494b8471456e984837e9e8518443 : SUCCESS in 34m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c5235ae051ba4ee88b522737fa975a94 : SUCCESS in 40m 18s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/193de15da5c14abd807e8974abdd3f6f : FAILURE in 12m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2a2375d515f14583a2acd0e59eb05b82 : SUCCESS in 30m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7454edf5bbbf4cbbbeaab0de8f21e508 : SUCCESS in 1h 53m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ceeb28d8194d4eb5baafced146901f67 : FAILURE in 10m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8c69cf847259420ca61c00843dd7e683 : FAILURE in 7m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e4c0ca951a70472d8a722551770f35fc : SUCCESS in 1h 55m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3cba92d0323c41dbb960b10229a9f8ac : SUCCESS in 1h 34m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c21fd4685804db4b325bed0a86b6f4f : FAILURE in 40m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9ca0d1a32c5f55583aaa6d4990314b0c4b0e274f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 14:00:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"549b78829a9e900452e55a300d41d61dec5b52a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 15:25:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/363c0858beea4d2dbe059226b46d6d5f\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb7c11feebe441139500f37a67df4ba0 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f3077eee3444462aa4cc036de664ac7 : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d90470cfb45041748933731fb3f9c2fa : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d34451be830e4bdc92dcb4d92cbc40c8 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1de92c21624a423ea1039dbdb21082b8 : SUCCESS in 11m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/75e65e969ea64ab5b9cab308374703c1 : SUCCESS in 5m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c0a3bbd46ac84fa781997c1426691f69 : SUCCESS in 7m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bf93f01e5d8243fa860e1e8b56c0fc30 : SUCCESS in 1h 18m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/df46bbc0119649bc8c2f881f9c515acb : SUCCESS in 1h 24m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0f38ee437e3c484b87229244a47cee30 : SUCCESS in 34m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/038a139ac5c240a294f250e1e6393e65 : SUCCESS in 36m 05s","accounts_in_message":[],"_revision_number":1},{"id":"53fa8d76fc95dad6bf751acb7e43cc2d88c2669b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 15:25:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"739d904a8f54390cd663f8b6b47e1b3378ffd1b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 15:27:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5121eda71e8a4284aed01b592f07eb78\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae8bf6b3f2774604914856f5b199a038 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4cfb3e5cb2b5953ef7c45eb0fe4a5680fbb9d63","revisions":{"e4cfb3e5cb2b5953ef7c45eb0fe4a5680fbb9d63":{"kind":"REWORK","_number":1,"created":"2023-04-07 09:08:41.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/74/879874/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/74/879874/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/879874/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/879874/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/879874/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/74/879874/1"}}},"commit":{"parents":[{"commit":"0c8c5037ff63f465588df16f0cb11f8ff2f03f49","subject":"Fix octavia to accept [ipv6]:port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c8c5037ff63f465588df16f0cb11f8ff2f03f49"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-01-31 19:22:43.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-07 09:08:24.000000000","tz":120},"subject":"Remove incorrect info message","message":"Remove incorrect info message\n\nIn amphorav2 the controller now waits for the completion of the job to\nsend a keepalive to taskflow.  This patch is a follow up patch to\nI6619ba117f7051fe44086389789bc6e2810fd23d, it removes a INFO message\nthat no longer makes sense because it is displayed after the\ncompletion of the flow.\n\nChange-Id: Iccc50867b28dd20ee45cad623ab19f23ff865682\n(cherry picked from commit 24ce12a15669a5ebe997655a4b42ba00142dc687)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e4cfb3e5cb2b5953ef7c45eb0fe4a5680fbb9d63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e4cfb3e5cb2b5953ef7c45eb0fe4a5680fbb9d63"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
