)]}'
{"id":"openstack%2Foctavia~879877","triplet_id":"openstack%2Foctavia~stable%2Fxena~I3859c7fcefc89e91fdaecd139143e6a74d8b3c1b","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-04-13 22:22:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3859c7fcefc89e91fdaecd139143e6a74d8b3c1b","subject":"Fix serialization of lists in data_models.to_dict","status":"MERGED","created":"2023-04-07 09:15:21.000000000","updated":"2023-04-13 22:24:13.000000000","submitted":"2023-04-13 22:22:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879877","meta_rev_id":"ca44e67f9ffb54dbf131f87020ad811e1c6f7da8","_number":879877,"virtual_id_number":879877,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-13 22:22:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-04-13 18:13:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-11 10:50:25.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-04-13 18:13:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-07 12:08:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-11 10:50:25.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-04-13 18:13:39.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"768049226ba5ef61fd00dd439222b4264fe43f0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-07 09:15:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e99fa0c407d5f705bde4f1ea44dee7511bee1a19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 12:08:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dc76c4b5157401eb5e422fa54b35012\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16579f983bc143a086212f79cdd03506 : SUCCESS in 9m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1176755bb338471cb57695e025242a62 : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d764e24a59d4d5ab706e623e7bd3256 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f23a3be14f80454089931b479853c371 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08a17f1d669342aaa014d930cc7a122b : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/659fe4fddc324492952b71c8a89991ef : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e7fc95877d9443dbe66572667bb6ba4 : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8895548e80b6404ba395c7d3bc897c06 : SUCCESS in 5m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1cf72cb19087488386d8e439ea8d67a0 : SUCCESS in 4m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6daff6b44b8942a3932d601dcd87bb9e : SUCCESS in 9m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1e13f863d74a489eac364ca656623279 : SUCCESS in 4m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/69f4373abd104be38af227f66605854d : SUCCESS in 8m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/05024250a5974f79980e151997d5be82 : SUCCESS in 1h 22m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a8279b40dae40df811611a1ae7cbdd0 : SUCCESS in 2h 25m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7b922cd993194385b1c9919a3725adb4 : SUCCESS in 36m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/53e96b0ec9414397880c510cae1eeba0 : SUCCESS in 54m 31s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/61c6b5c789a941b0a3cda5581adec6fe : FAILURE in 9m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/90b29c194eab45e3a81d811123778119 : SUCCESS in 38m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6ce02e0e82cf41888b38e61c0a5b270f : SUCCESS in 2h 15m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/edb8315afb5b44a3966c65b6e2ad1278 : SUCCESS in 2h 49m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5bbeea2a37a14eed91e04ae55914404e : SUCCESS in 2h 09m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/38e040804c3a458e8f012ac3a4dd00b5 : SUCCESS in 2h 07m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c6c96fd4a1cd48f783645f2940a98259 : SUCCESS in 1h 56m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/db3e75bab7c54398947194386b1d4f80 : NODE_FAILURE Node request 300-0020926152 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d80e6ec326c713ddaca877c51d1861ecf3ecb1d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-11 10:50:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"af6818f964eba68c3bd247e82549e345f7e143c0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-13 18:13:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dfb8114890bf9bf9c04ac35d9f72f6126229c0c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 18:14:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7104c41f27edc743df6cc0d83cc554aa7aeb249b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 22:22:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4669632ce44f4507a67a87c82920df57\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70036c348afc4fbda880762dd65146ce : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbae3573644e4b00bafdec6e33469b74 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/589a0be4c91e4a1890ab381cfb594ffa : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e98421a3cd0423ebee1e9474ac4e1a8 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e44d1c3817dd4b6e8d838c4ae1fd018f : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ad056f63e1d41c1b5d8301fcc30e9e5 : SUCCESS in 4m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/35c823098d6c403d9067ebf6f1b357ff : SUCCESS in 4m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/244d444c24604e30960305edb8f4f6db : SUCCESS in 7m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a6788dac6017435eb73fd6cdb1e49f21 : SUCCESS in 1h 17m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e16da2934b754cacb94ee1c226f30241 : SUCCESS in 1h 37m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/02b268c8df7940ddad4990ef7b34452a : SUCCESS in 38m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/29e9b52b8df7488dac7658f06b5d8319 : SUCCESS in 45m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a2038b0f435b0d3ca1b85eae1041766399f094ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 22:22:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ca44e67f9ffb54dbf131f87020ad811e1c6f7da8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 22:24:13.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac661d91630c4ed5979e6ddce3e5864d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82faf4742c174e26806d9f7225b10920 : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/deddd89bd01b48efa3daf9b25e58e833 : FAILURE in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a799b443a54440d0cff7853d2209a4576388b07","revisions":{"4a799b443a54440d0cff7853d2209a4576388b07":{"kind":"REWORK","_number":1,"created":"2023-04-07 09:15:21.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/77/879877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/879877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/879877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/879877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/879877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/879877/1"}}},"commit":{"parents":[{"commit":"f66f3cd855ab7616225cdfc7e93d993e95eb8442","subject":"Merge \"Fix full graph loadbalancer creation if jobboard is disabled\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f66f3cd855ab7616225cdfc7e93d993e95eb8442"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-05 16:08:19.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-07 09:15:16.000000000","tz":120},"subject":"Fix serialization of lists in data_models.to_dict","message":"Fix serialization of lists in data_models.to_dict\n\ndata_models.to_dict didn\u0027t convert values inside lists.\nWhen serializing TLSContainer, the intermediates list that contains\nbytes types wasn\u0027t converted properly.\n\nto_dict now converts the items of a list recursively.\n\nAdd a test that checks if TLSContainer.to_dict is serializable.\n\nStory 2009310\nTask 43699\n\nChange-Id: I3859c7fcefc89e91fdaecd139143e6a74d8b3c1b\n(cherry picked from commit d25ed7aba6753285671e4f3b8c478c0e5e022e10)\n(cherry picked from commit db22c0b9da5bd167e97ba021c22e9cf46488ea8d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a799b443a54440d0cff7853d2209a4576388b07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4a799b443a54440d0cff7853d2209a4576388b07"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
