)]}'
{"id":"openstack%2Foctavia~879879","triplet_id":"openstack%2Foctavia~stable%2Fyoga~I29fcad9e121a30d6e8f8178f2f078cf10771a32a","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-07-03 11:20:20.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I29fcad9e121a30d6e8f8178f2f078cf10771a32a","subject":"Prevent incorrect reschedule of resumed tasks with jobboard","status":"MERGED","created":"2023-04-07 09:25:39.000000000","updated":"2023-07-03 15:23:07.000000000","submitted":"2023-07-03 15:20:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879879","meta_rev_id":"d46e11b6dd8373ac9e92dbbab1eafe00bcee398d","_number":879879,"virtual_id_number":879879,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-03 15:20:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-13 18:20:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-04-13 15:57:16.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-13 18:20:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-07 11:54:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-13 15:57:16.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-04-13 18:20:04.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"14bd8aac588ceb98954358dff91ca5eb1f356027","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-07 09:25:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9733615cdeed214e59de1f53764c8be72f764ca0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 11:54:28.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/db42d7308c124f1dadf54cf8e3761002\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2249bff6fbb8484092f770dcf5e56251 : SUCCESS in 9m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f17a9e9bc814cc6bbfdb6b986b6fd81 : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b50c465333214309a730371d363a9cce : FAILURE in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59230a4141c0446db2656d11833a39cb : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5334584d094d49f093e5bace12f5fc7d : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20454a3892684756b550eea9e5152ce5 : SUCCESS in 12m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a65d40ef7d5d4463a4ff2afd726e64e1 : SUCCESS in 4m 41s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8c4cf4b27ed94b9e8aa18eaa97896a5d : SUCCESS in 5m 09s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1bd8b286809d42b6ac98de82a02a0568 : SUCCESS in 4m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/965a59eb889b43bdac52d7a0fb48e711 : SUCCESS in 5m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/767fefea0ebd4dd0bece05ca01ec6a58 : SUCCESS in 8m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a910594a09a748abbd56a32a425ee9d6 : SUCCESS in 1h 19m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a3034d4842347dcb93e5b141197ea75 : SUCCESS in 2h 11m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f397770bae9e4da4aff92784d9742391 : SUCCESS in 39m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9a7f4a40b9b343cfa5728df7bd9f6cdd : SUCCESS in 50m 51s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/af6e9eb68da24753a3318caa59f2c64d : FAILURE in 31m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f1b425ea9bf4b0bb2977a885dd99c1e : SUCCESS in 39m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a8b9549d013048e5bf628aeba0faa804 : SUCCESS in 2h 19m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0654e3cc15f546d7b22f8bb35a7e6d3e : SUCCESS in 2h 24m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ce624ecdafba48e19eab072218ad9923 : FAILURE in 19m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/75f854a0798b45a58ac5917a5fd07268 : SUCCESS in 2h 13m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/731f420758834c9391a868e395d425a1 : SUCCESS in 2h 19m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/4d20e4eb173e4a1587d62758b575bab4 : FAILURE in 42m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a2f57d1129b6c595a941d4b8e8e62b62343a6ff8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-11 13:54:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9440e56448c6b8357b8e47970a173cca143b29e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 16:09:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28c8f6cf8f5445ed9921d9d1179bf1e4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13444a809d344103a6dddfcbedec7b45 : SUCCESS in 10m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6721fdd76174973b0897c3b6549f8da : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9f4c3e0dc7e4cce8cfd14405a220d4f : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a88a944527b45fb9f5320d1858079ec : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4080e04a26045148057e2236daa420f : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ad79420b1f84b9397e406267daf3f23 : SUCCESS in 11m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7964ba1679d4621a192a880b58423b9 : SUCCESS in 4m 33s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ade57468d91743789561ce4719ce5489 : SUCCESS in 4m 58s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8ffb0a5626dc41d0929ef174c3197250 : SUCCESS in 5m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cdb2203c4c054d82b81b1661fcc8fa21 : SUCCESS in 5m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/56eafe46f4fb4acd9154d0c5e3b81331 : SUCCESS in 7m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/345b4637ba9b43c6a388046c78c7b9ff : SUCCESS in 1h 28m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b9371f2dbd5d40b985604657210392bc : SUCCESS in 1h 41m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/828eb1be1acf455baccb19fe42cf91fd : SUCCESS in 32m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2150470c38084bb4a887f863d16f1e20 : SUCCESS in 38m 14s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/98843878d391424097d03e767e2567b3 : SUCCESS in 1h 00m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/231cbba2f8ed4ed897b4ec5ba82cacac : SUCCESS in 35m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/91cd997cb64a42308e8333db2ff1a73a : SUCCESS in 1h 48m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/994270547e494ea9b5e262ab0cf8d78f : SUCCESS in 2h 13m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f77c8c21affb4ca898a3454372e44930 : SUCCESS in 2h 00m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/72cd4c65d1f64b218eb44241ccf07d0c : SUCCESS in 2h 02m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/de008a6df6064ad6b2197adbc198d8d3 : SUCCESS in 1h 55m 41s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0eee72197a34088a9e89dba0a636160 : FAILURE in 35m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d8fc5e539a9be60135abba41113f172b3ff777a0","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-13 15:57:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"09c5e4cdbcfcd03c1e9d98d6eacb639afbd6aba3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-13 18:20:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9e8725dba36bf37d8dbe6c43d165a34317ed4800","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 18:20:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8829ae0e620e467f0148da094bc532a5160a4a9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 22:47:55.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a3e06cbfc8f4c14bc21d17d385d5d8e\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05a58c605d3c49acbe67a25a7be56eab : CANCELED in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/616f64b64e7e402f87c6384e17aa67bb : CANCELED in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bfb4aed9ef64ef3900c6f2a1125e223 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0cd31287dbad4d3f958ed623aaa1ac27 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fddb67607ceb4323afebdd8d4ad4d975 : CANCELED in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8542bc0425d241ec963980693427d95d : CANCELED in 4m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/da16687bad4a4ea89b17ab30986f24b7 : CANCELED\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c5c69deffad4490986cbc72e0fb7cb0 : CANCELED in 4m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/62d9c717f6d143aaaa21e54a46ff25a7 : CANCELED in 5m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/722e02ae2c2f4620a9bf762d79c2638f : CANCELED in 2m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/87f4793ec1e1404892379e88347b77a7 : CANCELED in 2m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bdc0dea6bbac483d9276013c12f570ab : FAILURE in 5m 56s","accounts_in_message":[],"_revision_number":2},{"id":"50dd59b4aa5b07115993e0b5dbe6cb172ec582e2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-06-22 18:32:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3cf8933a7e9d8a9a5a5d42d249bba52edf58c6e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 20:43:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f631e7779b2422280bd28b4f5871230\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/164a1952ddff4265b709f90f92c38095 : SUCCESS in 9m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0179ca447a74ff9a0d74278573f898b : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11a7a4cdac5e47719ba6f6eb66711c3c : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17e3ac26680046cc842cebf081eb84e0 : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe2d3c5edf1243a796aa39f882af2d75 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8639b50fb074b64a7b657e961be9d9c : SUCCESS in 12m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcff35e0bc524f02839e9c30e8434551 : SUCCESS in 5m 09s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/68ae7179897d4a7b85a97d3b90fbb9cf : SUCCESS in 5m 29s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f955506031a247adb56e4a0cd2625d59 : SUCCESS in 5m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a06f386144f34e69b88859e369c2e92a : SUCCESS in 4m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d573a122cb4f4f4a8d77bf764c5a58e8 : SUCCESS in 8m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d6b8acdade0a4162a027cdd1b529633b : SUCCESS in 1h 25m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/12f61996c35b4e89af7617bd929cc14e : SUCCESS in 1h 35m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7aa467e61bf549068b3e5481c373079f : SUCCESS in 42m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/57831e40ba3f4e57a4fc0d17a367e45c : SUCCESS in 43m 58s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/60b8d596604948a58b908fccdbd2cbfb : NODE_FAILURE Node request 300-0021520571 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8410e7eea3264f979a5f8010859422ff : SUCCESS in 38m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c26391937e1f4f5a99529f4d26c49366 : SUCCESS in 2h 08m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6600523065c94f7eae2892f1a373af9b : FAILURE in 14m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f5d76ef02f434104ad4a0b5a2aaafedc : FAILURE in 5m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7a4ffefe33c04b9d840848cd3fff7de4 : SUCCESS in 1h 42m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4c1d1f505b8d476e8127a1f1a54ab100 : SUCCESS in 1h 55m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/3b9baad3ea004ca48245b080fbff1da6 : NODE_FAILURE Node request 300-0021520578 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"81c03a292de876f8f2fe7ae5cdaa5ff7757d2017","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 20:44:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2e1e06d5cb76395b07b556821ab8d69c37e95fe4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 23:03:00.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a7e646d93d84dc684d9670952777ee6\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c8bef2944b941eb8e484e394dc7dc27 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/834509f4afb4495bb9a2797a0b86634c : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07001a857d4e4813935290e6412099bd : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/625897f0fa7e4a59b11e27a326d076d0 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59f2887828c1420d9b42059e6a6baa47 : SUCCESS in 10m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a79f9529831948dca793c7ab61fa4e64 : SUCCESS in 5m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d62160d3d8cb45d3b2b4890d9ec7b5b4 : SUCCESS in 5m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e6df5d831c9840a6bca0fb40d0d24e57 : SUCCESS in 9m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/54ec2b5561a143e88c24657e37733d1a : SUCCESS in 1h 08m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c5503773d15f4099991281300e909487 : FAILURE in 1h 18m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/17290c685ddb45f9a79304d0881f1fd6 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a65657b77af5443a8c1e3abd0a929653 : SUCCESS in 43m 47s","accounts_in_message":[],"_revision_number":2},{"id":"7917f96010706aa463f9cf99b263697f516c73ba","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-03 11:20:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bed068cfa5b0f68181795ffdae299a29f7767ab8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 13:19:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd63c6f3c8d94970a50bdce8a7eff296\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25c2ef94b0ab484fb8b999dc259cf6af : SUCCESS in 12m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/424e066d8d8948ed81ce21a4e7090f8a : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5aa774485390408495a157aad1092054 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af46878e48d64f85ab510f7dedd47899 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d1889328c7f45ba80759dcd149f1430 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ba03b379eb3447dbd25fa9bccdec39e : SUCCESS in 11m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de4b6225adb8419dbacab93cb99151c3 : SUCCESS in 4m 30s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/56d34f45c7c746fdafa49d9503a094b9 : SUCCESS in 7m 14s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fdf8fb1d3d6e4c689a7f4eedef04ff48 : SUCCESS in 7m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/10be0faaf741417187d6b8cb4b14adfc : SUCCESS in 4m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3ef1e290a42045799bd9bafd7a25e306 : SUCCESS in 8m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b0132ed27d8843e9984d8d3aa837c1b7 : SUCCESS in 1h 17m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a2e1004d9af849f993bb53e309ddb145 : SUCCESS in 1h 42m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/99378bd8b8ce4b7fb3f35d3eac9899fb : SUCCESS in 37m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0778edbb6cc54aac93437a998d165a97 : SUCCESS in 38m 15s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/79128bb04a364f6a95411970c2203e0f : FAILURE in 11m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2fa6e087a52e4a30b780ff88f1533c74 : SUCCESS in 33m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2f66862352b7494990c287f10adafb42 : FAILURE in 26m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/909ef82fdb1f485b8e26e67c45216021 : FAILURE in 11m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bc3d21e7d2564f83a036dab429bc9f19 : FAILURE in 6m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/06b5e15477d544bf992e1042943b5e85 : SUCCESS in 1h 50m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0fc73c34540343fd9c6d13637169618e : SUCCESS in 1h 57m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/7b635483e6224e98a719d58d239d0e53 : FAILURE in 36m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"69365b10fdacf39520271d2a7f297887e5ba4918","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 13:20:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a091ea68f8d1eb103ba44110cee57c0b4db3988f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 15:20:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eeaef776f4244eb8a175ab220211845c\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70f2014e5f2c42f9983bb5ddbb45f968 : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59bad53f21c64c189b4dcde8e8e55267 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/117cd1338258405a9e586af4cda9ac6e : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c28b48f05ae94665b6c92d9fc39bf7a2 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca731399f85b4af2bcbf0e4abe1c75f9 : SUCCESS in 12m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5c3d5ce7f374f6b89324f2b77962064 : SUCCESS in 4m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3744edfcea86455bbd5bbcafe4eff7e4 : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/88ab9d5ca9774bacaccf890806e86bce : SUCCESS in 8m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8da87c913ae04da48b3dd0d7cbd46c85 : SUCCESS in 1h 33m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/50cb30b2ecab406db249525b80bfe560 : SUCCESS in 1h 45m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fb91a17b011149ba8d6788d128681d0d : SUCCESS in 35m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/57786a3b1ac84201837065d8d631a794 : SUCCESS in 35m 54s","accounts_in_message":[],"_revision_number":2},{"id":"961c0b7665028bdf4b276578a17b694f0958afd0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 15:20:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d46e11b6dd8373ac9e92dbbab1eafe00bcee398d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 15:23:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8eb705017ef944dfab9f35792fa23db4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e1023538b9d4db88848b56226b51d2e : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f28379cdcf484845887bafd6403b3f9a : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dd29742d0e37682b9e46c48e8f63579ad7152b97","revisions":{"0d5bcb296352d37307978a8b0550a50b80402024":{"kind":"REWORK","_number":1,"created":"2023-04-07 09:25:39.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/79/879879/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/879879/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/879879/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/879879/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/879879/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/879879/1"}}},"commit":{"parents":[{"commit":"0c8c5037ff63f465588df16f0cb11f8ff2f03f49","subject":"Fix octavia to accept [ipv6]:port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c8c5037ff63f465588df16f0cb11f8ff2f03f49"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-15 12:53:05.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-07 09:22:40.000000000","tz":120},"subject":"Prevent incorrect reschedule of resumed tasks with jobboard","message":"Prevent incorrect reschedule of resumed tasks with jobboard\n\nA previous patch [0] introduced a keepalive mechanism that allows a\ncontroller to notify taskflow that a task is being executed.\nBut it doesn\u0027t handle \"resumed\" tasks (tasks that have been rescheduled\nfrom one controller to another controller).\n\nThis patch moves this keepalive task in a thread that is attached to a\ntaskflow conductor, the thread will extend the expiration of the tasks\nthat are executed by this conductor.\n\n[0] I6619ba117f7051fe44086389789bc6e2810fd23d\n\nStory 2009998\nTask 45089\n\nConflicts:\n\toctavia/tests/unit/common/test_base_taskflow.py\n\nChange-Id: I29fcad9e121a30d6e8f8178f2f078cf10771a32a\n(cherry picked from commit d367b47639af02aaaff80cd62629f5c8e7f3f91a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d5bcb296352d37307978a8b0550a50b80402024"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0d5bcb296352d37307978a8b0550a50b80402024"}]},"branch":"refs/heads/stable/yoga"},"dd29742d0e37682b9e46c48e8f63579ad7152b97":{"kind":"REWORK","_number":2,"created":"2023-04-11 13:54:24.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/79/879879/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/879879/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/879879/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/879879/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/879879/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/879879/2"}}},"commit":{"parents":[{"commit":"0c8c5037ff63f465588df16f0cb11f8ff2f03f49","subject":"Fix octavia to accept [ipv6]:port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0c8c5037ff63f465588df16f0cb11f8ff2f03f49"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-04-15 12:53:05.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-11 13:54:08.000000000","tz":120},"subject":"Prevent incorrect reschedule of resumed tasks with jobboard","message":"Prevent incorrect reschedule of resumed tasks with jobboard\n\nA previous patch [0] introduced a keepalive mechanism that allows a\ncontroller to notify taskflow that a task is being executed.\nBut it doesn\u0027t handle \"resumed\" tasks (tasks that have been rescheduled\nfrom one controller to another controller).\n\nThis patch moves this keepalive task in a thread that is attached to a\ntaskflow conductor, the thread will extend the expiration of the tasks\nthat are executed by this conductor.\n\n[0] I6619ba117f7051fe44086389789bc6e2810fd23d\n\nStory 2009998\nTask 45089\n\nConflicts:\n\toctavia/tests/unit/common/test_base_taskflow.py\n\nChange-Id: I29fcad9e121a30d6e8f8178f2f078cf10771a32a\n(cherry picked from commit d367b47639af02aaaff80cd62629f5c8e7f3f91a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dd29742d0e37682b9e46c48e8f63579ad7152b97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dd29742d0e37682b9e46c48e8f63579ad7152b97"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
