)]}'
{"id":"openstack%2Foctavia~879898","triplet_id":"openstack%2Foctavia~master~Iecedb9e9c8f293cd4c067900e1696f5050769ee8","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-04-18 09:34:42.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-04-13 07:12:23.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Iecedb9e9c8f293cd4c067900e1696f5050769ee8","subject":"Send IP advertisements when plugging a new member subnet","status":"MERGED","created":"2023-04-07 15:00:05.000000000","updated":"2023-04-18 09:35:50.000000000","submitted":"2023-04-18 09:34:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"879898","meta_rev_id":"19e63770936a51ab4988e0049f60eb11f4c942b8","_number":879898,"virtual_id_number":879898,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-18 09:34:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-04-17 16:41:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-13 07:12:23.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-18 07:38:00.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-13 07:12:23.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-07 17:40:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-11 17:00:20.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-04-17 16:41:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"29edc8e674836d1460e92953524cc0b89fdd27f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-07 15:00:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"732542a7b789b4e59eabacb5f957a472b18de6a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 17:40:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/12af095a7ef04d38817a72aff97dd896\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ab7157ee4e54e6fb5019ad82188da90 : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/931da7aa314f4d8abc439e49843f256e : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9e336365ab149b5b8bf263ba73daa88 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae98fc66b953422eaf2d0123a32d4ee2 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/529cd5523d884b3ba951ad956ac16783 : SUCCESS in 11m 02s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b7991e69c7f0457c8b08bbde36cd99d4 : SUCCESS in 4m 54s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/01991224c4b1411cbbea6239e610ec88 : SUCCESS in 7m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ca06b85551d140adb44851a491619882 : SUCCESS in 4m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a0ee786faef846659c36a0ec6339ac1e : SUCCESS in 7m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/827d0d4db708453faa2d859d018cab80 : SUCCESS in 7m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/530282e8d8304f59b81d59133039fbe8 : SUCCESS in 6m 58s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3263636f40df4766b74682420aa16e90 : SUCCESS in 1h 21m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4b289cb04f3b43cfaab3ae4c02b111f3 : FAILURE in 2h 11m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e37f373924f24264b35c7d68c26f6b9d : SUCCESS in 46m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0b512582e17d48888ad059580cbb8fe4 : SUCCESS in 50m 42s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/23fe80e02c5a49d884b5ee7d6368ed6a : FAILURE in 29m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1f7ef41b132645788c5c434bd2a452a2 : SUCCESS in 35m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fe8f59c8011a4778979f3524550bae16 : SUCCESS in 1h 49m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b1eb83da3959488fbca90e6185859837 : SUCCESS in 2h 25m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9003963062654db8b7df0fc5f5cbb45b : TIMED_OUT in 2h 34m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c7da4afa698e40a78016f0e54b32d7d2 : SUCCESS in 2h 08m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/75201bc5c6fc4e77b647b053fe5c9033 : FAILURE in 5m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/41eae208ddab416cb59dc2968acbc9da : FAILURE in 22m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"38f83251b796434eb97b345900ea0eb2e401c606","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-11 13:00:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d2897db0d9d6a1ff4cb40c5d665bac91271de323","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 15:48:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d9baa7cb95c442489bbca51bb47ad4d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a9dda8cb5c84870ac88d26188ee3481 : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33a634f16be846acb2ce65125d4f7866 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ae5fb3c2cfc43ffbb989a60d88be194 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f46ecab8b4504c3b8352c31b5fb86fc5 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d476dff2ba7b4f0586156cacd1e9d408 : SUCCESS in 9m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2db2441fb08347d398a71f53ec1176b9 : SUCCESS in 4m 36s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6a9e8a1d48de4ee99e6fbc3ebecfd267 : SUCCESS in 4m 54s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a240545619594440b6c76071243b89cb : SUCCESS in 6m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0100a0cbd94b4c308dc51fc7f7775b5d : SUCCESS in 4m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3c235f621b53465597100081b9abfbb1 : SUCCESS in 6m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8437df4621354389bb5764c96fa60a57 : SUCCESS in 7m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9028c1150b774fd0ac31df606e4e2e78 : SUCCESS in 7m 25s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bdf1c21c728a4478b1e7bb3553c4d4d4 : SUCCESS in 1h 12m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53b2001dd63848ffbef0f02b3346b51e : SUCCESS in 2h 26m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/86c86b705fdd4b6a9606bd71d95a3836 : SUCCESS in 39m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1288085beacd427e91790ef372fb92b0 : SUCCESS in 52m 22s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d98105b9d55943978c9bcb82ea5529a6 : SUCCESS in 1h 01m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6ac122ab643547f6b460f2848aa0f6c9 : SUCCESS in 43m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b31ed6d9e308461da8677c5081068851 : SUCCESS in 2h 21m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1543e83a0c72428a8083f8da8161069b : SUCCESS in 2h 32m 29s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b5b521a990c340ad87f59f10708ce06e : FAILURE in 2h 27m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ea83effd580e4772bdaa09f08054a9cf : TIMED_OUT in 2h 32m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/97453eb5afe04e22aceb0be25cd1f414 : SUCCESS in 2h 18m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5733ebb9057b425ca62df97bedcf5fbe : TIMED_OUT in 2h 35m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8a04d7be40dec52d00f4942607601b4481ea71be","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-11 17:00:20.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"abdcdc115fefb6e7c1197fe2c7437f47868d844f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-12 19:33:46.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d95149ad6e0d9378d8b0d70956adef7dcd494219","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 22:05:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9e6341c5669426ea8716ad205ad9481\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3b2db07783d40d596f688dd5e5851e5 : SUCCESS in 11m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aab5b4b75f8e4fc8b613e312d7e0212f : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4e9b37e4fa74e4fbea92de5917a960d : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc61636647b647f4ad177d8d2f2b16ce : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/488abad6d1344614af4a289d8939d4df : SUCCESS in 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c72d72b20d94d0db9a68f8c0fc1b05e : SUCCESS in 6m 13s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/78a8de6e6f284a38a697cfd0829e40a5 : SUCCESS in 5m 23s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/15ce97d857e94a389187ffaa68942ef2 : SUCCESS in 9m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f55510cf60524267bb9d95d26298f6cf : SUCCESS in 5m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b30a1532f599409ca40515d6a176ede8 : SUCCESS in 7m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bed48d79b4b34592a433e43f86ff7c82 : SUCCESS in 8m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d8d127d2e1d7430fa945d8f3708efc60 : SUCCESS in 8m 41s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cc5af5ee368d4c82aadd294228deb830 : SUCCESS in 1h 51m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0248f19cd06a46f6ae894d7fadcb2fd6 : SUCCESS in 2h 02m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c0581b3e203a42e2a0b3115e5d75f97c : SUCCESS in 38m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/515ed5a3ec254ef98b786d5d57475920 : SUCCESS in 42m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c80220c9efa947f89f3eb9955ae4c921 : SUCCESS in 1h 07m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67fdb15edac3403f82b8781e918fcc24 : SUCCESS in 34m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9d46e52b63b8487ca84aa056dcb4bebe : SUCCESS in 2h 15m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a6feaab0191d4aad8f4122fb8ca1ae81 : SUCCESS in 2h 28m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1bf2e4a4fa684c17b324d5f2fcbc770b : SUCCESS in 2h 06m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b50cf51a210b47169549c56317d988c7 : SUCCESS in 2h 26m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/28d27390d91243609434a43aab7c499e : SUCCESS in 2h 28m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ea05265261e49bcabac9cab6f466cf5 : SUCCESS in 2h 20m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"25e4bce846719f11ff71df1c17e00bd9740a440c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-13 06:28:45.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3818e7e10dc6baf0152a49787dfacff4b300c4a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-13 06:46:43.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"693382dcb2295b9f21bb538b3f7e80b9a9164de2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-13 06:46:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dd2109c945bd7f6dff95fa3cf16d005499a64709","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-13 07:12:23.000000000","message":"Patch Set 4: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8c01f4daf5d05df7f8dc53d22de64a5cdebb8714","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 21:37:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47b975e4bb9b4aef9ad24af22aa2e5b6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32fa9a66c4374961a7904e2c08d3bdc2 : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fcaa285fb0b407589e70c3a07b4da1a : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a71fedf15234caf8c27b2567db6e444 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d2d338e73494b349e0893d3499696d1 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab82bafdcacc410c9eee44648e768dc6 : SUCCESS in 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77dbf0e487fb496ab554b99b8330c54f : SUCCESS in 4m 56s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7d8c058dfdfd4f989de989016c5108be : SUCCESS in 4m 55s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4df9ba187b094393a18bf73256db8d0e : SUCCESS in 7m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/62c7375eb6864e5f8cf8b3bcf3ee75ce : SUCCESS in 4m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a8cd9eace99549e5b388ec3d9ef56f2b : SUCCESS in 7m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c2cebe30406642d6b4bb3b932665b8dd : SUCCESS in 8m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/536795be08e2411d83736478c712d12d : SUCCESS in 7m 57s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/36f525f2cdb74fedbe9a43e3a1b93ad3 : SUCCESS in 1h 35m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7025d8673ed14033bc46d911e2e50402 : FAILURE in 8m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b93cf5838c6440f88fe9b45c3dfd9ed8 : SUCCESS in 38m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a57131947f3e42ff940f211d475b7376 : FAILURE in 7m 30s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/cc40cc91bbc34bcaa055e93be2b519e2 : SUCCESS in 1h 04m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de6dc370b9d94dc9b83acf897a6fd899 : SUCCESS in 35m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8330024aeb484a40810302fb79b93c19 : SUCCESS in 1h 45m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d24523ec17854346b949a5b30410ce86 : SUCCESS in 2h 45m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/06e02469c01c4323ab1206f4302b51b6 : SUCCESS in 2h 28m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7ea43b09fe46480d9796cbb759650af2 : SUCCESS in 1h 43m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/43964ad91dc844678b0e82e9d7066168 : SUCCESS in 1h 43m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e263e504ef34e739be49ee477964c54 : SUCCESS in 2h 17m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c5b459771e4d7d885d6d7d05b990f205a39c9f3b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-17 15:32:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7d22a39b514902422895e7d1169a27a74d719c5e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-04-17 16:41:16.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b44aeb901547028f745783c8d006eef34cdd867a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 18:10:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96c2e8665a894152b2616c477a5a22af\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cda41b88ef3f4473accb1024102b590b : SUCCESS in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7f267aa4f954aa084089617f51a3c90 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31530fcb2fa943849d7645f9c0fa5d1b : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8eec9a6883454b87b211b6c8ecad6af0 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/197d30052ac9411796bfab79f241bf7b : SUCCESS in 11m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5de3a796343e4b298679f32edd047ccd : SUCCESS in 5m 01s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c240cd0129e1443cbe17ced5b9ddfa80 : SUCCESS in 4m 48s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5208e4dfe7604768bcc11e09bc6b348d : SUCCESS in 7m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/36ee3367804043f99c3988ee5c063353 : SUCCESS in 4m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aa11451d558b4f1f8b1de021a1ac47e0 : SUCCESS in 7m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4b2c4eb4749d484e87fc4f75b5dbdaf2 : SUCCESS in 7m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a35f690f595e41149e16f9b3e766cb66 : SUCCESS in 8m 02s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3fc305eaf83142c390ab409d561329ae : SUCCESS in 1h 37m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c13fb55e6df2494ab878df8eb9a72104 : SUCCESS in 2h 03m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fe4365013b08456f8920fe273a8c356d : SUCCESS in 33m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f0e88d71b2de4a469c69b4ca26e22229 : SUCCESS in 37m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c43326b44d1a491aa03d5de680767494 : SUCCESS in 1h 04m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b8e1b8342577481aba84e2c584d3ed24 : SUCCESS in 33m 28s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/26fb4341b1124f14a0d61dd6bc14dad8 : SUCCESS in 2h 03m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b894a5315dfb4385b25b52acecec5806 : SUCCESS in 2h 18m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f2053afa84d44dcba70b7568f0e6836d : SUCCESS in 1h 49m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4543d71dc2a344e8811cbb56d780cce0 : SUCCESS in 1h 57m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/31fe0a97af6f4ca7ae19ed58f2780d4e : SUCCESS in 1h 50m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/57ef4fddd71c4a88b31259acf06949de : SUCCESS in 2h 29m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"381c7351e0a24918b5dd386ac62399eda8ba0c3f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-18 07:38:00.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5bb0d300a6331ce1e5bf6f48bbbb72db858ab97e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 07:38:34.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fccb7fcdf97d131546a65692ca02e22f04aa9fdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 09:34:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49447ad268054451bc1e10a5b3be5d50\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17b48fadd7af431f8e9b25b6ac97ada7 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a9da99732ee4f31a53faed023b3706c : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ebb4bc7215f44b13835ce80bef2b7f8b : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0e23088ab414df8a901e8bb82037fc8 : SUCCESS in 12m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/829f002bf5cd4d0498837df9c91f8701 : SUCCESS in 5m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c54241ae90794be7812910ce4702c512 : SUCCESS in 5m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e266103c0692484c91f1fcc198e20e68 : SUCCESS in 6m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0c00481ae4e34e7491a4b43dca46a85a : SUCCESS in 1h 32m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f64c7a85cd01415cb506528ed690c63f : SUCCESS in 1h 55m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/550ed6ba0ff24029b933c35500987e13 : SUCCESS in 37m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5c08fb399a7a4e058be3e0bf665bf07f : SUCCESS in 38m 39s","accounts_in_message":[],"_revision_number":4},{"id":"a59f9421bd16e35ce29b6f667a83c8be713af552","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 09:34:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"19e63770936a51ab4988e0049f60eb11f4c942b8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 09:35:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32755c1cb4654ecca136bc849d902608\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb952fa6a7a44219b590002041efdb95 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/258330447fb64483b04d770bc3e6bf6b : SUCCESS in 39s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"18cd1f34c8a918ab6f969a3271ac74132c4ac736","revisions":{"8dd5e590b855f144bb32fdeca30b54a715701450":{"kind":"REWORK","_number":1,"created":"2023-04-07 15:00:05.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/98/879898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/98/879898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/879898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/879898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/879898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/98/879898/1"}}},"commit":{"parents":[{"commit":"cad146f1bb538b5ac442da0944a42206b69f1798","subject":"Merge \"amphorav1 removal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cad146f1bb538b5ac442da0944a42206b69f1798"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-07 12:04:45.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-07 14:57:58.000000000","tz":-240},"subject":"WIP Send GARP when plugging a new member subnet","message":"WIP Send GARP when plugging a new member subnet\n\nChange-Id: Iecedb9e9c8f293cd4c067900e1696f5050769ee8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8dd5e590b855f144bb32fdeca30b54a715701450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8dd5e590b855f144bb32fdeca30b54a715701450"}]},"branch":"refs/heads/master"},"23fdf27be9ce343f52c9749a1386c3bb47a311b4":{"kind":"REWORK","_number":2,"created":"2023-04-11 13:00:17.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/98/879898/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/98/879898/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/879898/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/879898/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/879898/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/98/879898/2"}}},"commit":{"parents":[{"commit":"cad146f1bb538b5ac442da0944a42206b69f1798","subject":"Merge \"amphorav1 removal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cad146f1bb538b5ac442da0944a42206b69f1798"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-07 12:04:45.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-11 12:57:10.000000000","tz":-240},"subject":"Send GARP when plugging a new member subnet","message":"Send GARP when plugging a new member subnet\n\nWhen plugging a new subnet, the amphora sends a GARP advertising the IP\naddress of the port. In case previously deallocated IP addresses are\nre-used, it helps the other resources on the same L2 network to flush\nthe old ARP entries.\n\nCloses-Bug: #2015572\n\nChange-Id: Iecedb9e9c8f293cd4c067900e1696f5050769ee8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23fdf27be9ce343f52c9749a1386c3bb47a311b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23fdf27be9ce343f52c9749a1386c3bb47a311b4"}]},"branch":"refs/heads/master"},"3716eba9fa0d5ee1398cfedd5fc78ecdec050c59":{"kind":"REWORK","_number":3,"created":"2023-04-12 19:33:46.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/98/879898/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/98/879898/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/879898/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/879898/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/879898/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/98/879898/3"}}},"commit":{"parents":[{"commit":"cad146f1bb538b5ac442da0944a42206b69f1798","subject":"Merge \"amphorav1 removal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cad146f1bb538b5ac442da0944a42206b69f1798"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-07 12:04:45.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-12 19:29:04.000000000","tz":-240},"subject":"Send IP advertisements when plugging a new member subnet","message":"Send IP advertisements when plugging a new member subnet\n\nWhen plugging a new subnet, the amphora sends a packet advertising the\nIP address of the port. In case previously deallocated IP addresses are\nre-used, it helps the other resources on the same L2 network to flush\nthe old ARP entries.\n\nCloses-Bug: #2015572\n\nChange-Id: Iecedb9e9c8f293cd4c067900e1696f5050769ee8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3716eba9fa0d5ee1398cfedd5fc78ecdec050c59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3716eba9fa0d5ee1398cfedd5fc78ecdec050c59"}]},"branch":"refs/heads/master"},"18cd1f34c8a918ab6f969a3271ac74132c4ac736":{"kind":"REWORK","_number":4,"created":"2023-04-13 06:46:43.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/98/879898/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/98/879898/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/879898/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/879898/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/879898/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/98/879898/4"}}},"commit":{"parents":[{"commit":"cad146f1bb538b5ac442da0944a42206b69f1798","subject":"Merge \"amphorav1 removal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cad146f1bb538b5ac442da0944a42206b69f1798"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-07 12:04:45.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-13 06:44:32.000000000","tz":-240},"subject":"Send IP advertisements when plugging a new member subnet","message":"Send IP advertisements when plugging a new member subnet\n\nWhen plugging a new subnet, the amphora sends a packet advertising the\nIP address of the port. In case previously deallocated IP addresses are\nre-used, it helps the other resources on the same L2 network to flush\nthe old ARP entries.\n\nCloses-Bug: #2015572\n\nChange-Id: Iecedb9e9c8f293cd4c067900e1696f5050769ee8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/18cd1f34c8a918ab6f969a3271ac74132c4ac736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/18cd1f34c8a918ab6f969a3271ac74132c4ac736"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
