)]}'
{"id":"openstack%2Foctavia~880435","triplet_id":"openstack%2Foctavia~stable%2Fzed~I30b81866989c22b94fb77e62e7abd180f0f0af50","project":"openstack/octavia","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-04-29 11:29:54.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I30b81866989c22b94fb77e62e7abd180f0f0af50","subject":"Pass config to castellan","status":"ABANDONED","created":"2023-04-14 05:45:34.000000000","updated":"2024-04-29 11:29:54.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"40b63e85b038286d556cd34520149fb276b9b02e","_number":880435,"virtual_id_number":880435,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-04-17 17:59:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-23 11:01:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-14 05:45:34.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2023-04-17 17:59:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-20 06:45:48.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"CC"},{"updated":"2024-01-23 11:01:34.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-04-24 16:47:30.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2024-04-29 11:29:54.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"ae2062ef9a5845ac243bb620cd07ab1a86178de0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-14 05:45:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9f32b33b7243d6de28ecf765bafde2a65c63986","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 17:59:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/febc29364e1c447294d775a922fec47c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb9b34e12a754142a8546d0cd1f35d41 : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d1adcb4064d4cd9bd3442d493dc8975 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4454123655694dc5bf6a30d86b2e9c50 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8f133e7795444f78bfbc8be4713ad83 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7738c245c1c4aa69b9ecf8b88e5bf90 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2625a9803ccf44079aec3b86d05d316b : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff57b6652cda435baaa08573a5268142 : SUCCESS in 4m 47s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c30de5cc226744adba7748f2d9778521 : SUCCESS in 4m 49s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1b423c240958496ba65398b0b2a70783 : SUCCESS in 8m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/79491ba0bcfc490cab56b9b98a61debe : SUCCESS in 5m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/97e5b0ad99e8464fb9150575b7299e43 : SUCCESS in 6m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5dd0224dd72549d8b4304aa62e1e4093 : SUCCESS in 6m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5f30513ab2f84ff3b392365be24f2ec1 : SUCCESS in 7m 54s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d0779095b5ed4c3aa9e4e46f6419e43e : SUCCESS in 1h 26m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2856169a24e840b3bda89c38f48a6ee2 : SUCCESS in 2h 27m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c9d09506cf8f470a9d4ced8fdb277b0b : SUCCESS in 33m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e8a29dd97f8f4321b8996b7721e74985 : FAILURE in 8m 35s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0a1ab0de66c84bb3a754485888f49746 : NODE_FAILURE Node request 300-0020974963 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c14743de69844232908fcef359b56da6 : SUCCESS in 29m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e0f46c8050ea4d05af498a509c6611b2 : SUCCESS in 2h 08m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/87c5162f8fc149bfbd9f9391c88ed0ea : SUCCESS in 2h 29m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/82855299a8cb4296b43d471c3bc10eaf : SUCCESS in 2h 31m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/02aabcc3a9d448c29e4b2c466da10924 : SUCCESS in 1h 40m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3e87576a6c9344ac9160549cfa711de6 : SUCCESS in 1h 42m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d802fe3e20ff4248b75c43d80ea72605 : SUCCESS in 2h 20m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3631000ff796a9443a7c6d074c6acf00714945b5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-04-20 06:45:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4952fbd48943b7be611997cdfd4de05696c50fb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 09:02:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0dc09aa7a77f4379820dcc380b1b8350\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cf033f6240f479a98249115264de135 : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4949d3e3820f4aae9b213122da5578a6 : FAILURE in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c03517de53004be48e941ec6a7a0edf2 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa60818665594a7a9a7804bfdbeb0b27 : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/73441c0a1e6f4684b801060490faec98 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e95d5627d9f4586bb4d9daf1d23ea0f : SUCCESS in 11m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f296690e6664f229b7859c965bf3cae : SUCCESS in 5m 08s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/020dcadd307a49b1ad8984c1e0f90cad : SUCCESS in 4m 58s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/37ee96cbf9cd4ee0bcdf99a24551f1c2 : SUCCESS in 7m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/41edf016d38844dbafde99a0fb828a62 : SUCCESS in 4m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cebe67aee80546e08da48267efecaaf7 : SUCCESS in 7m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1d2c7c337da740a394601b286f08c558 : SUCCESS in 6m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/be2131a0e4bb4a5f82ee65de2a825f7e : SUCCESS in 7m 53s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aba87291f6e243ba9c2fe1792a1d339f : SUCCESS in 1h 23m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/309387b40b144043b1091bb0c8949941 : FAILURE in 5m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b19f531419784680a03b6bc3df4e23c9 : FAILURE in 25m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/82ba61ad65584917984e8c5a98c2f937 : FAILURE in 5m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7afb4d4a4906493b81c063d125353266 : SUCCESS in 1h 12m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4eda643fef8f42ecaa81d2162744ab0b : FAILURE in 7m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8c523913e6b24991a15987f6dcdeacb2 : SUCCESS in 1h 58m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ca9fcf5691ec451e8768e7e3949d66d2 : SUCCESS in 2h 04m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e1d8da8658964af89c441a56f8c4eb61 : SUCCESS in 2h 00m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7ccc5e81539b4a658a9c8aee404c524d : SUCCESS in 2h 10m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/16b3c6f13eb04edfabe7d0ff255d8b94 : SUCCESS in 1h 36m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a93cdef9a7748cb87ced29ee4efb42f : SUCCESS in 2h 14m 17s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/hacking/checks.py","accounts_in_message":[],"_revision_number":1},{"id":"0842f072fc2124388b2b467e8b07e3d2f395a031","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-23 11:01:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d07ec821856e71057e27274bc579b973cdb37501","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-04-24 16:47:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"20f886b1dae38e91cdba2d5d00e1a347e2f12f51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 19:40:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f857c120e31c4a3f81fafd6579aea46b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba6ab871dac44f7c9e0c776f2ba1da42 : SUCCESS in 10m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46be4c9425554bac87f82971908b5170 : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16999bb6c32c42b5a7f6c088e36937e1 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/daa5eddb247549a095086b5f1d4f5845 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e58018e712e844b3b256f13a4026e5e9 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e745e8b89b1c44a4997eeec1f3b41f3b : SUCCESS in 13m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4b367662e19405487a0453c597c9714 : SUCCESS in 4m 40s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/bb2101dec71a4375b57f083edeaf38bc : SUCCESS in 6m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fb80793b9fcd414bb966eadfecb2680e : SUCCESS in 6m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/88cc2bc355d648f3a67fa4ce5080ac7a : SUCCESS in 5m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7ad86654698d4b84af59b7672cb1a00a : SUCCESS in 7m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/573e58a1694a4c8d84b9a52ce9414107 : SUCCESS in 7m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c2b5ce89a06a4c80bf794c9331d6c11a : SUCCESS in 7m 33s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3da0a971b5104f9e82c3d71c6fd5f4c9 : SUCCESS in 1h 37m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c215788b73d64060bbba909f910549c5 : TIMED_OUT in 2h 32m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/170a4e0acff14354ac6aeff7472e7ca0 : SUCCESS in 47m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1abdb3b4cb7a45fda61d152c467ec6ac : SUCCESS in 42m 45s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3d510fee00d94996a55318a9faa720a5 : SUCCESS in 1h 00m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f2218b905104440bb01e7fccc026efb8 : SUCCESS in 38m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/94739df04b784e8195cc7a128916719b : SUCCESS in 1h 47m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b6f8c71bcdea4e219b01f560d9023d64 : SUCCESS in 2h 11m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8bd9d67f3e904037980eb98337ed5899 : FAILURE in 40m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5bad7ce341314a85897b1e0ecb987de8 : SUCCESS in 2h 18m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9291ee5218d548ca9babb7b678473da8 : SUCCESS in 1h 48m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/43b92babe6504129a78440fb1c32ea30 : TIMED_OUT in 2h 34m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7c7a39bfc41d29fd04efd454b67a4e43edf5f9b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-25 19:00:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ebe63d727d5d316f4731caa28907a3508aab1f89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 21:36:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac30caaa38844edc8af7f92b95a02d59\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd5c766602d74a61b75b77545a2c159a : SUCCESS in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/211d6b3a118541b3b1284fc470c95597 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b34dcb805c444749d8334fd20dfff92 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19251d9fac274c1692465f658e957970 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55161d1c7ff3460f9e5c222ecb3de8ea : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a28ff1e1e72479882cb0055de0f7379 : SUCCESS in 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/937b6373fe87419da15673c059c2b62f : SUCCESS in 7m 29s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0ab4075bab8946aa80cd59f85068d9d6 : SUCCESS in 4m 39s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/595c31073934433cb09cac856dd3547a : SUCCESS in 6m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0538db8a61484defa5b326dba17e2309 : SUCCESS in 5m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c4bd61c7a35540ef8b638ad3e7aae3a8 : SUCCESS in 6m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/599011e695554d6ab628f925100fa366 : SUCCESS in 8m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4c92caec87e9482cb04e1a45acf7cf4c : SUCCESS in 7m 02s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9bd540d48cc249589bb7d35353f1af6e : SUCCESS in 1h 40m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/76526aec21e045e89f8463eb546be2a9 : TIMED_OUT in 2h 33m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a5e30691af7e4ad78096a145b7555bcd : SUCCESS in 53m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cd30fa2256754c52b30cbc218fbd87aa : SUCCESS in 55m 44s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a61c7a586b7b445ebe59dc93654cd4e1 : SUCCESS in 1h 07m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/23131cbe13dc427fbcb93413274eb865 : SUCCESS in 47m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2c4d948ab1c24772a74fd31eb8eb3c71 : SUCCESS in 2h 18m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/85bcd327a7384c99b455e35388fc0c2f : SUCCESS in 2h 15m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c6d907792c774cb28037609222c0e53a : FAILURE in 37m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1bfb2c67067e4bf98610b3ad0ba93694 : SUCCESS in 2h 08m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1e084b999cae425a87abfbdb0b23dca1 : SUCCESS in 2h 09m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c33f81c480494a80b4ec8876be03cb96 : SUCCESS in 2h 11m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06c9992594a57e054b4fdd6da4b65772d1638f3a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-29 09:32:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40b63e85b038286d556cd34520149fb276b9b02e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-29 11:29:54.000000000","message":"Abandoned\n\nstable/zed branch of openstack/octavia is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/zed if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e8ada397f9a7acc3d25d5557e7ecf6a7742ff319","revisions":{"e8ada397f9a7acc3d25d5557e7ecf6a7742ff319":{"kind":"REWORK","_number":1,"created":"2023-04-14 05:45:34.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/35/880435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/35/880435/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/880435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/880435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/880435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/35/880435/1"}}},"commit":{"parents":[{"commit":"332d7deeba740b9fd8a67065f5818010a5d2990b","subject":"Fix octavia to accept [ipv6]:port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/332d7deeba740b9fd8a67065f5818010a5d2990b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-02-02 09:36:14.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-14 05:44:23.000000000","tz":120},"subject":"Pass config to castellan","message":"Pass config to castellan\n\nCurrently castellan can\u0027t be configured through octavia.conf as\nconf is not passed while initializing backend.\n\nAlso document castellan configuration options in reference.\n\nBackports on stable branches also include [0] which adds a release note.\n\n[0] Iacc796737bad8881873da7db5273338c2cff9e68\n\nConflicts:\n    etc/config/octavia-config-generator.conf (doesn\u0027t exist on stable\n    branches)\n\nChange-Id: I30b81866989c22b94fb77e62e7abd180f0f0af50\n(cherry picked from commit f5ac714a7b22687fbb5b12db7f41c283cee12aee)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e8ada397f9a7acc3d25d5557e7ecf6a7742ff319"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e8ada397f9a7acc3d25d5557e7ecf6a7742ff319"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
