)]}'
{"id":"openstack%2Foctavia~880438","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I30b81866989c22b94fb77e62e7abd180f0f0af50","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-01-23 16:03:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I30b81866989c22b94fb77e62e7abd180f0f0af50","subject":"Pass config to castellan","status":"MERGED","created":"2023-04-14 05:46:24.000000000","updated":"2024-01-23 16:04:45.000000000","submitted":"2024-01-23 16:03:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880438","meta_rev_id":"356b805472134972ac5134a74cfd845ac4ff3dd9","_number":880438,"virtual_id_number":880438,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-23 16:03:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-23 11:00:54.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-23 11:00:54.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-14 05:46:24.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2023-04-14 07:56:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-23 11:00:54.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"96d125f0c9600c9868898ee9eb84ee634b2c29eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-14 05:46:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6817ef898cc04a36b3f26939e0dfb2c49547fa48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 07:56:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07aac467c133423ca858faf712df55ba\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7082144aa4f4942aa0c0cc290573331 : SUCCESS in 10m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5292062f402447f1b52cc28b7633a8a6 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83bb4fc8b4c0477aa4ab7a36972cf6af : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56af267d25e44da3ba9dfa13550e3abd : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7755cd2c72e74135a30a6b739a5036de : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96665f32e8cc412aaafeb6a6ddbb0915 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd0221266d594733b0c57966ea2e3ced : SUCCESS in 12m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0fb06d4c80d48a99d7d082c127c8d41 : SUCCESS in 4m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f01cca2c635b44a9ae6bc299255e4119 : SUCCESS in 5m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/034ea32599f04de19ec3a6ba5748b558 : SUCCESS in 7m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7a8a3425e3f04f00bc7d56e42d9f4800 : SUCCESS in 4m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6d7ea0ac23ac4b4f9ecb86968a422abb : SUCCESS in 7m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/43466e3464614a5f8de5ee3e3944adc9 : SUCCESS in 1h 12m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1b0e9900e430477cb3b04b342ea5e49e : SUCCESS in 1h 42m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4997fedbd3274921a6e4d23743ae28d0 : SUCCESS in 33m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8e71905b3b5845fdb4d3a759f2e54923 : SUCCESS in 38m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/26fdb3e58bb74e4a8443c7b8713d8f14 : SUCCESS in 30m 23s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9a95373ef279401e85aac2f28460590b : FAILURE in 9m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9f7adb3fedd64b609bf9ba2d182ef85f : SUCCESS in 1h 32m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f8730ec67b414f6ead50d71628a58d3c : FAILURE in 34m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c12582860b69412198404f9353f1279d : SUCCESS in 2h 06m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/119998dae54645f59603e774861889a4 : SUCCESS in 1h 58m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/02dbf2f8fc7546089e212705df7a7746 : SUCCESS in 1h 35m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/07112a23848440479579bc3133371f76 : NODE_FAILURE Node request 300-0020975058 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"824e40457492e03f13cfd29efd74edc9cd48c56d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-23 11:00:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7f5a92d199ed3a881eb88ce4f7198dcaea886ba7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 11:01:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"21681f557e9ca86b5684c6f7b63ee1d54246cfde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 16:03:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ac4dd77fcaf40b4ab3c988e6631a397\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8c6b47b855946d18ef506fae39c9848 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e90fdeab0e154f619275332f29bff386 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a8e7e588d4145329f813cfe8803b9a1 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7079214971a45be81ef52dff213a78a : SUCCESS in 11m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f3d90ef897b4646a95fc26a2b62abd7 : SUCCESS in 4m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0bba468bf4a14e8497aacb1cf51edf94 : SUCCESS in 5m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/24a08714bbbd4c36b2c3775a8cc41d56 : SUCCESS in 6m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f6682154dbc5483eb3bea5d27411c4a6 : SUCCESS in 1h 20m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/38c4828ed6a648fb86f511c0765a818b : SUCCESS in 1h 30m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/27a9374a9ee9406397cc8d6bf59197fd : SUCCESS in 37m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0e24f33fd89e45a09a876bf81252e15a : SUCCESS in 32m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/21833a9a6c41405bba1295e5a35fb799 : SUCCESS in 34m 05s","accounts_in_message":[],"_revision_number":1},{"id":"6604280f340e7819ff101d83275c04fa02dbd659","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 16:03:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"356b805472134972ac5134a74cfd845ac4ff3dd9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 16:04:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bebc02c2be2943bcbbf8ed1431bb0f6a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9770984e6c449da91ed9d5cd1752bd0 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a58b4830e58043498ea17bb8acf5c710 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e19279f8edb0a54f3d74ee97ba85815d32c85baa","revisions":{"e19279f8edb0a54f3d74ee97ba85815d32c85baa":{"kind":"REWORK","_number":1,"created":"2023-04-14 05:46:24.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/38/880438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/38/880438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/880438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/880438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/38/880438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/38/880438/1"}}},"commit":{"parents":[{"commit":"a1c40d42ddc327fc8dfdc06b48a49f640ed0e4c6","subject":"Enable taskflow retry feature when waiting for compute","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1c40d42ddc327fc8dfdc06b48a49f640ed0e4c6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-02-02 09:36:14.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-14 05:46:17.000000000","tz":120},"subject":"Pass config to castellan","message":"Pass config to castellan\n\nCurrently castellan can\u0027t be configured through octavia.conf as\nconf is not passed while initializing backend.\n\nAlso document castellan configuration options in reference.\n\nBackports on stable branches also include [0] which adds a release note.\n\n[0] Iacc796737bad8881873da7db5273338c2cff9e68\n\nConflicts:\n    etc/config/octavia-config-generator.conf (doesn\u0027t exist on stable\n    branches)\n\nChange-Id: I30b81866989c22b94fb77e62e7abd180f0f0af50\n(cherry picked from commit f5ac714a7b22687fbb5b12db7f41c283cee12aee)\n(cherry picked from commit e8ada397f9a7acc3d25d5557e7ecf6a7742ff319)\n(cherry picked from commit 8fbe035cd13920f174beb8f528cc93a52a96570f)\n(cherry picked from commit 006e0c5ff0b1dc106e3585491e56bb307a8f908e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e19279f8edb0a54f3d74ee97ba85815d32c85baa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e19279f8edb0a54f3d74ee97ba85815d32c85baa"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
