)]}'
{"id":"openstack%2Foctavia~880446","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~Ic08ad89b4e07e0fe8a80b3e5db6d50276aafff82","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-03-05 19:08:08.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic08ad89b4e07e0fe8a80b3e5db6d50276aafff82","subject":"Rename Context to RequestContext","status":"ABANDONED","created":"2023-04-14 06:01:57.000000000","updated":"2024-03-05 19:08:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"960a848d79fd3583502e83f1b780766bd8ccda82","_number":880446,"virtual_id_number":880446,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-04-14 08:30:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":1,"date":"2024-01-24 16:54:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-23 11:07:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-14 06:01:57.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2023-04-14 08:30:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-23 11:07:24.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-01-24 16:54:11.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2024-03-05 19:08:08.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"e4cac3d58bc2b158e2d79d57e69d8985ddbf5628","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-14 06:01:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"132bc1422b3b29824fef0316f1eee70e34207fd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 08:30:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ed6e55184e44ea8b0a64466e519fc00\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8480a5ee52d49db89c9cc6fa27c099c : SUCCESS in 9m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b3d5347075649e6bd1cc6427c0a404c : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c35972a26cf4639aa81c0d41ae21810 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b71cbf9acd34ad28530aff21145549e : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adba006bc9fd40b395d9dd2fb6a064c6 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69d42627e96c4efbb6623896361ecd8b : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/977b4e5aa9d34d7eb826aab7dee3a5cc : SUCCESS in 12m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8cbb4ce51cef475d8197ea3e8862aa83 : SUCCESS in 5m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ede1b5875112467eabb60477c1980ed6 : SUCCESS in 6m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/88be3cd1844d401f93fbad1fb7aa7c49 : SUCCESS in 4m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/21cd731571b3402eb944b76f1f122c96 : SUCCESS in 6m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/19236efdaf424ba386c9f778ce565a2e : SUCCESS in 1h 05m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/90e347f053a046e88d142a8d0c7362f5 : SUCCESS in 1h 54m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/746f7f536f854a20a73806776f5f8346 : SUCCESS in 40m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/fc001948e49c4063985a8561360a9b44 : SUCCESS in 33m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f97d8639aa234398b21f1c20482ed2ee : SUCCESS in 37m 39s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7169d5ba6f424381a3bacc5aea8569e1 : FAILURE in 10m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/753c41f6b29344d484608ae2e38269a2 : SUCCESS in 1h 57m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4c7f21e31ab648578adc32b0f936aba2 : FAILURE in 31m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/727ffea1396743ecb3187e98abc6b0d1 : SUCCESS in 1h 52m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/354bef3b9fb14c198c9c76c5d42e00ef : SUCCESS in 2h 25m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b16539f2092748ca97e0618d7110e7b0 : SUCCESS in 2h 03m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2c4b1b177cb843afb6fe8bf1ce7f43a1 : NODE_FAILURE Node request 300-0020975191 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b68b07e6c79c84da28c85382a621473cd8ee3235","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-23 11:07:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7fbcc66d645bef8c0653f3a2fdbb4cfd0c6839d2","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2024-01-24 16:54:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"960a848d79fd3583502e83f1b780766bd8ccda82","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-05 19:08:08.000000000","message":"Abandoned\n\nstable/wallaby branch of openstack/octavia is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/wallaby if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1e909db4e94c5422244f7ab1a7bf3545a7e9e37","revisions":{"a1e909db4e94c5422244f7ab1a7bf3545a7e9e37":{"kind":"REWORK","_number":1,"created":"2023-04-14 06:01:57.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/46/880446/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/46/880446/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/46/880446/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/46/880446/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/46/880446/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/46/880446/1"}}},"commit":{"parents":[{"commit":"a1c40d42ddc327fc8dfdc06b48a49f640ed0e4c6","subject":"Enable taskflow retry feature when waiting for compute","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1c40d42ddc327fc8dfdc06b48a49f640ed0e4c6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-02-02 09:18:01.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-14 06:01:50.000000000","tz":120},"subject":"Rename Context to RequestContext","message":"Rename Context to RequestContext\n\nIf castellan is used, it requires one of the following context types to\nbe provided during init: KeystonePassword, KeystoneToken or\nRequestContext [1]\n\nAlso it makes sense to be consistent across projects and with\noslo_context regarding context class naming.\n\n[1] https://opendev.org/openstack/castellan/src/commit/86712360f345866e108e12eda1075101635dd1ec/castellan/key_manager/barbican_key_manager.py#L208-L210\n\nConflicts:\n\toctavia/controller/worker/v2/tasks/notification_tasks.py\n\toctavia/tests/unit/controller/worker/v2/tasks/test_notification_tasks.py\n\toctavia/tests/unit/common/test_policy.py\n\nChange-Id: Ic08ad89b4e07e0fe8a80b3e5db6d50276aafff82\n(cherry picked from commit 62e1d871cbbbeefc4919c0c12a3cfff606c53557)\n(cherry picked from commit 48b9d42a571659240dfa737d99781c9713edaf0b)\n(cherry picked from commit 111194847f0c0bc2fe672513485ba9ce70767db4)\n(cherry picked from commit 15b01f4c3d94d00017bab87035dc90d42922b900)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1e909db4e94c5422244f7ab1a7bf3545a7e9e37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1e909db4e94c5422244f7ab1a7bf3545a7e9e37"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
