)]}'
{"id":"openstack%2Foctavia~880447","triplet_id":"openstack%2Foctavia~stable%2Fzed~I1a9a9177cafb8c52f13a646b12cff1c35afd8679","project":"openstack/octavia","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-04-24 19:11:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1a9a9177cafb8c52f13a646b12cff1c35afd8679","subject":"Fix SQLAlchemy warning about conflict relationship with Tags","status":"MERGED","created":"2023-04-14 06:18:30.000000000","updated":"2024-04-24 19:11:10.000000000","submitted":"2024-04-24 19:11:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880447","meta_rev_id":"7cd91936cb9b950112194a4efa27d2e40eb7113a","_number":880447,"virtual_id_number":880447,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-24 19:11:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":1,"date":"2024-01-24 16:54:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2024-04-24 16:49:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-23 11:09:51.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2024-04-24 16:49:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-14 09:18:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-23 11:09:51.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-01-24 16:54:17.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2024-04-24 16:49:56.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"93db2cf7899d29153ab2f880f640aa57c4210bbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-14 06:18:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d70de101d8186bb735fc6580df40694a2e33fae6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 09:18:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8abef8cea12347b2b894b201ce2ecde2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4ad515b65a947c5a599ad8cbea58251 : SUCCESS in 12m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a81e332f3cc491ab68282848cdc75a0 : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23f22651e2584b7e8da11038d04e4c81 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/217af1cfccda4d30a1459113e13c5a50 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f4642cb9b4a4659a9ceb2f063892a66 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/963473c0f18a4b68a662007c86581afa : SUCCESS in 11m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf5601e0528d4b50a8beb0b57f4192c0 : SUCCESS in 5m 02s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/417f5a1d146c45e4b8043d020229565c : SUCCESS in 4m 44s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/95487db377f6408ba2a6103e2ab62e33 : SUCCESS in 8m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fed63bb0fb06458fb051d722ad912652 : SUCCESS in 4m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bdf3326d761c46fe968550d52eba5f36 : SUCCESS in 7m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7b64173772c94b3db67cb713b3bb6de8 : SUCCESS in 7m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9e4e404b40784820bcdabdb28d4dee30 : SUCCESS in 7m 53s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/366ec874a1d54eddb64277e3585748f5 : SUCCESS in 1h 35m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d4aaaab8788842b98cd77dfb1792eb59 : SUCCESS in 2h 14m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/43c163babf094536832c56800a68b9be : SUCCESS in 38m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e4e1fac1741247a7a04375bf4dd2c743 : SUCCESS in 48m 33s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7d119d04c5b248dc8f5398b22cd9e187 : SUCCESS in 1h 00m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d4a0b761d8e444d0bba6449b7dfbc899 : SUCCESS in 36m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/10f31765d58b455abc93e5e36a6fbcb4 : SUCCESS in 2h 11m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5073034e12ff4e57b25ae9cac20cc486 : SUCCESS in 2h 56m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bd70a760d5f14d3eb92f9f36d7d3cc0e : SUCCESS in 2h 29m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/49311e43fe7642c59a7c3402a6939655 : TIMED_OUT in 2h 33m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8c797c3abb0e4d77ac3e90c0d5e540d1 : SUCCESS in 2h 04m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/681cba220aed4fedbd90538835efe9e1 : SUCCESS in 2h 07m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c0fed9c16702d665440da74d0611cfeeb3004f6","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-23 11:09:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"02ee86b349535b474cb7bd406d3d93ab1c25a6e0","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2024-01-24 16:54:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"01044fa1d5b779fe1183f5923eb725b500b8f9a3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-04-24 16:49:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d33935385ea83084556ecffeafb10dd67ede2394","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 16:50:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"046a848cbe04c4a86723fc3d1145739eb84cf807","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 19:11:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d768c67e2d124b4fbb2059a3e1130e67\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/068289548f2f490d9099fcc129c46feb : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb2142b38f94482ca4674322a56b3e7a : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e48c350210d4159b68bd9fb4fe24f4e : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c26cc4b208454531af7d185cc667d0bc : SUCCESS in 9m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db1757a0ce7647c597454707505487fc : SUCCESS in 6m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/168c2c3a50b941dc8c1794a6d3f98154 : SUCCESS in 5m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/848528f7cd2e4328ab5759c44dd5c827 : SUCCESS in 7m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f12dc473619841f6923898aa5d4cac15 : SUCCESS in 1h 12m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/33a471b577d4481e9a07405db28933fd : SUCCESS in 1h 23m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c20c557368fa4ba9a729ad5eb78fab4d : SUCCESS in 34m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/27d8b8a7ec4c42f58ee4b72663505bbc : SUCCESS in 1h 00m 04s","accounts_in_message":[],"_revision_number":1},{"id":"7cd91936cb9b950112194a4efa27d2e40eb7113a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 19:11:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ae577279fee13b80e3b8e50252d15c70668d388","revisions":{"6ae577279fee13b80e3b8e50252d15c70668d388":{"kind":"REWORK","_number":1,"created":"2023-04-14 06:18:30.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/47/880447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/47/880447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/880447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/880447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/47/880447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/47/880447/1"}}},"commit":{"parents":[{"commit":"332d7deeba740b9fd8a67065f5818010a5d2990b","subject":"Fix octavia to accept [ipv6]:port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/332d7deeba740b9fd8a67065f5818010a5d2990b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-23 16:19:38.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-14 06:17:45.000000000","tz":120},"subject":"Fix SQLAlchemy warning about conflict relationship with Tags","message":"Fix SQLAlchemy warning about conflict relationship with Tags\n\nSQLAlchemy shows a warning about a conflict with a relationship with the\nTags object:\n\n  SAWarning: relationship \u0027\u003cObj1\u003e._tags\u0027 will copy column \u003cobj1\u003e.id to\n  column tags.resource_id, which conflicts with relationship(s):\n  \u0027\u003cObj2\u003e._tags\u0027\n\nRemove this warning by using the overlaps [0] keyword in the definition\nof the relationship.\n\n[0] https://docs.sqlalchemy.org/en/20/orm/relationship_api.html#\\\n    sqlalchemy.orm.relationship.params.overlaps\n\nChange-Id: I1a9a9177cafb8c52f13a646b12cff1c35afd8679\n(cherry picked from commit 8409c06cda0bb3e3c0795b14ec3917ca2b5afe7b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ae577279fee13b80e3b8e50252d15c70668d388"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ae577279fee13b80e3b8e50252d15c70668d388"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
