)]}'
{"id":"openstack%2Foctavia~880449","triplet_id":"openstack%2Foctavia~stable%2Fxena~I1a9a9177cafb8c52f13a646b12cff1c35afd8679","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-01-23 16:11:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1a9a9177cafb8c52f13a646b12cff1c35afd8679","subject":"Fix SQLAlchemy warning about conflict relationship with Tags","status":"MERGED","created":"2023-04-14 06:18:55.000000000","updated":"2024-01-23 16:13:01.000000000","submitted":"2024-01-23 16:11:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880449","meta_rev_id":"a72432439a3162360aba0943003468a983d6b6df","_number":880449,"virtual_id_number":880449,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-23 16:11:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-10-12 17:57:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-23 11:12:31.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-23 12:44:32.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-17 16:41:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-12 17:57:28.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-01-23 11:12:31.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"1416cc7cb3f8160f96f4999218edb3ca4d7619af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-14 06:18:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40664b1b5e7f1e4b11af94371c16b508eba0aea9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 16:41:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dcf7325960e144aba35941553565b1ca\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ca65b6069e34343bd6635dbf8164c97 : SUCCESS in 10m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/147ad9c6779041c29257237c9ac8a31d : FAILURE in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfa6bccde23f4572a6abc75523fa28a1 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a32b344c0833497ca41188313f9ca19d : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4872a851ef9a411e8e44e650dd1451f4 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec89457c07af40cdae2e576d0c002ceb : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1824728f187a44889195dad5400ed92c : SUCCESS in 11m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c70e2555cb5545b6b6efbd39b9975a81 : SUCCESS in 5m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a99e4d06606a47768ba2213e5deaa3b7 : SUCCESS in 4m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3da0ccd058374e2da621b9913895cb75 : SUCCESS in 7m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/18c72050ff3c42458416fd30bef81e7c : SUCCESS in 4m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3ebd3e9f37c64b359d5e0173beb69330 : SUCCESS in 7m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/44e8bbe99bd04ca584c6ad90bb64807f : SUCCESS in 1h 23m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c3e69f8901e4f9ba1bcdcddd56b5ad3 : FAILURE in 7m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5faa6b0970cc4856b3f542961c2b1ba1 : SUCCESS in 35m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f9b7a59bbd224f3287ee1f36f7101e3f : SUCCESS in 47m 34s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/eb3cfe832b1e4e0c8289d4ffacccffa7 : FAILURE in 10m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b440e49407d641c2a3b505f78739b25c : SUCCESS in 27m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1a7753a619e340078e06cfd68186833b : SUCCESS in 1h 46m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1eb3a850aa644e008fd7054c3e286072 : SUCCESS in 2h 22m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1ce8b9125b4c43bb8bff0586ae106ef0 : SUCCESS in 2h 10m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/975353a8137a4016b5af6c0c17df0330 : SUCCESS in 2h 10m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2c8e279a76c745a9b7b545985cff53f3 : SUCCESS in 2h 11m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c7e6e9b87bd9485cb89f2b10d58337a6 : NODE_FAILURE Node request 300-0020975405 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63e3cdfaad406889d6f60e3a014be0a77f97ed65","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-14 06:44:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84ecbc7c2414a28dcd454ac35be979dbde26cf73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 09:14:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50985b2209764469b224288ab1f9ec63\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96c2dbf4d9124e7aba907d5eede9f0ee : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3c6836247e044489105ba3e16075d96 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/886f7373720d4a67be47f1d4850200a6 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a10fae892f764dd29cbc0809e225eaa5 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35fe8730c06d41feb180cb78ebbb2f97 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1e098a7ba034f938fa2bb3e55bd225b : SUCCESS in 12m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa70544b4ef048ddac2dae8a6b22b5e7 : SUCCESS in 4m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/38f79711d46d43bda6b127d7d24a6c12 : SUCCESS in 6m 42s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b43f256b53fb4f10a453bc7ed5fb74cd : SUCCESS in 8m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/84cbb1012d0641b6a67e65b50c32ad41 : SUCCESS in 4m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/501f773fe91847fc94bd708a630a2c74 : SUCCESS in 9m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/67a7501ea83a48289455ffd1377f5ed7 : SUCCESS in 1h 30m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f4ebff6af324366a8fedff6f2ae54b1 : SUCCESS in 2h 23m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f3dc3e242a734f83a15cd55d6eacac81 : SUCCESS in 39m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aebf79b4614c4b979cc4c621cf51b361 : SUCCESS in 57m 38s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e21d9ae39f5a457b9cb30ffcd73fe70b : FAILURE in 9m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8723370ed0834170897bebfeec2adfb5 : SUCCESS in 43m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ab39c563b78c44ada0c0ac2e0a9009a6 : SUCCESS in 2h 00m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/08c1c87e906a4335b903fc773dfc4c70 : FAILURE in 14m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/11730f9445224d0b815578cebb3ccbe1 : FAILURE in 5m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1667c878d2da40ac81fd0ebbf7fae196 : SUCCESS in 1h 57m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1b10ad3107074dc5890601095f4c96df : SUCCESS in 1h 33m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3ec50582365c40eeadc489e0a3cca06c : NODE_FAILURE Node request 300-0022227470 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2195af56b15814754ce84bc1541040a2ebff0d3b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 17:57:28.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d1eb878e348b3bda746e01de070609c415bf83be","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-23 11:12:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"705b5f8cf555f8c8bf97dc94e395c25b853c0d64","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-23 12:44:32.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a6aa5e6f113fee3015b1c406ab50ec3c892a6d74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 12:45:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e5ba27dd29ae935c95d7bc6d7566298d8dab3b26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 16:11:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c52a2aa187641299ef33f4df3826168\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b0e62cc50604baa85378b2137dda61c : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/186dbbfd9c144a0d97f1377535c0a4a6 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4035b56294d949079722d47f0b331873 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93ca75cc43b1462089c2db34d8ebbd32 : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e15de12ee084c7f8f0e317aa6596e7f : SUCCESS in 5m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/eea86d07d91040c7a4165b79e0ed7f61 : SUCCESS in 5m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/98b77f27e8ab4094ac70f39ddea9928c : SUCCESS in 6m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2d6dcbab555e414bb2b08e4cdcec62e8 : SUCCESS in 1h 18m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1cb6180b339540f6946f6c5af1740e08 : SUCCESS in 1h 26m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/64062bdd936e4d1ba2f1d80b8d02238e : SUCCESS in 32m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/daaf5f4e001047368e584bcdfeffad44 : SUCCESS in 40m 45s","accounts_in_message":[],"_revision_number":1},{"id":"eb492628aee30a9b734f1c14b9c2ffadf8de5b7b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 16:11:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a72432439a3162360aba0943003468a983d6b6df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 16:13:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03b7639c56424beba89b8757dd131bef\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a96adeabeec4eceb220f954d0202d57 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c8cd369158744be8a7a13af83aea750 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"218d667892332e0d548d36eb563dcfac482087c2","revisions":{"218d667892332e0d548d36eb563dcfac482087c2":{"kind":"REWORK","_number":1,"created":"2023-04-14 06:18:55.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/880449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/880449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/880449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/880449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/880449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/880449/1"}}},"commit":{"parents":[{"commit":"195383c0c2ecf5b35afcd6028b45b22b52bfc9aa","subject":"Merge \"Prevent incorrect reschedule of resumed tasks with jobboard\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/195383c0c2ecf5b35afcd6028b45b22b52bfc9aa"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-02-23 16:19:38.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-14 06:18:48.000000000","tz":120},"subject":"Fix SQLAlchemy warning about conflict relationship with Tags","message":"Fix SQLAlchemy warning about conflict relationship with Tags\n\nSQLAlchemy shows a warning about a conflict with a relationship with the\nTags object:\n\n  SAWarning: relationship \u0027\u003cObj1\u003e._tags\u0027 will copy column \u003cobj1\u003e.id to\n  column tags.resource_id, which conflicts with relationship(s):\n  \u0027\u003cObj2\u003e._tags\u0027\n\nRemove this warning by using the overlaps [0] keyword in the definition\nof the relationship.\n\n[0] https://docs.sqlalchemy.org/en/20/orm/relationship_api.html#\\\n    sqlalchemy.orm.relationship.params.overlaps\n\nChange-Id: I1a9a9177cafb8c52f13a646b12cff1c35afd8679\n(cherry picked from commit 8409c06cda0bb3e3c0795b14ec3917ca2b5afe7b)\n(cherry picked from commit 6ae577279fee13b80e3b8e50252d15c70668d388)\n(cherry picked from commit 64b8846eda6ffed48a5964a8c285028dc0fcd99c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/218d667892332e0d548d36eb563dcfac482087c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/218d667892332e0d548d36eb563dcfac482087c2"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
