)]}'
{"id":"openstack%2Foctavia~880451","triplet_id":"openstack%2Foctavia~stable%2F2023.1~I02804b7075edac72776b0377b7b283d0c7bfd8a2","project":"openstack/octavia","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-12 20:52:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I02804b7075edac72776b0377b7b283d0c7bfd8a2","subject":"Fix failover when the last listener is deleted","status":"MERGED","created":"2023-04-14 06:46:10.000000000","updated":"2023-10-12 20:53:34.000000000","submitted":"2023-10-12 20:52:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880451","meta_rev_id":"25c60dcaa6da10d8c843c30ebe180f4988ce8bf2","_number":880451,"virtual_id_number":880451,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-12 20:52:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2023-05-05 06:04:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-10-12 18:22:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-17 12:13:19.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-10-12 18:22:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-14 06:46:10.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2023-04-17 12:13:19.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-04-17 18:40:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-05 06:04:40.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-10-12 18:22:43.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"9615565d36a1c5e0ea2cb86c66610d6e3e6fc8a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-14 06:46:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf4e2b8f968b222d2f894b540763893322cd5fb9","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-17 12:13:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"326eabe4b24659c92e95946fbbcdfc697c97ace1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 18:40:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b34e73da88094c7c87478195b73ed381\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52971fd7ac2648628f101ac5d4e1635b : SUCCESS in 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff34e3df204141dab590e5cfced7d9f7 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17c71728eb9f4d7bb859f2c6ac94a0f6 : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b97590cc9bfa4c19a78a72700393310a : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1369174ee09e4a20808249968d4a4ee2 : SUCCESS in 12m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2497e43b569945a981d5aabfec65a953 : SUCCESS in 4m 37s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/aeb9205d21e24a9e852ef92960694b4e : SUCCESS in 5m 06s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a5535cc9e551485c8982b485346cc56c : SUCCESS in 8m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3bf78f04a7794c5597d40de54276146e : SUCCESS in 4m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e3026befd00041179222c647541c0c1d : SUCCESS in 8m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/51fd95bfc8b044038ba97d8a4fec6d42 : SUCCESS in 7m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fbb9acae37884723b85da50edb067add : SUCCESS in 8m 49s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9a41c77f22764865addedbf2cb107c71 : SUCCESS in 1h 01m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e66f88d9ced14ae09eca853030050c49 : SUCCESS in 2h 21m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6852948425644535ac74adc5a1de7acc : SUCCESS in 35m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/24a5678f72804c01b59165a1c3c2749e : SUCCESS in 38m 33s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/4c224c0d87f44e96b193e49529ebcd8e : SUCCESS in 55m 12s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/61d0a7df44d7407dafd33018f7a3ab46 : SUCCESS in 1h 01m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/882b57f17ddd44e99a3b3d84661d977f : SUCCESS in 36m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3c944e51f1bc4d4b920e585878b4dd9a : SUCCESS in 1h 56m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/44e262ef55ea40dfa8f83f532bca6e1c : SUCCESS in 2h 13m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0ef58e05f0254e74aec892603e29098b : SUCCESS in 1h 53m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c5d9088b3c17401c9f388c4233e5125e : SUCCESS in 2h 01m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d870ba780efd43c69b9437052c6b53e1 : SUCCESS in 1h 38m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b32c90c0344f43f184538fad29800c2f : SUCCESS in 2h 04m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6693f7314c14f27894704a19710056c634ba9fe5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-05-05 06:04:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9413308904f5b2ab5fc9ec5821b294f058f0b39e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 18:22:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f3e8aad4c86dfbd95f30e24b992f8dedc913bc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 18:23:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"adf09da3366b458ee7ccff73fa0fea48299f6a4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 20:52:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bfb26ff51a64b06b46a251b8cd3c0aa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa0e4b04fbbe4fd990c4b764c5d9f125 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb1e156e600b48228ee49018f67c0789 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a72a678fb4b48328e18effaba2ac4ee : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1f6bb0914b64821a6858370011121fe : SUCCESS in 13m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/043e7d982b6d490d92ffe47564b8b2c4 : SUCCESS in 5m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/48f20abc92ab4decbf177d337dfa7673 : SUCCESS in 5m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c40ea78bba57416587240c49a15854d6 : SUCCESS in 8m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/16873a08a05b4875bf1b9d1990679fd2 : SUCCESS in 57m 55s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9da5eef229724821a003918924a20f90 : SUCCESS in 1h 21m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/42e0ae11039d424b9ef119fa313c8cea : SUCCESS in 1h 01m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4e74544d737a49bc99fc681b48e3610d : SUCCESS in 37m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3f5d388f26d6443ab0936bf840034e26 : SUCCESS in 40m 45s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/5a2fe0d0326d417b97f6a9ccec75ac7f : SUCCESS in 40m 44s","accounts_in_message":[],"_revision_number":1},{"id":"ccdda77984876c22e87363717ff342bde63ecc50","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 20:52:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"25c60dcaa6da10d8c843c30ebe180f4988ce8bf2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 20:53:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/523270061b5546bf8922b0d0fed57daf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d9b111ceced464c8529e267f50a3ff5 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd9c71ea60e04d609c091f183ee0af5b : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7f4b1b4c21d0a2b12022032caa38602f5341a20","revisions":{"e7f4b1b4c21d0a2b12022032caa38602f5341a20":{"kind":"REWORK","_number":1,"created":"2023-04-14 06:46:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/51/880451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/51/880451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/880451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/880451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/51/880451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/51/880451/1"}}},"commit":{"parents":[{"commit":"11512bf61292f889fbdce33510ae2aa239d05abf","subject":"Merge \"Add octavia-grenade-slurp CI job\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/11512bf61292f889fbdce33510ae2aa239d05abf"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-17 22:17:05.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-14 06:19:53.000000000","tz":120},"subject":"Fix failover when the last listener is deleted","message":"Fix failover when the last listener is deleted\n\nThis patch fixes an issue when deleting the last listener from a load balancer\nmay trigger a failover.\n\nStory: 2010652\nTask: 47683\nChange-Id: I02804b7075edac72776b0377b7b283d0c7bfd8a2\n(cherry picked from commit ba92aeb946343c045e8363a2aae78e0a17a6cb82)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e7f4b1b4c21d0a2b12022032caa38602f5341a20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e7f4b1b4c21d0a2b12022032caa38602f5341a20"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
