)]}'
{"id":"openstack%2Foctavia~880452","triplet_id":"openstack%2Foctavia~stable%2Fzed~I02804b7075edac72776b0377b7b283d0c7bfd8a2","project":"openstack/octavia","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-05-05 06:04:45.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-12 20:52:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I02804b7075edac72776b0377b7b283d0c7bfd8a2","subject":"Fix failover when the last listener is deleted","status":"MERGED","created":"2023-04-14 06:47:02.000000000","updated":"2023-10-12 20:53:13.000000000","submitted":"2023-10-12 20:52:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880452","meta_rev_id":"b09e14772a0e2e9116f91231b2e71b41207ff59e","_number":880452,"virtual_id_number":880452,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-12 20:52:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2023-05-05 06:04:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-10-12 18:22:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-17 12:16:07.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-10-12 18:22:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-14 06:47:02.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2023-04-17 12:16:07.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-04-17 17:33:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-24 08:06:40.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-10-12 18:22:29.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"954776aff94b6bd79b1982b7270780b344c12cd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-14 06:47:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"969345f37dd9a27adcbd732e059d376f1f43c954","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-17 12:13:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e35f73bbf118fad962d53d0fe04ed5ffa95acec2","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-17 12:16:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1802ae5a3d2f064d5a0efd049b1dbb987074c984","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 17:33:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8553a011c0244de4ba2318ca14177fc0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f97aba1bdbce479cbb6db23826bb0d54 : SUCCESS in 10m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8f7d4d9f5fc45aaba0cf04e4f6a3b3c : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1a2b896352e4ef3801054bd9c5e55f9 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcb024945ccb4e2fb24c4053a587b0dc : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2d67310307b43b8a77e8e85113b681c : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7576d24f28245038bb8188acd4e3823 : SUCCESS in 11m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ed3ec29344a4b6e9364d28df7cfbcd3 : SUCCESS in 5m 08s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c8dd22a5d9434f9c8742379a7e15591e : SUCCESS in 4m 48s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ed324ea956d34081a79ebefbad67b4a9 : SUCCESS in 8m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/15af8acc7b0447d3b6cc1ea821d2eedf : SUCCESS in 4m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/88c8ca6c70e54a7aaab1d7d3946a8f2e : SUCCESS in 7m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ad20f1b589934d4d98a603d2c213f1d1 : SUCCESS in 7m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a17302e619ba44a1b2d68bc1a3a80447 : SUCCESS in 7m 28s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/03869e1a728449edae8d05eb21d0677d : SUCCESS in 1h 29m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3f201935131e40d88431775c129261b0 : SUCCESS in 1h 36m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/11f64f36755a4139834248985a5f07c2 : SUCCESS in 33m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e88a01252bb048d293bfe83ca707c84e : SUCCESS in 36m 20s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5233b4156d3043299cd551aa83af9bf1 : SUCCESS in 1h 00m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c6382531d84248539ab574f89cee52e5 : SUCCESS in 29m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9b4456e982a04c048eb6136727456c28 : FAILURE in 6m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b787cad1196c4ad9bf493917c02a195e : SUCCESS in 2h 48m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8a4da5499034414e808a47708c894520 : SUCCESS in 1h 57m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e6e4637250164913a7d7b056ef650641 : SUCCESS in 1h 58m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/02b1b37474bb4b83a04069e2d30b7815 : SUCCESS in 1h 53m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/59372f56665a47ff9ac4a9d686b65c63 : SUCCESS in 2h 22m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d36dd6b3c0a46aa8b1d17c102b3499d244987e20","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-05-05 06:04:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1344536a68860cf5c7a6fa863f8596337c610f79","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 18:22:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0b4397387a9c09e3b82d02772d3e3efcbd3087b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 18:23:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed646d2ccbc07516df26a24a24f793610e545509","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 20:52:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd242fd156bc49fca6608017175f675e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3967b3569d0349ac80f1ea3e5327ce90 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33128ed1054b41a0bfeabbe08c51dacf : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c2a070302df46c2ab8037a2cabb00cb : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a7dfcab496047c999efb90ed9a45954 : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aab011200e5f453ea026b5d80b4e3811 : SUCCESS in 5m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/84166ec3d8484655a8b749301405afe4 : SUCCESS in 5m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bfa0e809512644aaa9de50652938709b : SUCCESS in 8m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/24e2a3d38b5c4613bf436e445dcd6a52 : SUCCESS in 1h 19m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b03d1e666730467cad4649a0689834a2 : SUCCESS in 2h 27m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/45ed5ac56d1e4965888eb61ccac849d1 : SUCCESS in 42m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/128c5859123c4f7283195d54333daa36 : SUCCESS in 51m 24s","accounts_in_message":[],"_revision_number":1},{"id":"e9ed2e25506f20f4554cd40a91040345ccf47614","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 20:52:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b09e14772a0e2e9116f91231b2e71b41207ff59e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 20:53:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d596addcf5e4624b166da47cbb093b3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85d4fc32f1ad4c36b4f7d58224477c57 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/877fcd7a41b0469a9ec8f5007b521bf4 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"658173e5b7f5dcbe5e309801b7d8488442f858e4","revisions":{"658173e5b7f5dcbe5e309801b7d8488442f858e4":{"kind":"REWORK","_number":1,"created":"2023-04-14 06:47:02.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/52/880452/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/52/880452/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/880452/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/880452/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/880452/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/52/880452/1"}}},"commit":{"parents":[{"commit":"332d7deeba740b9fd8a67065f5818010a5d2990b","subject":"Fix octavia to accept [ipv6]:port","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/332d7deeba740b9fd8a67065f5818010a5d2990b"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-03-17 22:17:05.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-14 06:46:21.000000000","tz":120},"subject":"Fix failover when the last listener is deleted","message":"Fix failover when the last listener is deleted\n\nThis patch fixes an issue when deleting the last listener from a load balancer\nmay trigger a failover.\n\nStory: 2010652\nTask: 47683\nChange-Id: I02804b7075edac72776b0377b7b283d0c7bfd8a2\n(cherry picked from commit ba92aeb946343c045e8363a2aae78e0a17a6cb82)\n(cherry picked from commit e7f4b1b4c21d0a2b12022032caa38602f5341a20)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/658173e5b7f5dcbe5e309801b7d8488442f858e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/658173e5b7f5dcbe5e309801b7d8488442f858e4"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
