)]}'
{"id":"openstack%2Foctavia~880503","triplet_id":"openstack%2Foctavia~stable%2F2023.1~I3706fd5e12e17be37edce974563c6806d4f09709","project":"openstack/octavia","branch":"stable/2023.1","topic":"multi-vip-allowed-cidr-stable/2023.1","attention_set":{},"removed_from_attention_set":{"11290":{"account":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"last_update":"2024-01-25 12:49:25.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-01-25 12:49:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3706fd5e12e17be37edce974563c6806d4f09709","subject":"allowed_cidr validation for additional_vips","status":"MERGED","created":"2023-04-15 20:30:36.000000000","updated":"2024-01-25 12:51:48.000000000","submitted":"2024-01-25 12:49:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"880503-multi-vip-allowed-cidr-stable/2023.1","cherry_pick_of_change":876042,"cherry_pick_of_patch_set":1,"meta_rev_id":"7ae797c6d9687ea0e3630cd8f285df051a46b5d9","_number":880503,"virtual_id_number":880503,"owner":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-25 12:49:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-07-06 12:50:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-25 10:33:11.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-25 10:33:11.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-15 22:42:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-06 12:50:42.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-10-12 18:21:17.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2024-01-25 10:33:11.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"491aef5d1da42825842ebd59bec941e2dfc7f1cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2023-04-15 20:30:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"79f46f4af9650d6fcd85c6a512295a5abbb76cb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-15 22:42:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72550d5d8cdd488599b16f01190534b0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd0bb73447ac4cae8c4fa09b679eab39 : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67a3b70ab15e455daacca511d8d79524 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd95aaa2a4334f8580cb8476c61a9145 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/448a5f3e4af04035849ea13edfcf115a : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13e66e03c96546a7b3c8f6ba71b7b17b : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/077255b79d3440e0a3e551f9ea629fad : SUCCESS in 4m 22s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/47eb1569d5384191be1c294723b57353 : SUCCESS in 4m 59s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d1f0c385b59b48e9b66d9f3f4fc07dd5 : SUCCESS in 7m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/466e7fc85f914da4b2d1d83d49405188 : SUCCESS in 4m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6483176ce28244d8b3daebcae862c16b : SUCCESS in 6m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0018ff41beaf4ceeba826df221f99b73 : SUCCESS in 6m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0737a442a2974cf29bd3b18f69d51ece : SUCCESS in 6m 27s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f3f655f79e804100b903dbe62b0faac4 : SUCCESS in 1h 06m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9fbcb29003db41648a238ec1acbf7467 : SUCCESS in 1h 36m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/90f81a54453d43d49d0cd0d3ea208a50 : SUCCESS in 36m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/900f434d99084ab59e0f55efa229a936 : SUCCESS in 38m 58s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/ebc91a33575643f98a484254f1e182e6 : SUCCESS in 45m 19s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/198294f844934429ab021a083c448a82 : SUCCESS in 51m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8bf8af7d83124e6d89c5ac6402bf0ea0 : SUCCESS in 33m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a525072da1c848ca862d4687c22bc150 : FAILURE in 7m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/481148c074324c3296b4aad6c4e0cfe3 : SUCCESS in 2h 09m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/05fb199478f44b31a04fbb27016752bf : SUCCESS in 1h 36m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f7e46de2796b4d6eb5cb35b24f43bc3c : SUCCESS in 1h 52m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/86ba12c4c8d14163951a6604d8e9b072 : SUCCESS in 1h 33m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/47bc414a969b43e0b46ca1b2ca2430d3 : SUCCESS in 2h 00m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fece38d87154d1b007df10cd71d8042b9ccc7c2f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-07-06 12:50:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"29fba24e8f2475ac3852143537913c47af64b983","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 18:21:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"37adfc192469c116272131191e42bbefa53d104b","author":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"date":"2023-10-26 09:25:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b977ba445c09658519c153d650aa59dbb60a6a68","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-25 10:33:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a04f78cb7e2ad1be8386b81cc6ef87305d979195","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 10:34:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ccd2800b783be380e2193fcba48339eb8ce43f35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 12:49:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f85f717ed3e4076b429d1ff73b67a8f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/740c7dab10f54259a91d380650c6d075 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbe12865a87447d19abbec17325501fa : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a37852415b3249e4a94e448d0de18181 : SUCCESS in 12m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3692d9e6316f447788da65b30376865b : SUCCESS in 5m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5bacee23a21c441293f240883a2e599b : SUCCESS in 5m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4c2254557a38496391aa6b75797da79f : SUCCESS in 9m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/571bf27e884d4351a2a9d325a3a27ee1 : SUCCESS in 1h 26m 02s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/45e59017cc8a40fd820343c1580d481a : SUCCESS in 1h 52m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/75f9934daded4674b0811521b7a6f197 : SUCCESS in 1h 19m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1ba6c3fa94a14e8383315b999aed1554 : SUCCESS in 33m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/00b4e70bfba44276a1c4a87c35ea6b43 : SUCCESS in 1h 03m 53s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/93162f905d0b4529b8fc7b4c441e77ee : SUCCESS in 56m 55s","accounts_in_message":[],"_revision_number":1},{"id":"9579abac5272a786f146aff406fa7d6ba4dba7cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 12:49:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7ae797c6d9687ea0e3630cd8f285df051a46b5d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 12:51:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4ed523d0aa84a909314827c5642a3df\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a93d06dba9e45bcb39cec896da04e0f : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a88acf1d67443ff9548b613c4be4b9b : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5783c6767f682af25b2e4e8da560af98a3109913","revisions":{"5783c6767f682af25b2e4e8da560af98a3109913":{"kind":"REWORK","_number":1,"created":"2023-04-15 20:30:36.000000000","uploader":{"_account_id":11290,"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","username":"gaudenz"},"ref":"refs/changes/03/880503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/880503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/880503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/880503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/880503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/880503/1"}}},"commit":{"parents":[{"commit":"11512bf61292f889fbdce33510ae2aa239d05abf","subject":"Merge \"Add octavia-grenade-slurp CI job\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/11512bf61292f889fbdce33510ae2aa239d05abf"}]}],"author":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-03-01 08:52:00.000000000","tz":60},"committer":{"name":"Gaudenz Steinlin","email":"gaudenz.steinlin@cloudscale.ch","date":"2023-04-15 20:30:36.000000000","tz":0},"subject":"allowed_cidr validation for additional_vips","message":"allowed_cidr validation for additional_vips\n\nThe validation for the allowed_cidr parameter did not take into account\nthe IP version of additional vIPs. The parameter was rejected if the IP\nversion did not match the IP version of the primary vIP. As additional\nvIPs can have a different IP version from the primary vIP all vIPs must\nbe checked during validation.\n\nChange-Id: I3706fd5e12e17be37edce974563c6806d4f09709\n(cherry picked from commit 60f579b64a8f603e5a67860a90cd413a0e9ca381)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5783c6767f682af25b2e4e8da560af98a3109913"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5783c6767f682af25b2e4e8da560af98a3109913"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
