)]}'
{"id":"openstack%2Foctavia~880723","triplet_id":"openstack%2Foctavia~stable%2F2023.1~Iecedb9e9c8f293cd4c067900e1696f5050769ee8","project":"openstack/octavia","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-06-07 17:16:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iecedb9e9c8f293cd4c067900e1696f5050769ee8","subject":"Send IP advertisements when plugging a new member subnet","status":"MERGED","created":"2023-04-18 09:37:21.000000000","updated":"2023-06-07 17:17:06.000000000","submitted":"2023-06-07 17:16:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880723","meta_rev_id":"5ce1f6ad958d9ad31d1b95bba7184d00cbb3c5d9","_number":880723,"virtual_id_number":880723,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-07 17:16:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2023-05-05 06:05:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-06-07 15:24:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-24 08:12:22.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-06-07 15:24:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-18 12:26:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-24 08:12:22.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-05-05 06:05:49.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-06-07 15:24:24.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"4f666195055429c263a021d59ac3e6c894e4ae3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-04-18 09:37:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9674b64a79a24a5562bf88f938e2d2af41fab62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 12:26:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a683ee50b914ac4bba7c1b4a5f3d816\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f876add864f141b4bb75a3412b0b3e08 : SUCCESS in 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd494ec4935c46faad9a8cf8f3d09aed : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f122563f859e4a6a8f8830f605602708 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94b95e6fb05a4cd5b06e419f01cb456d : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f497b2c6c53e440d8d658e964039302a : SUCCESS in 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b47237f2f4e4a18811094324c187fc7 : SUCCESS in 4m 57s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/bfe3440c55f14e17bc6d9d4f8f473798 : SUCCESS in 5m 36s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/6d7ba5e1ea0a4e6f967c5a4102972765 : SUCCESS in 8m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cf3c57d98cf746fab3e23fc76177079e : SUCCESS in 4m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d2395052bc5e4bd19386a5e86e18ed70 : SUCCESS in 8m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/57aaafade9674edd99a7e467d922f799 : SUCCESS in 7m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0f77c0343e37441e8f94a5c786fd1434 : SUCCESS in 8m 45s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bd243f50e3a7431bbebf92e227ffe5aa : SUCCESS in 1h 36m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d36a3eeaef94b73a5b97fa3ae9229fc : SUCCESS in 2h 20m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3452cb4c6c954687aaa5e691616cc051 : SUCCESS in 37m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c23881ad6a4843fe8a5192b400dfe2d7 : SUCCESS in 54m 20s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/2b3dc051695c498cb800842d42c7777c : SUCCESS in 54m 15s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5308c91b34c14ff98269af9f4eb0d68e : SUCCESS in 1h 09m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a83cce75f6084cedaa22960cee124c71 : SUCCESS in 34m 08s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b046ca02c08f4df2a016f168b4329fd2 : TIMED_OUT in 2h 33m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/af20f93a6fe14eff92bd4eec35eb13ee : SUCCESS in 2h 31m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/01ae6ba1235d435098af4d431f9bf855 : SUCCESS in 2h 15m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/35673b479220449d9a3811a6b9ddfcf5 : TIMED_OUT in 2h 34m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/dc8bccd85db340569f322eb223ba3e7d : NODE_FAILURE Node request 300-0020999367 failed in 0s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f1b0c35f04b451db4a822d0aa4e2071 : SUCCESS in 2h 32m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"99071839fc94da1ff93bff7f006a1d13b47c087c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-24 08:12:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6e4129d294aa347c7251b0ea9be041d8c20e57e8","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-05-05 06:05:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"847024b21eb6a9f641c84fdb037f3c534e9c01da","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-06-07 15:24:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e1b1a35bd523f0c5c938233e99c464d09a8abab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 15:24:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f79ae00db150472b43da30c76fe6e5c432b9fbe3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 17:16:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/653abd4d96b54dd5b9e257be15059d05\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4bf10a53443454ca441f4aed1090004 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/040a021ae5694cbd918016ff29145a57 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44c3bb349c2543d4a59785400367f0ff : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abdcbc3dfdc24fa7b16c83608d8128e2 : SUCCESS in 10m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/239333fd47b64700a2d498600acb4983 : SUCCESS in 5m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8893cdabba524cba961162fab7051529 : SUCCESS in 5m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3711c984843c4e02bc8639cb7e86b7f1 : SUCCESS in 8m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/545ead87c1234134ba91c5ad2824675e : SUCCESS in 1h 15m 17s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/057468827dc542f3ae7249aa9eb50ebd : SUCCESS in 1h 49m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2a64fb00903e41558ffdfe12ae473530 : SUCCESS in 1h 21m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4dec10abced14bf4b865d840af322acc : SUCCESS in 36m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a87774ca25814bad9a10587bc789a347 : SUCCESS in 38m 33s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/696db8faa2ac4576bc3d1927825ddeab : SUCCESS in 39m 25s","accounts_in_message":[],"_revision_number":1},{"id":"476023f0c6bd9480d9839f238bd0d5569028e1d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 17:16:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5ce1f6ad958d9ad31d1b95bba7184d00cbb3c5d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 17:17:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4441e6c09559488b9cccd23c1d4b70d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c73c3cb5f6a4526ba80f33093c4b404 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71028a34f17049399eff7ce52d192681 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f03a2df9aa10f6901020292b2f2f4eaa912c4cb8","revisions":{"f03a2df9aa10f6901020292b2f2f4eaa912c4cb8":{"kind":"REWORK","_number":1,"created":"2023-04-18 09:37:21.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/23/880723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/880723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/880723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/880723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/880723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/880723/1"}}},"commit":{"parents":[{"commit":"11512bf61292f889fbdce33510ae2aa239d05abf","subject":"Merge \"Add octavia-grenade-slurp CI job\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/11512bf61292f889fbdce33510ae2aa239d05abf"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-07 12:04:45.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-18 09:37:08.000000000","tz":120},"subject":"Send IP advertisements when plugging a new member subnet","message":"Send IP advertisements when plugging a new member subnet\n\nWhen plugging a new subnet, the amphora sends a packet advertising the\nIP address of the port. In case previously deallocated IP addresses are\nre-used, it helps the other resources on the same L2 network to flush\nthe old ARP entries.\n\nCloses-Bug: #2015572\n\nChange-Id: Iecedb9e9c8f293cd4c067900e1696f5050769ee8\n(cherry picked from commit 18cd1f34c8a918ab6f969a3271ac74132c4ac736)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f03a2df9aa10f6901020292b2f2f4eaa912c4cb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f03a2df9aa10f6901020292b2f2f4eaa912c4cb8"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
